)]}'
{"id":"openstack%2Fnova-specs~824191","triplet_id":"openstack%2Fnova-specs~master~I463273d35050116933b4e6bad44b6bddedf02fa3","project":"openstack/nova-specs","branch":"master","topic":"nova-support-lightos-driver","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-01-13 10:46:17.000000000","reason":"sean mooney replied on the change"},"34459":{"account":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"last_update":"2022-01-12 18:02:02.000000000","reason":"Muli Ben-Yehuda replied on the change"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-01-13 12:20:32.000000000","reason":"Change was submitted"},"33612":{"account":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"last_update":"2022-01-13 12:20:32.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2022-01-13 11:55:57.000000000","reason":"Sylvain Bauza replied on the change"}},"hashtags":[],"change_id":"I463273d35050116933b4e6bad44b6bddedf02fa3","subject":"lightos volume driver spec","status":"MERGED","created":"2022-01-11 14:36:30.000000000","updated":"2022-01-13 12:23:21.000000000","submitted":"2022-01-13 12:20:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":78,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"824191-nova-support-lightos-driver","meta_rev_id":"d545b9674814761e25d5caa2035fc31a41bf645e","_number":824191,"virtual_id_number":824191,"owner":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-13 12:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2022-01-13 10:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2022-01-13 08:47:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-13 11:55:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-13 11:55:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"}],"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-11 14:53:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-11 15:08:40.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-01-12 09:51:52.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2022-01-12 10:09:36.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2022-01-12 12:13:58.000000000","updated_by":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"reviewer":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"state":"CC"},{"updated":"2022-01-13 08:47:46.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"0a5c7d3c45e46cf35ce7fc855021d82083f73366","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-01-11 14:36:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36972e4625b9e19fe99518b0d0879571ac3fbef1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 14:53:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f9cf4b34bbf46679b2fcf49bd39fd32 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a792a6515a5488da3d7524093fd1dca : FAILURE in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"9192fff0089683c1848eca8926ae71664b2ac9d4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-01-11 15:08:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a19314fdb980263221acfa98f6f34a4689cf6194","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-01-11 15:26:45.000000000","message":"Patch Set 1:\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f91cb314a26a83b28239108b135b31f58e8aee0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-01-12 08:52:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8fd9d77560a69d25f8099eb260765b3efd41ee28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 09:18:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91de57722fcc41c7870825e500ed815a : FAILURE in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a4a78ec05e8452c82262b8fb1f940d8 : FAILURE in 6m 57s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/yoga/approved/lightos_volume_driver.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/yoga/approved/lightos_volume_driver.rst","accounts_in_message":[],"_revision_number":2},{"id":"3fe6b7dcef9958fb761dd03893adb539000e501d","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-01-12 09:51:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3e09948df3805df94326f9426c333fd98ce08dec","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-01-12 10:09:36.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"522358eb440bf097fb95baea7d6079f42e1fcce2","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-01-12 11:17:07.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2f0683182fd6ca8298a0f6974c3137878743366e","author":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"date":"2022-01-12 12:13:58.000000000","message":"Patch Set 2:\n\n(20 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f4dc8f3c3c57eefeb3666cc7e87fa5babd26e931","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-01-12 12:24:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"36941f61865c539074ce4df04fe5973aef3b8979","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-01-12 12:35:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"255ade5a52272a0ffb883b95ea19f22453f1abb9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-01-12 12:49:18.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"458865de6de79945b6817c5adbf72e09d1ad8cb2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-01-12 12:53:59.000000000","message":"Topic 824191 removed","accounts_in_message":[],"_revision_number":3},{"id":"431a426a66cd20155cdc318208f6b212e7c5f60b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-01-12 12:54:02.000000000","message":"Topic set to nova-support-lightos-driver","accounts_in_message":[],"_revision_number":3},{"id":"33b9380ecd2023683667f174d6f08f94ab3de782","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-01-12 12:54:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"06b6b85649568b59bd72bc8e89a6b4da4bf51d02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 12:57:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91681c8cf70341a5948f2451ce17300f : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb0458d999447f7b891d7cc8247d42c : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":3},{"id":"0d85379cbde7cef5f6cca180f29266f3c75f615e","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-01-12 13:13:34.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bbdb6865249c63e3dda2f62170bd502fcfb55679","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-01-12 14:00:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"afc76d03b934e751dd898b9b5221a8a45d795d1a","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-01-12 14:07:02.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"63f0b40788e2635617ac2957888763f511aebfe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-01-12 14:43:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a4ac7f763ec377fae7bf8884f9f9375be5483ecb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 15:07:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d33678c8a17747ffb528efe42c108760 : FAILURE in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b38dca86d5d47f3b844c61ea80455ba : SUCCESS in 5m 14s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/yoga/approved/lightos_volume_driver.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/yoga/approved/lightos_volume_driver.rst","accounts_in_message":[],"_revision_number":4},{"id":"715a0f9948e4621e8a4d6429f3d97149c3c8d4d9","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-01-12 15:41:02.000000000","message":"Patch Set 4: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5e1605137640fb002f428bb49748e76805cbd3fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-01-12 15:47:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"99b5c35ff76476fc6156df8ee5f431ef61563451","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 16:13:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b7e195d6ae4a3e8e7671a56262d43b : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b51b174ae9c4c79900416f13d61afd3 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":5},{"id":"0daaf57f61d52b71c97179c64cf36014775dcea5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-01-12 17:01:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"73f89476648c095b883fa5d2ee3da7ba6e1de6ef","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-01-12 17:03:07.000000000","message":"Patch Set 6:\n\n(8 comments)","accounts_in_message":[],"_revision_number":6},{"id":"325cb1306c26ebe471033bf724e3954a22544a74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 17:19:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ca33f5d5a1b4f44a2f6670dd1e655a2 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3b4a11439ca4efcb2d205a3c64e860e : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":6},{"id":"6f31696cb54ad46d5d092945369ed49bc83b2b49","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-01-12 17:59:15.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"28fb5645b0ee4a4c433b01937be1d297dd8a95aa","author":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"date":"2022-01-12 18:02:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7f62dbc49f937b0130cef8296f1ac0cd80cf66ca","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-01-13 08:47:46.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"28a8929ddca3074f8b7fc8e8fa5ff3e366d79727","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-01-13 10:46:17.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"abfa47f656bfb4c8112a9830d555ef6dc0ff52f1","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-01-13 11:55:57.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bf24a8af9057c8c9984492f60c55b5a9658caa16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 11:57:26.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1f06eb674420efb4b543588a675cc6d943423a2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 12:20:32.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48d7d5c7ceda450ebd0bcbfdec387001 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e270fcd6f0a34284b8487bedbcccdcc8 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":6},{"id":"9b5ec056674a7581617e5e9476b93e954aeb1cbc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 12:20:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"d545b9674814761e25d5caa2035fc31a41bf645e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 12:23:21.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/d5a05364012845599e0c30726e1fdf61 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f752ed139783416e2cefa53bda08d8100d685355","revisions":{"682c81f5d0b60ba9f8b17cca7f8dd00ad16b9c0a":{"kind":"REWORK","_number":1,"created":"2022-01-11 14:36:30.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/91/824191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/824191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/1"}}},"commit":{"parents":[{"commit":"2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82","subject":"libvirt: Allow Manila shares to be directly attached to instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-01-11 14:36:12.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-01-11 14:36:12.000000000","tz":120},"subject":"lightos volume driver spec","message":"lightos volume driver spec\n\nChange-Id: I463273d35050116933b4e6bad44b6bddedf02fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/682c81f5d0b60ba9f8b17cca7f8dd00ad16b9c0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/682c81f5d0b60ba9f8b17cca7f8dd00ad16b9c0a"}]},"branch":"refs/heads/master"},"9b6d768249a82053cefd5d4178fc2d513581d083":{"kind":"REWORK","_number":2,"created":"2022-01-12 08:52:21.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/91/824191/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/824191/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/2"}}},"commit":{"parents":[{"commit":"2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82","subject":"libvirt: Allow Manila shares to be directly attached to instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-01-11 14:36:12.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-01-12 08:52:08.000000000","tz":120},"subject":"lightos volume driver spec","message":"lightos volume driver spec\n\nChange-Id: I463273d35050116933b4e6bad44b6bddedf02fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9b6d768249a82053cefd5d4178fc2d513581d083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9b6d768249a82053cefd5d4178fc2d513581d083"}]},"branch":"refs/heads/master"},"b3ace207f2689627b313367b524b51d5c43df27f":{"kind":"REWORK","_number":3,"created":"2022-01-12 12:24:58.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/91/824191/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/824191/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/3"}}},"commit":{"parents":[{"commit":"2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82","subject":"libvirt: Allow Manila shares to be directly attached to instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-01-11 14:36:12.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-01-12 12:24:44.000000000","tz":120},"subject":"lightos volume driver spec","message":"lightos volume driver spec\n\nChange-Id: I463273d35050116933b4e6bad44b6bddedf02fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b3ace207f2689627b313367b524b51d5c43df27f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b3ace207f2689627b313367b524b51d5c43df27f"}]},"branch":"refs/heads/master"},"eac4f80c3868a6899583c072d73a27ccb55be608":{"kind":"REWORK","_number":4,"created":"2022-01-12 14:43:44.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/91/824191/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/824191/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/4"}}},"commit":{"parents":[{"commit":"2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82","subject":"libvirt: Allow Manila shares to be directly attached to instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-01-11 14:36:12.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-01-12 14:43:09.000000000","tz":120},"subject":"lightos volume driver spec","message":"lightos volume driver spec\n\nChange-Id: I463273d35050116933b4e6bad44b6bddedf02fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eac4f80c3868a6899583c072d73a27ccb55be608"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eac4f80c3868a6899583c072d73a27ccb55be608"}]},"branch":"refs/heads/master"},"8e839ef178c1f443ff2cbc069c5cb3e95b51ef04":{"kind":"REWORK","_number":5,"created":"2022-01-12 15:47:20.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/91/824191/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/824191/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/5"}}},"commit":{"parents":[{"commit":"2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82","subject":"libvirt: Allow Manila shares to be directly attached to instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-01-11 14:36:12.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-01-12 15:47:10.000000000","tz":120},"subject":"lightos volume driver spec","message":"lightos volume driver spec\n\nChange-Id: I463273d35050116933b4e6bad44b6bddedf02fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8e839ef178c1f443ff2cbc069c5cb3e95b51ef04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8e839ef178c1f443ff2cbc069c5cb3e95b51ef04"}]},"branch":"refs/heads/master"},"f752ed139783416e2cefa53bda08d8100d685355":{"kind":"REWORK","_number":6,"created":"2022-01-12 17:01:44.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/91/824191/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/824191/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/824191/6"}}},"commit":{"parents":[{"commit":"2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82","subject":"libvirt: Allow Manila shares to be directly attached to instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-01-11 14:36:12.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-01-12 17:00:36.000000000","tz":120},"subject":"lightos volume driver spec","message":"lightos volume driver spec\n\nChange-Id: I463273d35050116933b4e6bad44b6bddedf02fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f752ed139783416e2cefa53bda08d8100d685355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f752ed139783416e2cefa53bda08d8100d685355"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
