)]}'
{"id":"openstack%2Fnova-specs~831506","triplet_id":"openstack%2Fnova-specs~master~I190970b24942706370b325655175270a656ac733","project":"openstack/nova-specs","branch":"master","topic":"bp/unshelve-to-host","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-05-23 17:14:02.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-05-23 17:14:02.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-05-23 12:52:37.000000000","reason":"gibi replied on the change"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2022-05-23 15:57:25.000000000","reason":"Sylvain Bauza replied on the change"},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2022-05-23 17:14:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I190970b24942706370b325655175270a656ac733","subject":"Allow unshelve to a specific host","status":"MERGED","created":"2022-03-02 10:32:17.000000000","updated":"2022-05-23 17:16:27.000000000","submitted":"2022-05-23 17:14:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":118,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"831506-bp/unshelve-to-host","meta_rev_id":"7fb698afc99915fc241b8ee74fa0521d9ed5082a","_number":831506,"virtual_id_number":831506,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-23 17:14:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2022-05-23 16:45:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2022-05-23 16:46:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2022-05-23 16:46:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-02 10:43:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 10:45:49.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2022-03-08 13:08:01.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-04-06 15:53:12.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-04-20 16:39:57.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"a5e8d1f77a052de26328cef44e7d0867e6f9f1ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-03-02 10:32:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43ff8b61955d45cc0d31ab2ca6e558e2dcca4ced","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 10:43:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbced44922fb417badee9183f8ec5fe5 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c21fb6e21b34d72a9c2f3ccc81d9adb : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"350d4e6180d91ba887fad63825387ce8fb38fd1a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-03-08 13:08:01.000000000","message":"Patch Set 1: Code-Review+2\n\n(9 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a58bbff8316b4fb61c8a250807eaa7bef13662da","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-04-06 15:53:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ee86ea4037bdf1b30dd361826dce7ffdf630182","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-04-13 08:49:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"28881a375f354e2611f9a1298589cbbf9cc9c1be","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-04-13 11:13:06.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b2fa5f9011a82caf6c9deb6b43da2b9be1aa2bc","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-20 16:39:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"35c6a68ac7c3e264e058c3f4ee1dcc888ca87334","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-04-28 09:51:00.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f213ea35069a53c5214a45b0076c79570e0cdc02","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-04-28 10:00:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1fc10785ba72504f851c1b95de8c315a83c1f702","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-03 13:59:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"423461950d003732d198d38d2b259f040f4ef467","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-03 14:01:17.000000000","message":"Patch Set 2:\n\n(14 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a607c0f8ea880a14a4654b2e29ba5785abc46462","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 14:07:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9595ecc7875c4c45a7945806c81d7e89 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf44e1f076c54293aa9b7ce04dd52c48 : FAILURE in 2m 45s","accounts_in_message":[],"_revision_number":2},{"id":"d521b1cb056159578207893feecf109cd999bf53","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-03 15:21:43.000000000","message":"Patch Set 2: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1a6b95f429202c415a039af9e15e5735828852a8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-04 08:00:23.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9f7cbc32e5ba530ec6856925e9a7e4a630c7502f","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-05-04 10:45:49.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1d776baad12dfa2902f0a05b10d03b492963cc3e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-04 11:49:21.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"31573bcec57765c37a98436ecd8e7aabd512e1db","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-05-04 14:07:03.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dcc779b7f1c26e133546e3c77ccf38dcf97a1cec","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-04 14:59:17.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c24790cd1b3c7ee7b2ea2a4e062d983bbc499afc","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-05-04 15:03:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fee6fc1e425dae57b0eb7a89f4ac03be1cf3081e","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-05 09:54:00.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ebe9f6a83f0c49b2f8bbbdb9d4c1ecb2d891c997","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-05 10:23:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b110e10caef87a98ffd5d7f1bfb73b142449858a","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-06 12:14:25.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d3369b86b2ef69efdf13f8d9bb1ac168c05d7762","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-06 12:15:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"825c0817edb0e95864eebf170ac1fe90cb5e5831","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 12:31:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b49d176fe834c4e8b98bb9ef9a6ad5e : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4e0b83e264e4500880683922baf4f07 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":3},{"id":"4bc203361a8913aac8fe1198b80de1b35ec27339","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-05-10 08:32:32.000000000","message":"Patch Set 3: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5cc7f96a53f8b28d42126b5cb21102c8777611a8","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-05-10 09:02:59.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1526a83ac0e19096c9fbf46d65017c8927f77fd1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-10 09:16:46.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"676620b4309f74b53a4f9815a93a468c511f132c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-10 12:06:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"43457f306e51b3683ddee1c7b12e56ce19f02923","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-11 13:10:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"be0b70deb95761a98e169ce8ceddfe7d160def80","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-11 13:11:24.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"789effa2677a42b0095f61a31ac11c30a15e9dd6","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-11 13:12:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"59dd0a6995a393e31dd6850b2a3c03a65563a1f2","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-11 13:22:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e1e64781a20d5b4f4a73f507bdb38a2bec39d988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 13:25:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acec86a5c6b445d7a37b3e78162e130a : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93dc912fb11244189a137db80eaa7a4a : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":4},{"id":"ad979e68602a768e0c817152bc1fe33ab6c5788e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-11 14:17:01.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a49720883fe488e7cbe83539cfa8fcf77853464f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-11 14:20:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dd671948a47918f85cbe9864d42bce1932b5aa73","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-17 11:53:02.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"04957cddcbfe821a387f1a12d83f49e6656dc88f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-17 13:00:19.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e5102811ec4cd168520f2d5fe427f0cd72f5c4f2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-05-17 17:49:10.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fedb74f7acb1723b10e76ae3362783cbcfe4271e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-17 17:54:19.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"02c9ce0af24564ab82f4d86f22ba12aa60f54225","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-17 17:54:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd86c9215b415cbd96d19e1da9b6ef02bdd8e72b","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-18 10:04:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"06b921fb168b5f1e2eac92c5b97c1b5705905511","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-18 12:08:48.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4f974108ae8344915e3a1e09fe911995ff13081e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-05-18 13:54:11.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c45eb6550e164b785ae8106393d7c92611d57b8d","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-05-18 14:54:24.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7d59a1d14206c958ad82d974af5f2a2c00ad255c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-18 15:02:27.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"60be31f532175ae0fbd332ba16a9f2d935c26bc6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-18 17:48:37.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8ebaaa1e37ffd8224b41285ee1009ebb531e498f","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-19 07:59:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"42392e7848f797cf18a3a02cae1181bd6421744a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-19 10:22:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c0059f55641cd6912cfdad38e928363079e72efa","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-19 11:51:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a3813fe612fd89d9ffa1cb63a888b3e01b3279fa","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-05-19 14:04:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c03b18d771440eb21eeeec6ea1d62c687d1636ed","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-23 12:52:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4cf2cefd776d9b0698d4e3f6b75ec6773da1f03d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-23 14:30:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fdf6cadaa23533e14a897b19cd62caa26d6fb962","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 14:40:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b70ca8ef18e64701b04c99ba667e1e4f : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12cdae01be614ff48ce3c898b8d9da79 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":5},{"id":"67c86f24b2bf638ce419769a7aafe08ee34effd5","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-05-23 15:57:25.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1f95ae4ea767d6abd3f6508996a72a22149aa10c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-23 16:25:19.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8c059f89d61d0bf59691ac94f89f4421058d5359","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-05-23 16:44:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6de356ec046a33e2fa8bffa83cf8ace52c86839a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-05-23 16:45:14.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"048c951ef2035b7902668d96e178f067713a821b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-05-23 16:45:25.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ffa9fc395e5dc26908c3c0b782b552d336139521","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-05-23 16:46:15.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ebd72f3ff20565c099889a4d38a007009a6a9a6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 16:57:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72e2a2cda9dd40c7873ebc4be02f8a27 : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41d7a70356e24c79af80dffc07d2a321 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":6},{"id":"b6668b3b4345e35a3f906e25ab903b18ab390727","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 16:57:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"46a5ff20664a1fb480bae4d0e0146c8391127015","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 17:14:02.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80d2b01b5e3747cdbf5bff7f0bea9720 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef1f41c27b4b4a3ca3a62ce1a4de2b01 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":6},{"id":"2a737bee5607a25a89fcee740142de5f8ec0c39e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 17:14:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"7fb698afc99915fc241b8ee74fa0521d9ed5082a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 17:16:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/29ac5e885ba840ac99bfe8fc30f6755a : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"0fc8a027736a4a46e74f7478d9f00590cab69db4","revisions":{"9bef200971035ea78c04b63c0a00402a68a1ac6e":{"kind":"REWORK","_number":1,"created":"2022-03-02 10:32:17.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/06/831506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/831506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/1"}}},"commit":{"parents":[{"commit":"eeb8b92a0176ba727387db71e8cbf2ac55f97c89","subject":"Create specs directory for Zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eeb8b92a0176ba727387db71e8cbf2ac55f97c89"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-01-26 09:07:29.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-03-02 10:15:19.000000000","tz":60},"subject":"Allow unshelve to a specific host","message":"Allow unshelve to a specific host\n\nThis spec proposes allow administrators to unshelve to a specific host.\n\nAPIImpact\n\nBlueprint: unshelve-to-host\n\nChange-Id: I190970b24942706370b325655175270a656ac733\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9bef200971035ea78c04b63c0a00402a68a1ac6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9bef200971035ea78c04b63c0a00402a68a1ac6e"}]},"branch":"refs/heads/master"},"cc4543293e46fec091a6389375c977a8dc120718":{"kind":"REWORK","_number":2,"created":"2022-05-03 13:59:23.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/06/831506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/831506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/2"}}},"commit":{"parents":[{"commit":"d345f4dbd4e2e2e7bb3afb3f8b8dc6efb51101b1","subject":"Add the BP link to owner-nova-trait-usage spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d345f4dbd4e2e2e7bb3afb3f8b8dc6efb51101b1"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-01-26 09:07:29.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-05-03 13:59:08.000000000","tz":120},"subject":"Allow unshelve to a specific host","message":"Allow unshelve to a specific host\n\nThis spec proposes allow administrators to unshelve to a specific host.\n\nAPIImpact\n\nBlueprint: unshelve-to-host\n\nChange-Id: I190970b24942706370b325655175270a656ac733\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cc4543293e46fec091a6389375c977a8dc120718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cc4543293e46fec091a6389375c977a8dc120718"}]},"branch":"refs/heads/master"},"1c874d0a03f7c7918ade2ef99cf61238ae275edd":{"kind":"REWORK","_number":3,"created":"2022-05-06 12:15:11.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/06/831506/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/831506/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/3"}}},"commit":{"parents":[{"commit":"d345f4dbd4e2e2e7bb3afb3f8b8dc6efb51101b1","subject":"Add the BP link to owner-nova-trait-usage spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d345f4dbd4e2e2e7bb3afb3f8b8dc6efb51101b1"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-01-26 09:07:29.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-05-06 12:15:01.000000000","tz":120},"subject":"Allow unshelve to a specific host","message":"Allow unshelve to a specific host\n\nThis spec proposes allow administrators to unshelve to a specific host.\n\nAPIImpact\n\nBlueprint: unshelve-to-host\n\nChange-Id: I190970b24942706370b325655175270a656ac733\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1c874d0a03f7c7918ade2ef99cf61238ae275edd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1c874d0a03f7c7918ade2ef99cf61238ae275edd"}]},"branch":"refs/heads/master"},"1681ab949fb28c8da5051580005820e692861860":{"kind":"REWORK","_number":4,"created":"2022-05-11 13:10:48.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/06/831506/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/831506/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/4"}}},"commit":{"parents":[{"commit":"d345f4dbd4e2e2e7bb3afb3f8b8dc6efb51101b1","subject":"Add the BP link to owner-nova-trait-usage spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d345f4dbd4e2e2e7bb3afb3f8b8dc6efb51101b1"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-01-26 09:07:29.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-05-11 13:10:38.000000000","tz":120},"subject":"Allow unshelve to a specific host","message":"Allow unshelve to a specific host\n\nThis spec proposes allow administrators to unshelve to a specific host.\n\nAPIImpact\n\nBlueprint: unshelve-to-host\n\nChange-Id: I190970b24942706370b325655175270a656ac733\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1681ab949fb28c8da5051580005820e692861860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1681ab949fb28c8da5051580005820e692861860"}]},"branch":"refs/heads/master"},"ce983fd5adeecbe4f01893dce32f0cb5bd061f03":{"kind":"REWORK","_number":5,"created":"2022-05-23 14:30:31.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/06/831506/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/831506/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/5"}}},"commit":{"parents":[{"commit":"d345f4dbd4e2e2e7bb3afb3f8b8dc6efb51101b1","subject":"Add the BP link to owner-nova-trait-usage spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d345f4dbd4e2e2e7bb3afb3f8b8dc6efb51101b1"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-01-26 09:07:29.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-05-23 14:30:25.000000000","tz":120},"subject":"Allow unshelve to a specific host","message":"Allow unshelve to a specific host\n\nThis spec proposes allow administrators to unshelve to a specific host.\n\nAPIImpact\n\nBlueprint: unshelve-to-host\n\nChange-Id: I190970b24942706370b325655175270a656ac733\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ce983fd5adeecbe4f01893dce32f0cb5bd061f03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ce983fd5adeecbe4f01893dce32f0cb5bd061f03"}]},"branch":"refs/heads/master"},"0fc8a027736a4a46e74f7478d9f00590cab69db4":{"kind":"REWORK","_number":6,"created":"2022-05-23 16:45:14.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/06/831506/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/831506/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/831506/6"}}},"commit":{"parents":[{"commit":"d345f4dbd4e2e2e7bb3afb3f8b8dc6efb51101b1","subject":"Add the BP link to owner-nova-trait-usage spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d345f4dbd4e2e2e7bb3afb3f8b8dc6efb51101b1"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-01-26 09:07:29.000000000","tz":60},"committer":{"name":"Dan Smith","email":"dms@danplanet.com","date":"2022-05-23 16:45:14.000000000","tz":0},"subject":"Allow unshelve to a specific host","message":"Allow unshelve to a specific host\n\nThis spec proposes allow administrators to unshelve to a specific host.\n\nAPIImpact\n\nBlueprint: unshelve-to-host\n\nChange-Id: I190970b24942706370b325655175270a656ac733\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0fc8a027736a4a46e74f7478d9f00590cab69db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0fc8a027736a4a46e74f7478d9f00590cab69db4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
