)]}'
{"id":"openstack%2Fnova-specs~833669","triplet_id":"openstack%2Fnova-specs~master~I12b00d732123986e9794d68588e4224ccdb715f3","project":"openstack/nova-specs","branch":"master","topic":"bp/manila_shares_attachments","attention_set":{},"removed_from_attention_set":{"33728":{"account":{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"},"last_update":"2022-07-05 17:10:58.000000000","reason":"Change was submitted"},"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2022-05-03 14:46:27.000000000","reason":"gibi replied on the change"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-07-05 17:10:58.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-07-05 16:55:40.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-07-05 17:10:58.000000000","reason":"Change was submitted"},"6796":{"account":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"last_update":"2022-07-05 17:10:58.000000000","reason":"Change was submitted"},"6413":{"account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"last_update":"2022-07-05 17:10:58.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2022-07-05 17:10:58.000000000","reason":"Change was submitted"},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2022-07-05 17:10:58.000000000","reason":"Change was submitted"},"32079":{"account":{"_account_id":32079,"name":"Carlos Augusto da Silva Martins","email":"carlos.martins@ifsuldeminas.edu.br","username":"casmssm"},"last_update":"2022-07-05 17:10:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I12b00d732123986e9794d68588e4224ccdb715f3","subject":"libvirt: Allow Manila shares to be directly attached to instances","status":"MERGED","created":"2022-03-14 17:12:41.000000000","updated":"2022-07-05 17:13:11.000000000","submitted":"2022-07-05 17:10:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":196,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"833669-bp/manila_shares_attachments","meta_rev_id":"434d9849e31323b64d2e98c0f3bdb51c3fba5563","_number":833669,"virtual_id_number":833669,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-05 17:10:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2022-07-05 16:55:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2022-07-05 12:12:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-05 08:01:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2022-07-05 16:55:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":32079,"name":"Carlos Augusto da Silva Martins","email":"carlos.martins@ifsuldeminas.edu.br","username":"casmssm"}],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32079,"name":"Carlos Augusto da Silva Martins","email":"carlos.martins@ifsuldeminas.edu.br","username":"casmssm"},{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-14 17:26:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-05 10:49:14.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-05-10 16:13:07.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-06-01 10:55:25.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"},"state":"REVIEWER"},{"updated":"2022-06-01 10:55:25.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":32079,"name":"Carlos Augusto da Silva Martins","email":"carlos.martins@ifsuldeminas.edu.br","username":"casmssm"},"state":"REVIEWER"},{"updated":"2022-06-01 10:55:25.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2022-06-01 10:55:25.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2022-06-01 10:55:25.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-06-02 15:54:57.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2022-06-17 15:00:52.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"}],"messages":[{"id":"f4d866e8cca04e21e9a6f9c265008d9d65cf2ff5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-03-14 17:12:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3602ce79bf0396691cf3b1fd0b75c9a4666c888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 17:26:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/212cd97c05314c6da564823a6df104f4 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0603d23c04004c7fb703dbc6006216a1 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c3fadabac480915b5352cb3c5d72e427aa7a37bb","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-04-05 10:49:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7d35b0521927eb55d646161834422be430f89a10","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-04-05 12:51:11.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"41077bb4a244e30554c17d4db974b187404b5b3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-04-06 13:33:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"87785b1bd11696c41d86abeeaccc232e89a7957a","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-04-06 13:35:34.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ab1ec83f7ce71f6421c3cc6630f8128193893523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 13:48:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b06d87a0a3146f6822c02492f13fe5e : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddbb32dafff647cab121ced6ca759d59 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"bf5917f760b7d30764680c04eff4c9a6c50d0ff2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-04-07 09:52:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e54bf85958c837b31cb4d5b516fcdc9f539250d","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-04-07 16:13:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"35ffdb46e7e03ddc1d212869b46972abb2ae6dc3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-04-21 08:26:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"bb53955b054dfd13ad9e7c202c0103f90ad5892f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-04-22 10:02:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dfe265584f26b7f85d94294417e9412075598392","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-02 16:53:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"32cbabc80347641e5b4678d4be7417f89d8b5569","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-02 16:54:05.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f7106813745bde2e95f9f83539eb2ec32f7c2eae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 17:18:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ea6203d4a44f54988c9d1bacf4275b : FAILURE in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7df26a3198404188a106186238967f8d : FAILURE in 3m 29s","accounts_in_message":[],"_revision_number":3},{"id":"13c79d73a389874bf47560aa532fdab4b5f4be73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2022-05-03 14:13:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a24cb64e71a1a8423cc430ec856d82d46a2435f8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2022-05-03 14:13:30.000000000","message":"Topic add_graphviz removed","accounts_in_message":[],"_revision_number":4},{"id":"7dcfd4d7b0002774c09b870d2344f882ef2a0b68","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2022-05-03 14:13:41.000000000","message":"Topic set to bp/libvirt-virtiofs-attach-manila-shares","accounts_in_message":[],"_revision_number":4},{"id":"aecee7af59e7d4dd3462b42cff6103174d3a177a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 14:28:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d84497c7c6664f1eadcaea03fd58f145 : FAILURE in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcb661bac1134aad98740a038d4cc556 : FAILURE in 3m 12s","accounts_in_message":[],"_revision_number":4},{"id":"744e64738432d4a22feb557c7bab4ab235babf5d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-03 14:40:50.000000000","message":"Patch Set 4: Code-Review-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":4},{"id":"383d652860b7f576ae2d9f72c261b496a3a5a089","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-03 14:44:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1786dd3bc17913b5d178d0a085a98f83b027a598","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-03 14:46:27.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"794b7a6619ac8cc3d640aaae098c49f6d6a30479","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-03 14:47:30.000000000","message":"Topic bp/libvirt-virtiofs-attach-manila-shares removed","accounts_in_message":[],"_revision_number":5},{"id":"b4e8706c06d0d9a4d85ca3deb2650c64f412550c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-03 14:47:36.000000000","message":"Topic set to bp/manila_shares_attachments","accounts_in_message":[],"_revision_number":5},{"id":"6217f72dbbae69acc4a14ce6f8e1aba890a87df7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 15:01:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98634050e04f45bba7a744b0f4e44dc6 : FAILURE in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffb45ec71b5942c9a811818257b7cb82 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":5},{"id":"0bfbcfd57df9cb1beab525157bcd15bae1fd4b13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-03 16:40:45.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"a067c8b16c29dfa1aca105bf6c07eb30c2414470","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 16:54:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47589e2e74b74403aaecd0535f3a2e8f : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26c53995ef954ad8bb1c49f8d00a55f4 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":6},{"id":"2237a81d84b8b450d7e13b93f89e82f285c3c2e8","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-03 16:55:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"39b769ddd8627e5cb4f788be00b09cdd93eb422d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-04 08:23:51.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"23daf0568bbb5dcd67c509ee9219ef6407536af0","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-06 17:21:29.000000000","message":"Patch Set 6:\n\n(15 comments)","accounts_in_message":[],"_revision_number":6},{"id":"34a222de42086d06689a3bf0dcfc003b1506ed42","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-09 10:33:29.000000000","message":"Patch Set 6:\n\n(11 comments)","accounts_in_message":[],"_revision_number":6},{"id":"42a6c0bfa4168b616d1ed8b97e7b8ed7baf9a4c8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-10 16:13:07.000000000","message":"Patch Set 6: Code-Review-1\n\n(23 comments)","accounts_in_message":[],"_revision_number":6},{"id":"724cf7842bde0e19475eb4d8facf6bb2005981e7","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-10 17:49:00.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8cff17b30d106366673b9afb45002935d30b603b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-11 09:43:12.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"bf1cc19d5e5d64fe728faf88c4758e3cc1d6ced9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-11 11:46:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9caa200a6f4e31c1313efc40c30a60b6827e4dbf","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-18 16:09:51.000000000","message":"Patch Set 6:\n\n(25 comments)","accounts_in_message":[],"_revision_number":6},{"id":"653058f08532ccae2bf2661c8b71018ac25b5454","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-18 16:11:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a3aeb0a75602fa79dcb64cc8d20f1d67528288f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 16:28:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2693caa92c294af793e1fbf96936b148 : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e14bc9b8137e419d8342384487686f35 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":7},{"id":"04fdc315a2ef4666d0021a2b2cd39acb02b75e7d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-19 08:48:12.000000000","message":"Patch Set 7: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e8a5863fe831e8f342fd98e8464280481d8e2ad3","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-23 12:53:02.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b379733d6ad467cfba968d9ae851818b3c8b9c0d","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-23 13:31:59.000000000","message":"Patch Set 7:\n\n(8 comments)","accounts_in_message":[],"_revision_number":7},{"id":"5dfe225f609df87f2afad6ca7285932b8c0e76bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-23 13:32:53.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"86418e12e9bd2e32fa5c7e5c56da07a7b9b00764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 13:43:32.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07d73e3b0abe488687f8719e8a0535cc : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8509dd0cf581431f8b8c3402206251bf : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":8},{"id":"825939ca1cbfb19f2397c54b4ab70d933ee13abb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-23 14:07:48.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8be7fe91fe9de18aae167a0514af2e612f2dbc50","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-23 14:08:21.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"61b5b6fb6e2af00d585b182b15e66b6c633f51ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-23 17:27:14.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"de24ff93c7b3760a930fa8fedbbaba1a4d6b514d","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-05-23 17:28:06.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4a6fd39338defeb9fe5cc184a446f9ad9fa7ed00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 17:40:23.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cdf8ea9b1d44f0f82bf46229a0121d7 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36745719da834c4e9b4cd14448da8edc : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":9},{"id":"995028d7f8178250238d2948a2736b7b8b582289","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-06-01 09:32:57.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"87e000ff78ea1892a244cc2f2a09b4d309ac8fe5","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-06-01 10:58:04.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"540a15928e8b69258fbfef1fba11c519d13f9d76","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-06-02 09:59:39.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6db24656f15f070a65d65b2fb07b910a3fbc35da","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-06-02 15:54:57.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"64d08f19652aa880715f5f710b9f35462b557dab","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-06-02 17:30:42.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"02cf26345551f00fdc9e69ff9c64a497cd41f169","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-06-08 13:19:24.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"30fbd49c47ac2d32a4123b3770f6c8badb38ca74","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-21 17:51:34.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0d8388ca5d4de26c85868bde10ac6500b0a546b0","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-06-22 09:18:31.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6ea37d4a3dda9a49b39ee0d35d6e5d7c2332a5eb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-06-22 12:23:57.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9115575567cce51a4c3a3539f8f74c574c2882f0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-06-22 12:25:14.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"72470bc72654099dc8926de11f182240bcbc2e71","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-06-23 13:59:06.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"6d479a017bf1977b3862a5496b472ff843130c21","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-06-24 14:03:29.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0c8c680b7181fb6b4cdcff9a0c13e69d816a7feb","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-06-30 16:06:46.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4736cee385d03ba990b758b710ae5f0df3d9a6f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-07-04 09:29:04.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"17ae966731dd74f931af453ee71305dce0b62938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 09:38:30.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc06be079a7341e7b0b5546c1b3075e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d09f20a954c14078a9a7e4fffaffd509 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d062de812694511b0bd9867854a43d9 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":10},{"id":"2bfeeeac7283ee1f11b5522b65c2d3ad0da4e6da","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-07-05 08:01:23.000000000","message":"Patch Set 10: Code-Review+2\n\n(7 comments)","accounts_in_message":[],"_revision_number":10},{"id":"fde5442519794d32b93ad2bb3a86bab280a7afa3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-07-05 12:12:47.000000000","message":"Patch Set 10: Code-Review+2\n\n(11 comments)","accounts_in_message":[],"_revision_number":10},{"id":"efb747f3a5cb48d2806a74c6562bf734cefa6815","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-05 15:55:02.000000000","message":"Patch Set 10: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":10},{"id":"03668430f6a44a0cf06f0478dc314d5549b5f977","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-05 16:03:35.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0f92975c7e0ee7d70ffbc50d988082de87370269","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-07-05 16:48:14.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"abf7f25a45de6edb0e30f8c1d9201f89efa2da9f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-05 16:55:40.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"64acca69ebba6da177ff4f143951214ba333a58b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 16:56:11.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"20ea6e9bd0571af1420462c849ef2c731a966b98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 17:10:58.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f3ace0e2fc74f528f7a3643946fc2e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb2c75d3b8348afbea1dc7c96f60046 : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23984abebcef46c2b770ae37e2954697 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":10},{"id":"b01471cbd5f4e3e974d827dd733adc20d43ce1b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 17:10:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"434d9849e31323b64d2e98c0f3bdb51c3fba5563","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 17:13:11.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9939150d57b44dd9aa1f8e2c6b7d756\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/0e10391b71bc45ec9d804d15f5279ebc : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"58a313367c52761f13f714637c34787a121bfcc2","revisions":{"122f6c0fc7bb93f146ac2313aa4309ea693404e3":{"kind":"REWORK","_number":1,"created":"2022-03-14 17:12:41.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/69/833669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/833669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/1"}}},"commit":{"parents":[{"commit":"eeb8b92a0176ba727387db71e8cbf2ac55f97c89","subject":"Create specs directory for Zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eeb8b92a0176ba727387db71e8cbf2ac55f97c89"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-03-14 17:08:45.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-03-14 17:08:45.000000000","tz":60},"subject":"libvirt: Allow Manila shares to be directly attached to instances","message":"libvirt: Allow Manila shares to be directly attached to instances\n\nManila is the OpenStack Shared Filesystems service. This spec will\noutline API, database, compute and libvirt driver changes required in\nNova to allow the shares provided by Manila to be associated with and\nattached to instances.\n\nAddendum, change object add share_proto field.\n\nblueprint: libvirt-virtiofs-attach-manila-shares\nChange-Id: I12b00d732123986e9794d68588e4224ccdb715f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/122f6c0fc7bb93f146ac2313aa4309ea693404e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/122f6c0fc7bb93f146ac2313aa4309ea693404e3"}]},"branch":"refs/heads/master"},"c3f1a90051fe1c12be3fcbe285551d10e626e02f":{"kind":"REWORK","_number":2,"created":"2022-04-06 13:33:39.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/69/833669/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/833669/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/2"}}},"commit":{"parents":[{"commit":"1a20b736348a8ad68a922338ce3bcdfede85684e","subject":"Move implemented specs for the Yoga release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1a20b736348a8ad68a922338ce3bcdfede85684e"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-03-14 17:08:45.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-04-06 13:33:20.000000000","tz":120},"subject":"libvirt: Allow Manila shares to be directly attached to instances","message":"libvirt: Allow Manila shares to be directly attached to instances\n\nManila is the OpenStack Shared Filesystems service. This spec will\noutline API, database, compute and libvirt driver changes required in\nNova to allow the shares provided by Manila to be associated with and\nattached to instances.\n\nAddendum, change object add share_proto field.\n\nblueprint: libvirt-virtiofs-attach-manila-shares\nChange-Id: I12b00d732123986e9794d68588e4224ccdb715f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c3f1a90051fe1c12be3fcbe285551d10e626e02f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c3f1a90051fe1c12be3fcbe285551d10e626e02f"}]},"branch":"refs/heads/master"},"7f992d5facd058cbade1480bcc3d15fea95d70f0":{"kind":"REWORK","_number":3,"created":"2022-05-02 16:53:06.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/69/833669/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/833669/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/3"}}},"commit":{"parents":[{"commit":"1a20b736348a8ad68a922338ce3bcdfede85684e","subject":"Move implemented specs for the Yoga release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1a20b736348a8ad68a922338ce3bcdfede85684e"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-03-14 17:08:45.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-05-02 16:52:49.000000000","tz":120},"subject":"libvirt: Allow Manila shares to be directly attached to instances","message":"libvirt: Allow Manila shares to be directly attached to instances\n\nManila is the OpenStack Shared Filesystems service. This spec will\noutline API, database, compute and libvirt driver changes required in\nNova to allow the shares provided by Manila to be associated with and\nattached to instances.\n\nAddendum, change object add share_proto field.\n\nblueprint: libvirt-virtiofs-attach-manila-shares\nChange-Id: I12b00d732123986e9794d68588e4224ccdb715f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7f992d5facd058cbade1480bcc3d15fea95d70f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7f992d5facd058cbade1480bcc3d15fea95d70f0"}]},"branch":"refs/heads/master"},"dd11bac6c04b5beaefedd2b18eb836436688a28e":{"kind":"REWORK","_number":4,"created":"2022-05-03 14:13:11.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/69/833669/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/833669/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/4"}}},"commit":{"parents":[{"commit":"e524b76807bedaec1ecd512ed83e958f02d08574","subject":"Allow to use graphviz into specifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e524b76807bedaec1ecd512ed83e958f02d08574"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-03-14 17:08:45.000000000","tz":60},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2022-05-03 14:12:59.000000000","tz":-240},"subject":"libvirt: Allow Manila shares to be directly attached to instances","message":"libvirt: Allow Manila shares to be directly attached to instances\n\nManila is the OpenStack Shared Filesystems service. This spec will\noutline API, database, compute and libvirt driver changes required in\nNova to allow the shares provided by Manila to be associated with and\nattached to instances.\n\nAddendum, change object add share_proto field.\n\nblueprint: libvirt-virtiofs-attach-manila-shares\nChange-Id: I12b00d732123986e9794d68588e4224ccdb715f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dd11bac6c04b5beaefedd2b18eb836436688a28e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dd11bac6c04b5beaefedd2b18eb836436688a28e"}]},"branch":"refs/heads/master"},"82610d9cc3acf138dd1a480038997760dee26ee8":{"kind":"REWORK","_number":5,"created":"2022-05-03 14:44:34.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/69/833669/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/833669/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/5"}}},"commit":{"parents":[{"commit":"e524b76807bedaec1ecd512ed83e958f02d08574","subject":"Allow to use graphviz into specifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e524b76807bedaec1ecd512ed83e958f02d08574"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-03-14 17:08:45.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-05-03 14:44:13.000000000","tz":120},"subject":"libvirt: Allow Manila shares to be directly attached to instances","message":"libvirt: Allow Manila shares to be directly attached to instances\n\nManila is the OpenStack Shared Filesystems service. This spec will\noutline API, database, compute and libvirt driver changes required in\nNova to allow the shares provided by Manila to be associated with and\nattached to instances.\n\nAddendum, change object add share_proto field.\n\nblueprint: libvirt-virtiofs-attach-manila-shares\nChange-Id: I12b00d732123986e9794d68588e4224ccdb715f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/82610d9cc3acf138dd1a480038997760dee26ee8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/82610d9cc3acf138dd1a480038997760dee26ee8"}]},"branch":"refs/heads/master"},"0e762f6d553b00ba0af0361c62de1e97b414bfc7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-05-03 16:40:45.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/69/833669/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/833669/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/6"}}},"commit":{"parents":[{"commit":"ac17d63e69b9be89a081a66d988ee3c96e33319f","subject":"Allow to use graphviz into specifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ac17d63e69b9be89a081a66d988ee3c96e33319f"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-03-14 17:08:45.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-05-03 16:22:49.000000000","tz":120},"subject":"libvirt: Allow Manila shares to be directly attached to instances","message":"libvirt: Allow Manila shares to be directly attached to instances\n\nManila is the OpenStack Shared Filesystems service. This spec will\noutline API, database, compute and libvirt driver changes required in\nNova to allow the shares provided by Manila to be associated with and\nattached to instances.\n\nAddendum, change object add share_proto field.\n\nblueprint: libvirt-virtiofs-attach-manila-shares\nChange-Id: I12b00d732123986e9794d68588e4224ccdb715f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0e762f6d553b00ba0af0361c62de1e97b414bfc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0e762f6d553b00ba0af0361c62de1e97b414bfc7"}]},"branch":"refs/heads/master"},"4ea6c9d22b9d9dccac5d3984b5236f41a45bb49c":{"kind":"REWORK","_number":7,"created":"2022-05-18 16:11:17.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/69/833669/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/833669/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/7"}}},"commit":{"parents":[{"commit":"ac17d63e69b9be89a081a66d988ee3c96e33319f","subject":"Allow to use graphviz into specifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ac17d63e69b9be89a081a66d988ee3c96e33319f"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-03-14 17:08:45.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-05-18 16:10:46.000000000","tz":120},"subject":"libvirt: Allow Manila shares to be directly attached to instances","message":"libvirt: Allow Manila shares to be directly attached to instances\n\nManila is the OpenStack Shared Filesystems service. This spec will\noutline API, database, compute and libvirt driver changes required in\nNova to allow the shares provided by Manila to be associated with and\nattached to instances.\n\nAddendum, change object add share_proto field.\n\nblueprint: libvirt-virtiofs-attach-manila-shares\nChange-Id: I12b00d732123986e9794d68588e4224ccdb715f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4ea6c9d22b9d9dccac5d3984b5236f41a45bb49c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4ea6c9d22b9d9dccac5d3984b5236f41a45bb49c"}]},"branch":"refs/heads/master"},"03686bf748f7c681823d2c05402058f7e98f623d":{"kind":"REWORK","_number":8,"created":"2022-05-23 13:32:53.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/69/833669/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/833669/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/8"}}},"commit":{"parents":[{"commit":"ac17d63e69b9be89a081a66d988ee3c96e33319f","subject":"Allow to use graphviz into specifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ac17d63e69b9be89a081a66d988ee3c96e33319f"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-03-14 17:08:45.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-05-23 13:32:35.000000000","tz":120},"subject":"libvirt: Allow Manila shares to be directly attached to instances","message":"libvirt: Allow Manila shares to be directly attached to instances\n\nManila is the OpenStack Shared Filesystems service. This spec will\noutline API, database, compute and libvirt driver changes required in\nNova to allow the shares provided by Manila to be associated with and\nattached to instances.\n\nAddendum, change object add share_proto field.\n\nblueprint: libvirt-virtiofs-attach-manila-shares\nChange-Id: I12b00d732123986e9794d68588e4224ccdb715f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/03686bf748f7c681823d2c05402058f7e98f623d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/03686bf748f7c681823d2c05402058f7e98f623d"}]},"branch":"refs/heads/master"},"8f8ea7265a785706d6199cec70a3c5b10080294b":{"kind":"REWORK","_number":9,"created":"2022-05-23 17:27:14.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/69/833669/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/833669/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/9"}}},"commit":{"parents":[{"commit":"d345f4dbd4e2e2e7bb3afb3f8b8dc6efb51101b1","subject":"Add the BP link to owner-nova-trait-usage spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d345f4dbd4e2e2e7bb3afb3f8b8dc6efb51101b1"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-03-14 17:08:45.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-05-23 17:27:08.000000000","tz":120},"subject":"libvirt: Allow Manila shares to be directly attached to instances","message":"libvirt: Allow Manila shares to be directly attached to instances\n\nManila is the OpenStack Shared Filesystems service. This spec will\noutline API, database, compute and libvirt driver changes required in\nNova to allow the shares provided by Manila to be associated with and\nattached to instances.\n\nAddendum, change object add share_proto field.\n\nblueprint: libvirt-virtiofs-attach-manila-shares\nChange-Id: I12b00d732123986e9794d68588e4224ccdb715f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8f8ea7265a785706d6199cec70a3c5b10080294b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8f8ea7265a785706d6199cec70a3c5b10080294b"}]},"branch":"refs/heads/master"},"58a313367c52761f13f714637c34787a121bfcc2":{"kind":"REWORK","_number":10,"created":"2022-07-04 09:29:04.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/69/833669/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/833669/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/833669/10"}}},"commit":{"parents":[{"commit":"d345f4dbd4e2e2e7bb3afb3f8b8dc6efb51101b1","subject":"Add the BP link to owner-nova-trait-usage spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d345f4dbd4e2e2e7bb3afb3f8b8dc6efb51101b1"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-03-14 17:08:45.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-07-04 09:28:54.000000000","tz":120},"subject":"libvirt: Allow Manila shares to be directly attached to instances","message":"libvirt: Allow Manila shares to be directly attached to instances\n\nManila is the OpenStack Shared Filesystems service. This spec will\noutline API, database, compute and libvirt driver changes required in\nNova to allow the shares provided by Manila to be associated with and\nattached to instances.\n\nAddendum, change object add share_proto field.\n\nblueprint: libvirt-virtiofs-attach-manila-shares\nChange-Id: I12b00d732123986e9794d68588e4224ccdb715f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/58a313367c52761f13f714637c34787a121bfcc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/58a313367c52761f13f714637c34787a121bfcc2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
