)]}'
{"id":"openstack%2Fnova-specs~850048","triplet_id":"openstack%2Fnova-specs~master~Ib7950d805db2891f270f02aa9b595aacdd219aee","project":"openstack/nova-specs","branch":"master","topic":"bp/configurable-instance-domains","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2022-07-19 13:46:00.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-07-19 13:46:00.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-07-19 13:32:22.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}},"hashtags":[],"change_id":"Ib7950d805db2891f270f02aa9b595aacdd219aee","subject":"Revert \"Configurable instance domains\"","status":"MERGED","created":"2022-07-19 12:14:42.000000000","updated":"2022-07-19 13:47:26.000000000","submitted":"2022-07-19 13:46:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"revert_of":849765,"submission_id":"850048-bp/configurable-instance-domains","meta_rev_id":"72d7582c86428740c710def443fab60d76d56eb1","_number":850048,"virtual_id_number":850048,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-19 13:46:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2022-07-19 13:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2022-07-19 13:32:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2022-07-19 13:32:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-19 12:14:42.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2022-07-19 12:14:42.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2022-07-19 12:14:42.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2022-07-19 12:14:42.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"CC"},{"updated":"2022-07-19 12:14:42.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2022-07-19 12:14:42.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-07-19 12:14:42.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-19 13:10:26.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"eea8de3a453977e3437effc527565073638a631c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-19 12:14:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6aa42385d50771518f025b8ecd6eb7971477a4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 12:26:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7fc840ee8ad4308adc1b44333b30fb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2238662715244e47b891deb57ffc3221 : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92d8d6c2bb8a4bb99ef285a78b39b477 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"aa869d7d862f8094fca7b9e68020b39ac8d11bc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-19 12:32:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fc7ae02e6a926befc6acf37ff6f6465635b0ed4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 12:46:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/147aa85174dc4ef38be9bbaf19825427\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bcdd62d96d743d28fcbed32c3d9f633 : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d326253bbd84c83853349af79c62a9d : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"3804eaa2e0cc5c02066cc45329b694b396800736","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-07-19 13:10:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7ee59c47be451172aabce4825e54a711a8edfc3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-07-19 13:32:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0057cf94242a3c0ef3c69a01d7729416720920f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 13:32:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"62ddcdc0f85537253fe98c2c98c6cf578c440b4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 13:46:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a087d23f8d514d65a96af1fe8c189c12\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d0f34611f214524b93c2d9d21e8fdcd : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b67abdb085564ad985b610e69cb344bf : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":2},{"id":"b4128a8df2a3f2409e71a43cfa607db873104f61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 13:46:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"72d7582c86428740c710def443fab60d76d56eb1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 13:47:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9c48eacb872456ca70fc7ed06c6dfbe\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/78e44e0291cb46aea383eb06d762121c : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aaad3b4e26fd12e474e0645d1054200ba0277f0d","revisions":{"e86f2982c007285129b3e55b6a1d19ee2c0eab12":{"kind":"REWORK","_number":1,"created":"2022-07-19 12:14:42.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/48/850048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/48/850048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/48/850048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/48/850048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/48/850048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/48/850048/1"}}},"commit":{"parents":[{"commit":"35aaec78108fabd1961274250a8fa460abed6f90","subject":"Configurable instance domains","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/35aaec78108fabd1961274250a8fa460abed6f90"}]}],"author":{"name":"sean mooney","email":"smooney@redhat.com","date":"2022-07-19 12:14:42.000000000","tz":0},"committer":{"name":"sean mooney","email":"smooney@redhat.com","date":"2022-07-19 12:14:42.000000000","tz":0},"subject":"Revert \"Configurable instance domains\"","message":"Revert \"Configurable instance domains\"\n\nThis reverts commit 35aaec78108fabd1961274250a8fa460abed6f90.\n\nReason for revert: after this was merged there emerged that other cores had several concerns with the direction of this spec. both in terms of is this in scope of nova and if it is is this the correct way to support passing fqdn/dns_domain info to the gust vs the other abandoned approch.\n\nas such we should follow the reverts-for-retrospective-vetos policy\n\nhttps://docs.openstack.org/nova/latest/contributor/policies.html#reverts-for-retrospective-vetos\n\nand revert this to give time for those issues to be discussed.\n\ni will also bring this up in the team meeting and if we think\nthe concenrs are adressable and agree with the direction of this spec we can grant a spec freeze excption otherwise we will defer this to then next release\n\nChange-Id: Ib7950d805db2891f270f02aa9b595aacdd219aee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e86f2982c007285129b3e55b6a1d19ee2c0eab12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e86f2982c007285129b3e55b6a1d19ee2c0eab12"}]},"branch":"refs/heads/master"},"aaad3b4e26fd12e474e0645d1054200ba0277f0d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-19 12:32:09.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/48/850048/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/48/850048/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/48/850048/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/48/850048/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/48/850048/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/48/850048/2"}}},"commit":{"parents":[{"commit":"35aaec78108fabd1961274250a8fa460abed6f90","subject":"Configurable instance domains","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/35aaec78108fabd1961274250a8fa460abed6f90"}]}],"author":{"name":"sean mooney","email":"smooney@redhat.com","date":"2022-07-19 12:14:42.000000000","tz":0},"committer":{"name":"sean mooney","email":"smooney@redhat.com","date":"2022-07-19 12:32:09.000000000","tz":0},"subject":"Revert \"Configurable instance domains\"","message":"Revert \"Configurable instance domains\"\n\nThis reverts commit 35aaec78108fabd1961274250a8fa460abed6f90.\n\nReason for revert: after this was merged there emerged that\nother cores had several concerns with the direction of this spec,\nBoth in terms of is this in the scope of nova and\nif it is, is this the correct way to support passing fqdn/dns_domain\ninfo to the gust vs the other abandoned approach.\n\nAs such we should follow the reverts-for-retrospective-vetos policy\nhttps://docs.openstack.org/nova/latest/contributor/policies.html#reverts-for-retrospective-vetos\nand revert this to give time for those issues to be discussed.\n\nChange-Id: Ib7950d805db2891f270f02aa9b595aacdd219aee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aaad3b4e26fd12e474e0645d1054200ba0277f0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aaad3b4e26fd12e474e0645d1054200ba0277f0d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
