)]}'
{"id":"openstack%2Fnova-specs~853837","triplet_id":"openstack%2Fnova-specs~master~I72fa3f605cfcf7c3dd0ff4c791be7df8f19f058b","project":"openstack/nova-specs","branch":"master","topic":"bp/stable-compute-uuids","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2022-11-15 13:23:20.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-11-15 13:14:04.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-11-15 13:23:20.000000000","reason":"Change was submitted"},"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2022-11-15 13:23:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I72fa3f605cfcf7c3dd0ff4c791be7df8f19f058b","subject":"Robustify Compute Node Hostnames backlog spec","status":"MERGED","created":"2022-08-19 18:58:21.000000000","updated":"2022-11-15 13:25:05.000000000","submitted":"2022-11-15 13:23:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"853837-bp/stable-compute-uuids","meta_rev_id":"6d179ac90d7c37dc79de95441463517ef5e2db8b","_number":853837,"virtual_id_number":853837,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-15 13:23:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2022-11-15 13:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-15 10:49:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2022-11-15 13:14:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-15 10:49:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-19 18:59:24.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2022-08-19 19:10:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-21 17:42:45.000000000","updated_by":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"reviewer":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"state":"CC"},{"updated":"2022-08-22 12:18:22.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-08-23 14:14:00.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"CC"},{"updated":"2022-10-19 17:06:45.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2022-10-20 14:36:15.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2022-10-31 08:59:34.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"CC"}],"messages":[{"id":"e7aa13dc676fd8be24ffec251ab3c2436026c04a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-08-19 18:58:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a2f60de83fec0248638fef832933d47caf68014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 19:10:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2962a6109324432fb85ec41724245d25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8323c89ed42d43bca5b13deb8f97489b : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1dc35383d3c428d875866d76675488c : FAILURE in 6m 16s","accounts_in_message":[],"_revision_number":1},{"id":"728aa8375a808fd38ac4bf176c93aba085b5868c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-08-19 19:30:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e3cb714cf45d0d8f4452717d9f99db4a32fa68c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 19:38:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30f7e5bf596d444abea15f09563c0f12\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4efb10a14ad41d19dffe089e44baf8b : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7a85fc5aa834ee78262f18044f4ec77 : FAILURE in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"12472f39cb84a7b44dfb10e7de74e1d1a40b6bd8","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-10-20 14:36:15.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"5fcb3e5c6d508967e6970299779c0b71340ef602","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2022-10-31 08:59:34.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"21f015f4bd5ea79c3916d3ea20f8bd8f0fb8f74e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-31 17:33:31.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d6b919c682d0ede6b006e730feba9b3e1a0fadc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-11-01 15:56:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dbddcc0afaa068062a0b1db20b0835d6334e0ba0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-11-01 15:57:46.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"66445b092ae1c847a8d381c8894982b3b770edb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 16:19:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/187b026f940c42bebc33ef9507ec488e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e28ecc3d1ad442b5a28628a66f518762 : SUCCESS in 15m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ff1a52cc0e74004a7c51bce92ad682c : FAILURE in 2m 46s","accounts_in_message":[],"_revision_number":4},{"id":"f2143c7c2f494c14f2511ec356b6d8686421919d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-11-01 17:09:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a6cc0f6fbad95d94b32ce20df299312b42e7e141","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 17:15:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5198b8ae201a46018d12e91a2a277ec1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18115c0d98e9425491f02bc4eca801df : FAILURE in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e3d41921fc4f36b9d2b6d61145c539 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":5},{"id":"74f95b9a501e6c20b94007b86643ae3c47cc68b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-11-01 18:07:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"acb0fd75fbbbaa94d13c61048b9359a832a0bd6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 18:22:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58df1321ce9548bd83fe807050e8c5c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/197ce8ca9e2d42e690af8e6dd8192c5a : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48ef1e2bd60a42ff9102d74550a07b7a : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":6},{"id":"a4fad41c72f33a3bd77f59f782806dc88cf57501","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-11-15 10:49:48.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"faa81539ef2654c3ff0b3eea542fc1c6aa571ebb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-15 13:14:04.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5d14a3c53ca9cfd59321740203a7c219e6894b31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 13:14:25.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0b948a85be174d31b4c1c9d60786e7606af75340","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 13:23:20.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf7851e9fbd945239eb7194a0bde3c65\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/752c42e291de4617b703c54ac32b21ae : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/916528900dea42a68b469e2e023d1e4f : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":6},{"id":"780deba251ed1ff65f26962cd395e24d47d55ed6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 13:23:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"6d179ac90d7c37dc79de95441463517ef5e2db8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 13:25:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13cf97f15a21481ba4513cd11219dba0\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/7db3f4ac507c4ddd8a9d121bc1846451 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3256eda85f6de908702482f2ef65ad15c58ea867","revisions":{"0de6b748f8ea0e7ec82cbdba8b2b2923e20b240f":{"kind":"REWORK","_number":1,"created":"2022-08-19 18:58:21.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/37/853837/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/37/853837/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/1"}}},"commit":{"parents":[{"commit":"aaad3b4e26fd12e474e0645d1054200ba0277f0d","subject":"Revert \"Configurable instance domains\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aaad3b4e26fd12e474e0645d1054200ba0277f0d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-08-19 18:55:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-08-19 18:58:15.000000000","tz":-420},"subject":"WIP: Robustify Compute Node Hostnames","message":"WIP: Robustify Compute Node Hostnames\n\nThis is mostly a brain-dump spec on the topic of compute hosts and\nhow fragile they are in terms of hostname handling. There has long\nbeen a requirement that computes can NEVER change hostnames, and we\nhave few tools to even detect the problem before we corrupt the\ndatabase if it happens.\n\nHere I have documented some of the things we could do to make that\nmore robust, should we choose to do so. This is based on a recent\nnear-catastrophe and thus reflects things that would have avoided\npain in a real scenario.\n\nChange-Id: I72fa3f605cfcf7c3dd0ff4c791be7df8f19f058b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0de6b748f8ea0e7ec82cbdba8b2b2923e20b240f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0de6b748f8ea0e7ec82cbdba8b2b2923e20b240f"}]},"branch":"refs/heads/master"},"82affbfe5973236f8c1de417259a838a1f9bea56":{"kind":"REWORK","_number":2,"created":"2022-08-19 19:30:15.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/37/853837/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/37/853837/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/2"}}},"commit":{"parents":[{"commit":"aaad3b4e26fd12e474e0645d1054200ba0277f0d","subject":"Revert \"Configurable instance domains\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aaad3b4e26fd12e474e0645d1054200ba0277f0d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-08-19 18:55:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-08-19 19:30:05.000000000","tz":-420},"subject":"WIP: Robustify Compute Node Hostnames","message":"WIP: Robustify Compute Node Hostnames\n\nThis is mostly a brain-dump spec on the topic of compute hosts and\nhow fragile they are in terms of hostname handling. There has long\nbeen a requirement that computes can NEVER change hostnames, and we\nhave few tools to even detect the problem before we corrupt the\ndatabase if it happens.\n\nHere I have documented some of the things we could do to make that\nmore robust, should we choose to do so. This is based on a recent\nnear-catastrophe and thus reflects things that would have avoided\npain in a real scenario.\n\nChange-Id: I72fa3f605cfcf7c3dd0ff4c791be7df8f19f058b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/82affbfe5973236f8c1de417259a838a1f9bea56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/82affbfe5973236f8c1de417259a838a1f9bea56"}]},"branch":"refs/heads/master"},"55051aaca95973f3313ec5a1147d580195b95911":{"kind":"REWORK","_number":3,"created":"2022-11-01 15:56:43.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/37/853837/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/37/853837/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/3"}}},"commit":{"parents":[{"commit":"7ad4e88db2648b070db8505f2e47f6991a1b3c8b","subject":"Merge \"Re-propose PCI Device Tracking In Placement for A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7ad4e88db2648b070db8505f2e47f6991a1b3c8b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-08-19 18:55:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-11-01 15:01:59.000000000","tz":-420},"subject":"WIP: Robustify Compute Node Hostnames","message":"WIP: Robustify Compute Node Hostnames\n\nThis is mostly a brain-dump spec on the topic of compute hosts and\nhow fragile they are in terms of hostname handling. There has long\nbeen a requirement that computes can NEVER change hostnames, and we\nhave few tools to even detect the problem before we corrupt the\ndatabase if it happens.\n\nHere I have documented some of the things we could do to make that\nmore robust, should we choose to do so. This is based on a recent\nnear-catastrophe and thus reflects things that would have avoided\npain in a real scenario.\n\nChange-Id: I72fa3f605cfcf7c3dd0ff4c791be7df8f19f058b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/55051aaca95973f3313ec5a1147d580195b95911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/55051aaca95973f3313ec5a1147d580195b95911"}]},"branch":"refs/heads/master"},"b7851c3c4b4301ac32a438c3eb91c7613451a061":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-11-01 15:57:46.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/37/853837/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/37/853837/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/4"}}},"commit":{"parents":[{"commit":"7ad4e88db2648b070db8505f2e47f6991a1b3c8b","subject":"Merge \"Re-propose PCI Device Tracking In Placement for A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7ad4e88db2648b070db8505f2e47f6991a1b3c8b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-08-19 18:55:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-11-01 15:56:51.000000000","tz":-420},"subject":"Robustify Compute Node Hostnames backlog spec","message":"Robustify Compute Node Hostnames backlog spec\n\nThis is mostly a brain-dump spec on the topic of compute hosts and\nhow fragile they are in terms of hostname handling. There has long\nbeen a requirement that computes can NEVER change hostnames, and we\nhave few tools to even detect the problem before we corrupt the\ndatabase if it happens.\n\nHere I have documented some of the things we could do to make that\nmore robust, should we choose to do so. This is based on a recent\nnear-catastrophe and thus reflects things that would have avoided\npain in a real scenario.\n\nPer discussion at the PTG, I am adding this as a backlog spec, to be\nan overarching guide for multiple smaller specs to provide more\ndetailed progress towards the goals described here.\n\nChange-Id: I72fa3f605cfcf7c3dd0ff4c791be7df8f19f058b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b7851c3c4b4301ac32a438c3eb91c7613451a061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b7851c3c4b4301ac32a438c3eb91c7613451a061"}]},"branch":"refs/heads/master"},"eff6a7ef32780f9415b02f0ed30215d1c174d76e":{"kind":"REWORK","_number":5,"created":"2022-11-01 17:09:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/37/853837/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/37/853837/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/5"}}},"commit":{"parents":[{"commit":"7ad4e88db2648b070db8505f2e47f6991a1b3c8b","subject":"Merge \"Re-propose PCI Device Tracking In Placement for A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7ad4e88db2648b070db8505f2e47f6991a1b3c8b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-08-19 18:55:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-11-01 17:08:52.000000000","tz":-420},"subject":"Robustify Compute Node Hostnames backlog spec","message":"Robustify Compute Node Hostnames backlog spec\n\nThis is mostly a brain-dump spec on the topic of compute hosts and\nhow fragile they are in terms of hostname handling. There has long\nbeen a requirement that computes can NEVER change hostnames, and we\nhave few tools to even detect the problem before we corrupt the\ndatabase if it happens.\n\nHere I have documented some of the things we could do to make that\nmore robust, should we choose to do so. This is based on a recent\nnear-catastrophe and thus reflects things that would have avoided\npain in a real scenario.\n\nPer discussion at the PTG, I am adding this as a backlog spec, to be\nan overarching guide for multiple smaller specs to provide more\ndetailed progress towards the goals described here.\n\nChange-Id: I72fa3f605cfcf7c3dd0ff4c791be7df8f19f058b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eff6a7ef32780f9415b02f0ed30215d1c174d76e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eff6a7ef32780f9415b02f0ed30215d1c174d76e"}]},"branch":"refs/heads/master"},"3256eda85f6de908702482f2ef65ad15c58ea867":{"kind":"REWORK","_number":6,"created":"2022-11-01 18:07:35.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/37/853837/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/37/853837/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/37/853837/6"}}},"commit":{"parents":[{"commit":"7ad4e88db2648b070db8505f2e47f6991a1b3c8b","subject":"Merge \"Re-propose PCI Device Tracking In Placement for A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7ad4e88db2648b070db8505f2e47f6991a1b3c8b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-08-19 18:55:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-11-01 18:07:27.000000000","tz":-420},"subject":"Robustify Compute Node Hostnames backlog spec","message":"Robustify Compute Node Hostnames backlog spec\n\nThis is mostly a brain-dump spec on the topic of compute hosts and\nhow fragile they are in terms of hostname handling. There has long\nbeen a requirement that computes can NEVER change hostnames, and we\nhave few tools to even detect the problem before we corrupt the\ndatabase if it happens.\n\nHere I have documented some of the things we could do to make that\nmore robust, should we choose to do so. This is based on a recent\nnear-catastrophe and thus reflects things that would have avoided\npain in a real scenario.\n\nPer discussion at the PTG, I am adding this as a backlog spec, to be\nan overarching guide for multiple smaller specs to provide more\ndetailed progress towards the goals described here.\n\nChange-Id: I72fa3f605cfcf7c3dd0ff4c791be7df8f19f058b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3256eda85f6de908702482f2ef65ad15c58ea867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3256eda85f6de908702482f2ef65ad15c58ea867"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}}]}],"submit_requirements":[]}
