)]}'
{"id":"openstack%2Fnova-specs~855514","triplet_id":"openstack%2Fnova-specs~master~I9adda039edbc5cf8cd14aa1af9aa96a1d700ad90","project":"openstack/nova-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-05-03 20:05:26.000000000","reason":"Change was abandoned"},"35239":{"account":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"last_update":"2023-05-03 20:05:26.000000000","reason":"Change was abandoned"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-11-15 09:43:03.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}},"hashtags":[],"change_id":"I9adda039edbc5cf8cd14aa1af9aa96a1d700ad90","subject":"Improve multiqueue network adapter support","status":"ABANDONED","created":"2022-09-01 13:54:47.000000000","updated":"2023-05-03 20:05:26.000000000","total_comment_count":25,"unresolved_comment_count":16,"has_review_started":true,"meta_rev_id":"ac93f0cb45fdf1f8e9e029304a5d1f33f3ea1bf9","_number":855514,"virtual_id_number":855514,"owner":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-02 10:25:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2022-09-02 11:08:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":-1,"date":"2022-11-15 09:43:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-01 14:09:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-02 11:08:13.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-09-02 11:28:22.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2022-11-15 09:43:03.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"6ddcb1c497861f535a474a7a739bc92c930719ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2022-09-01 13:54:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b335142d82225996ada2faf691737d14ea6d88b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 14:09:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/07bb5e778eea4be08bb14898a6c7f91b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d55534bb7bbd42498255af57d9c59fb8 : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647c74022371493f82efa6c4f50faa24 : FAILURE in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"469ea8c6762a8300ed8f49c96bceeac878077f3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2022-09-02 08:20:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48b1c5e674d5939a01926b40a0780d081f17e769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 08:26:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffce0418a86149a3ba76c40c02f6dbc6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbfd966cafd349a1b6f5fa35504d9eac : FAILURE in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55cd95bed2a84c9f8501e69dbac2168d : FAILURE in 3m 35s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/2023.1/approved/improve-nova-multiqueue-adapter-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/2023.1/approved/improve-nova-multiqueue-adapter-support.rst","accounts_in_message":[],"_revision_number":2},{"id":"b47335027e3f2a942246606753ed80318608a495","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2022-09-02 08:29:27.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a2a22991af20ca0f703d90190b4db0bfcffaed31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2022-09-02 08:32:28.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"8df04da4c8a34870fca04bd3fd2485ce24142113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 08:40:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbd82adbcdb24e40b558ff0c7f15d5b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/299b0a3d578d4ed3a94b2d05ac773b51 : FAILURE in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ea31103cf0d46e68acda1fd5cebe7e5 : FAILURE in 3m 47s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/2023.1/approved/improve-nova-multiqueue-adapter-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/2023.1/approved/improve-nova-multiqueue-adapter-support.rst","accounts_in_message":[],"_revision_number":4},{"id":"7e91dd2855c2ca359f5fcfd97b1c7fdf33aa019e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2022-09-02 08:52:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"608e84e06edfd12d010464dab03e3ab6220f8996","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 09:02:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2347a69655d41fa9e657e509b0edc24\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80d46a48817a4df6b98be84c35e27d9f : FAILURE in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3659f2f3f7548888ccc4abf16d4c9a5 : SUCCESS in 3m 52s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/2023.1/approved/improve-nova-multiqueue-adapter-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/2023.1/approved/improve-nova-multiqueue-adapter-support.rst","accounts_in_message":[],"_revision_number":5},{"id":"54dd9e82b770bc0e80b379fe6bd6701bc53164ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2022-09-02 10:14:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2da6102336ef62545266f4c35326fbf2f3e1bfb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 10:25:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be1d8566f00141fea00a4bc35752629a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e38f9c5cb69c438a83c2b89cfaf3716e : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9d776f51cc04f1e91eb22bc0402a2fc : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":6},{"id":"01571ce17afb58f76a4cdc5ee9351d380c67964d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-09-02 11:08:13.000000000","message":"Patch Set 6: Code-Review-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4c09f240ea635ec30c6529c8803ad32c677f494b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-09-02 11:11:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"82fa3e8dc018dae791b7966af2723c5234291b04","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-09-02 11:28:22.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8f81cfd5dc9db2e208e058f5c60eb95af25dcd1b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-09-02 11:50:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5dcb237ebbd70a4166048d9177281cf382fa91bd","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2022-09-02 12:23:57.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d9d132503035b5f4fa8117fe301bab3a23ebc339","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-11-15 09:43:03.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ac93f0cb45fdf1f8e9e029304a5d1f33f3ea1bf9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2023-05-03 20:05:26.000000000","message":"Abandoned\n\nSplit into three separate blueprints","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"909b5c405d9f1c43a7ca0f3156e14e0239379727","revisions":{"d593857d5c90f6616e139d588d6d0d961deaba51":{"kind":"REWORK","_number":1,"created":"2022-09-01 13:54:47.000000000","uploader":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"ref":"refs/changes/14/855514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/855514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/1"}}},"commit":{"parents":[{"commit":"aaad3b4e26fd12e474e0645d1054200ba0277f0d","subject":"Revert \"Configurable instance domains\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aaad3b4e26fd12e474e0645d1054200ba0277f0d"}]}],"author":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-09-01 08:17:41.000000000","tz":120},"committer":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-09-01 13:54:30.000000000","tz":120},"subject":"Improve multiqueue network adapter support","message":"Improve multiqueue network adapter support\n\nAt the moment, OpenStack has very limited support for multiqueue network adapters. Current implementation of multiqueue vhost(user) netdev defaults to the number of queues equal to the number of vCPUs. Such operations model does not allign well with the hardware-offloaded netdev usage model, where number of queues available in the underlying hardware is fixed. Ideally, OpenStack should allow the operator to start a VM with arbitrary number of vCPUs that is larger (or equal) to the number of queues configured on vNICs.\n\nChange-Id: I9adda039edbc5cf8cd14aa1af9aa96a1d700ad90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d593857d5c90f6616e139d588d6d0d961deaba51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d593857d5c90f6616e139d588d6d0d961deaba51"}]},"branch":"refs/heads/master"},"c1b49e687d4f148620a311f7d6b1e64427bafaf2":{"kind":"REWORK","_number":2,"created":"2022-09-02 08:20:37.000000000","uploader":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"ref":"refs/changes/14/855514/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/855514/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/2"}}},"commit":{"parents":[{"commit":"aaad3b4e26fd12e474e0645d1054200ba0277f0d","subject":"Revert \"Configurable instance domains\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aaad3b4e26fd12e474e0645d1054200ba0277f0d"}]}],"author":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-09-01 08:17:41.000000000","tz":120},"committer":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-09-02 08:17:52.000000000","tz":120},"subject":"Improve multiqueue network adapter support","message":"Improve multiqueue network adapter support\n\nAt the moment, OpenStack has very limited support for multiqueue network adapters. Current implementation of multiqueue vhost(user) netdev defaults to the number of queues equal to the number of vCPUs. Such operations model does not allign well with the hardware-offloaded netdev usage model, where number of queues available in the underlying hardware is fixed. Ideally, OpenStack should allow the operator to start a VM with arbitrary number of vCPUs that is larger (or equal) to the number of queues configured on vNICs.\n\nChange-Id: I9adda039edbc5cf8cd14aa1af9aa96a1d700ad90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c1b49e687d4f148620a311f7d6b1e64427bafaf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c1b49e687d4f148620a311f7d6b1e64427bafaf2"}]},"branch":"refs/heads/master"},"fc0f917fd23a7c251f80c9381f29e24ba5cdbcde":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-09-02 08:29:27.000000000","uploader":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"ref":"refs/changes/14/855514/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/855514/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/3"}}},"commit":{"parents":[{"commit":"0d0000f0939e8220b9d580d46e19fc8f022dea51","subject":"Create specs directory for 2023.1 Antelope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d0000f0939e8220b9d580d46e19fc8f022dea51"}]}],"author":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-09-01 08:17:41.000000000","tz":120},"committer":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-09-02 08:29:27.000000000","tz":0},"subject":"Improve multiqueue network adapter support","message":"Improve multiqueue network adapter support\n\nAt the moment, OpenStack has very limited support for multiqueue network adapters. Current implementation of multiqueue vhost(user) netdev defaults to the number of queues equal to the number of vCPUs. Such operations model does not allign well with the hardware-offloaded netdev usage model, where number of queues available in the underlying hardware is fixed. Ideally, OpenStack should allow the operator to start a VM with arbitrary number of vCPUs that is larger (or equal) to the number of queues configured on vNICs.\n\nChange-Id: I9adda039edbc5cf8cd14aa1af9aa96a1d700ad90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fc0f917fd23a7c251f80c9381f29e24ba5cdbcde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fc0f917fd23a7c251f80c9381f29e24ba5cdbcde"}]},"branch":"refs/heads/master","description":"Rebase"},"4dd68b8fe81d46fc00773e29b0282deb635d95f8":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-09-02 08:32:28.000000000","uploader":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"ref":"refs/changes/14/855514/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/855514/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/4"}}},"commit":{"parents":[{"commit":"0d0000f0939e8220b9d580d46e19fc8f022dea51","subject":"Create specs directory for 2023.1 Antelope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d0000f0939e8220b9d580d46e19fc8f022dea51"}]}],"author":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-09-01 08:17:41.000000000","tz":120},"committer":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-09-02 08:31:14.000000000","tz":120},"subject":"Improve multiqueue network adapter support","message":"Improve multiqueue network adapter support\n\nAt the moment, OpenStack has very limited support for multiqueue network\nadapters. Current implementation of multiqueue vhost(user) netdev\ndefaults to the number of queues equal to the number of vCPUs.\nSuch operations model does not allign well with the hardware-offloaded\nnetdev usage model, where number of queues available in the underlying\nhardware is fixed. Ideally, OpenStack should allow the operator to\nstart a VM with arbitrary number of vCPUs that is larger (or equal)\nto the number of queues configured on vNICs.\n\nChange-Id: I9adda039edbc5cf8cd14aa1af9aa96a1d700ad90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4dd68b8fe81d46fc00773e29b0282deb635d95f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4dd68b8fe81d46fc00773e29b0282deb635d95f8"}]},"branch":"refs/heads/master"},"a0ddeed2e282c321649e46116bd89dbf8e391f09":{"kind":"REWORK","_number":5,"created":"2022-09-02 08:52:16.000000000","uploader":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"ref":"refs/changes/14/855514/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/855514/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/5"}}},"commit":{"parents":[{"commit":"0d0000f0939e8220b9d580d46e19fc8f022dea51","subject":"Create specs directory for 2023.1 Antelope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d0000f0939e8220b9d580d46e19fc8f022dea51"}]}],"author":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-09-01 08:17:41.000000000","tz":120},"committer":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-09-02 08:52:05.000000000","tz":120},"subject":"Improve multiqueue network adapter support","message":"Improve multiqueue network adapter support\n\nAt the moment, OpenStack has very limited support for multiqueue network\nadapters. Current implementation of multiqueue vhost(user) netdev\ndefaults to the number of queues equal to the number of vCPUs.\nSuch operations model does not allign well with the hardware-offloaded\nnetdev usage model, where number of queues available in the underlying\nhardware is fixed. Ideally, OpenStack should allow the operator to\nstart a VM with arbitrary number of vCPUs that is larger (or equal)\nto the number of queues configured on vNICs.\n\nChange-Id: I9adda039edbc5cf8cd14aa1af9aa96a1d700ad90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a0ddeed2e282c321649e46116bd89dbf8e391f09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a0ddeed2e282c321649e46116bd89dbf8e391f09"}]},"branch":"refs/heads/master"},"909b5c405d9f1c43a7ca0f3156e14e0239379727":{"kind":"REWORK","_number":6,"created":"2022-09-02 10:14:36.000000000","uploader":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"ref":"refs/changes/14/855514/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/855514/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/855514/6"}}},"commit":{"parents":[{"commit":"0d0000f0939e8220b9d580d46e19fc8f022dea51","subject":"Create specs directory for 2023.1 Antelope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d0000f0939e8220b9d580d46e19fc8f022dea51"}]}],"author":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-09-01 08:17:41.000000000","tz":120},"committer":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-09-02 10:14:23.000000000","tz":120},"subject":"Improve multiqueue network adapter support","message":"Improve multiqueue network adapter support\n\nAt the moment, OpenStack has very limited support for multiqueue network\nadapters. Current implementation of multiqueue vhost(user) netdev\ndefaults to the number of queues equal to the number of vCPUs.\nSuch operations model does not allign well with the hardware-offloaded\nnetdev usage model, where number of queues available in the underlying\nhardware is fixed. Ideally, OpenStack should allow the operator to\nstart a VM with arbitrary number of vCPUs that is larger (or equal)\nto the number of queues configured on vNICs.\n\nChange-Id: I9adda039edbc5cf8cd14aa1af9aa96a1d700ad90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/909b5c405d9f1c43a7ca0f3156e14e0239379727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/909b5c405d9f1c43a7ca0f3156e14e0239379727"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
