)]}'
{"id":"openstack%2Fnova-specs~85673","triplet_id":"openstack%2Fnova-specs~master~Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246","project":"openstack/nova-specs","branch":"master","topic":"bp/metadata-service-network-info","hashtags":[],"change_id":"Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246","subject":"API: Proxy neutron configuration to guest instance","status":"MERGED","created":"2014-04-28 20:01:23.000000000","updated":"2014-12-18 21:43:27.000000000","submitted":"2014-12-18 21:43:25.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":84,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1a72b8603b33d1404f8a62d9815981bae5555244","_number":85673,"virtual_id_number":85673,"owner":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2014-12-18 21:43:25.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2014-12-17 23:12:19.000000000","_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},{"value":0,"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},{"value":0,"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":1082,"name":"Scott Moser","email":"smoser@ubuntu.com","username":"smoser"},{"value":0,"_account_id":11547,"name":"Andy Tylee","email":"andy.tylee@metaswitch.com"},{"value":0,"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},{"value":0,"_account_id":4573,"name":"yunhong-jiang","email":"yunhong.jiang@intel.com","username":"yunhong-jiang"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"recommended":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":1,"date":"2014-12-17 22:42:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},{"value":2,"date":"2014-12-18 21:40:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},{"value":0,"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},{"value":0,"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":2,"date":"2014-12-17 23:06:17.000000000","_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":1082,"name":"Scott Moser","email":"smoser@ubuntu.com","username":"smoser"},{"value":1,"date":"2014-12-15 23:52:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11547,"name":"Andy Tylee","email":"andy.tylee@metaswitch.com"},{"value":0,"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},{"value":0,"_account_id":4573,"name":"yunhong-jiang","email":"yunhong.jiang@intel.com","username":"yunhong-jiang"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},{"value":1,"date":"2014-12-18 21:40:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},{"value":0,"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},{"value":0,"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":1082,"name":"Scott Moser","email":"smoser@ubuntu.com","username":"smoser"},{"value":0,"_account_id":11547,"name":"Andy Tylee","email":"andy.tylee@metaswitch.com"},{"value":0,"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},{"value":0,"_account_id":4573,"name":"yunhong-jiang","email":"yunhong.jiang@intel.com","username":"yunhong-jiang"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},{"value":0,"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},{"value":0,"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":1082,"name":"Scott Moser","email":"smoser@ubuntu.com","username":"smoser"},{"value":0,"_account_id":11547,"name":"Andy Tylee","email":"andy.tylee@metaswitch.com"},{"value":0,"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},{"value":0,"_account_id":4573,"name":"yunhong-jiang","email":"yunhong.jiang@intel.com","username":"yunhong-jiang"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},{"_account_id":1082,"name":"Scott Moser","email":"smoser@ubuntu.com","username":"smoser"},{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4573,"name":"yunhong-jiang","email":"yunhong.jiang@intel.com","username":"yunhong-jiang"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},{"_account_id":11547,"name":"Andy Tylee","email":"andy.tylee@metaswitch.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-04-28 20:01:23.000000000","updated_by":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"reviewer":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"state":"REVIEWER"},{"updated":"2014-06-04 12:12:39.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2014-06-11 04:40:18.000000000","updated_by":{"_account_id":4573,"name":"yunhong-jiang","email":"yunhong.jiang@intel.com","username":"yunhong-jiang"},"reviewer":{"_account_id":4573,"name":"yunhong-jiang","email":"yunhong.jiang@intel.com","username":"yunhong-jiang"},"state":"REVIEWER"},{"updated":"2014-06-27 13:02:20.000000000","updated_by":{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},"reviewer":{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},"state":"REVIEWER"},{"updated":"2014-07-07 23:28:22.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2014-07-08 14:05:57.000000000","updated_by":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"reviewer":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"state":"REVIEWER"},{"updated":"2014-07-11 18:14:56.000000000","updated_by":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"reviewer":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"state":"REVIEWER"},{"updated":"2014-07-14 00:10:08.000000000","updated_by":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"reviewer":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"state":"REVIEWER"},{"updated":"2014-10-01 20:05:49.000000000","updated_by":{"_account_id":1082,"name":"Scott Moser","email":"smoser@ubuntu.com","username":"smoser"},"reviewer":{"_account_id":1082,"name":"Scott Moser","email":"smoser@ubuntu.com","username":"smoser"},"state":"REVIEWER"},{"updated":"2014-10-21 17:37:49.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2014-10-29 14:45:02.000000000","updated_by":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"reviewer":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"state":"REVIEWER"},{"updated":"2014-11-04 13:59:34.000000000","updated_by":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"reviewer":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"state":"REVIEWER"},{"updated":"2014-12-12 13:38:11.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2014-12-14 12:59:03.000000000","updated_by":{"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},"reviewer":{"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},"state":"REVIEWER"},{"updated":"2014-12-15 23:52:21.000000000","updated_by":{"_account_id":11547,"name":"Andy Tylee","email":"andy.tylee@metaswitch.com"},"reviewer":{"_account_id":11547,"name":"Andy Tylee","email":"andy.tylee@metaswitch.com"},"state":"REVIEWER"},{"updated":"2014-12-17 22:42:24.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2014-12-17 23:06:17.000000000","updated_by":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"reviewer":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"state":"REVIEWER"},{"updated":"2014-12-17 23:12:19.000000000","updated_by":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"reviewer":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"state":"REVIEWER"},{"updated":"2014-12-18 21:40:14.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2014-12-18 21:43:25.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"cc274d3719269a64ff7b27d58a5738080b30a37e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"9fe8526f28e9177ba1559184436c8e43060068aa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/1/check/gate-nova-specs-python27/be44f96 : SUCCESS in 38s\n- gate-nova-specs-docs http://logs.openstack.org/73/85673/1/check/gate-nova-specs-docs/f80982b : FAILURE in 27s\n\n","accounts_in_message":[],"_revision_number":1},{"id":"7bebad8d8e10ef33246b1cfddea58edaeb4dac97","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 2:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"e7947c960cc1e06c2ff0c31386cc6b098f79af54","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/2/check/gate-nova-specs-python27/def03a6 : SUCCESS in 40s\n- gate-nova-specs-docs http://logs.openstack.org/73/85673/2/check/gate-nova-specs-docs/9401b3b : FAILURE in 40s\n\n","accounts_in_message":[],"_revision_number":2},{"id":"ed1cd9310a884e8b3fd6bcb0948615d5f6bbff9e","author":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"date":"2014-04-28 20:01:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"927ec50eee476fd5eea3a4abd8b47758e5bf5080","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/3/check/gate-nova-specs-python27/299df09 : SUCCESS in 40s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/3/check/gate-nova-specs-docs/3a88cd1/doc/build/html/ : SUCCESS in 57s\n\n","accounts_in_message":[],"_revision_number":3},{"id":"a0558ea5fac91d4d4d989d306926c681c8309b26","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 3:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"a1cffe4584f928f7c68acb47306ea04a2c876d16","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 3: I would prefer that you didn\u0027t merge this\n\n(11 inline comments)\n\n","accounts_in_message":[],"_revision_number":3},{"id":"bb4f9f7274bc3661b5ba3a0ed1fd49036e45b0ac","author":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 3: (11 inline comments)\n\n","accounts_in_message":[],"_revision_number":3},{"id":"bf241d8ce8fd60fdec64a8192419942f4fb30c64","author":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"date":"2014-04-28 20:01:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8479f83aa136dc148bdf77eec425bc78ea74d7db","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 4:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"8107acf823dd5ce45cc5ec0a9eb2daeec7a7ba26","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 4: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/4/check/gate-nova-specs-python27/f9f30c9 : SUCCESS in 47s\n- gate-nova-specs-docs http://logs.openstack.org/73/85673/4/check/gate-nova-specs-docs/478c719 : FAILURE in 1m 04s\n","accounts_in_message":[],"_revision_number":4},{"id":"63c1a8bb0710d8472823382e4478b54a4a5761c0","author":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"date":"2014-04-28 20:01:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e7295b61ae7851fea550ec8b784fdc6ecc15e9e2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 5: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/5/check/gate-nova-specs-python27/7d985b9 : SUCCESS in 1m 03s\n- gate-nova-specs-docs http://logs.openstack.org/73/85673/5/check/gate-nova-specs-docs/e7bbdde : FAILURE in 1m 01s\n","accounts_in_message":[],"_revision_number":5},{"id":"260163b3d1ee733162e45017d5e8752fc994f4e9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 5:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":5},{"id":"bbf1e556bd773fa6a6509064e340455dffaebc71","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 5:\n\nNew patchset patch-id matches previous patchset, but whitespace content has changed.","accounts_in_message":[],"_revision_number":5},{"id":"b2c74f0d1bba9557cf94dc6412bb6e4a43bd95c7","author":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"date":"2014-04-28 20:01:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5e7964fbe62dda0b63e7b289bec29632bbd089c0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 6: Works for me\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/6/check/gate-nova-specs-python27/38405ce : SUCCESS in 27s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/6/check/gate-nova-specs-docs/b68bb4a/doc/build/html/ : SUCCESS in 48s\n","accounts_in_message":[],"_revision_number":6},{"id":"f450d9cd50d6b605fdd7544f15e470846086fed9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 6: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":6},{"id":"bb91aacf30cd8e0266331f80862aa071b2246374","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 6:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":6},{"id":"1fd29e3d1c6395f655b537ba69a2318e62371893","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 6:\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":6},{"id":"9ca632b18cc15f145a581fd7c6ab030b4da920f5","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 6: Abandoned\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":6},{"id":"ffc8f4b3201daf2b691eb06e190fb349fbe880d2","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 6: I would prefer that you didn\u0027t merge this\n\n(5 inline comments)\n\nLooking close, just a few details that need sorting out a little bit I think.","accounts_in_message":[],"_revision_number":6},{"id":"fda917cb4daed4d06a5aeecd52194a6d0b2107dd","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 6:\n\nNew patchset patch-id matches previous patchset, but whitespace content has changed.","accounts_in_message":[],"_revision_number":6},{"id":"be3ae578feea363901aa23655af115ea33e78d9e","author":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"date":"2014-04-28 20:01:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c2585d089e08ac5ef70c8e30c947d9a7bf3b2898","author":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 6: Restored","accounts_in_message":[],"_revision_number":6},{"id":"3d117879b2caae6c6b60d6ca1092319fa3ae735b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 7:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":7},{"id":"695d9b09adaa9a0d8ab7faa5390145a4cc1264db","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 7: Works for me\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/7/check/gate-nova-specs-python27/628f451 : SUCCESS in 55s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/7/check/gate-nova-specs-docs/0ee49c8/doc/build/html/ : SUCCESS in 1m 09s\n","accounts_in_message":[],"_revision_number":7},{"id":"733397f6ebee5db2ba6a63aed15d287c55a85ac6","author":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"date":"2014-04-28 20:01:23.000000000","message":"Patch Set 7: Work In Progress","accounts_in_message":[],"_revision_number":7},{"id":"db1e7f4fae6bd325ba91ae4ff07cc39b122a01e9","author":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"date":"2014-04-28 20:01:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"325b9318dad9feb5516038e8fbe7e473be98c2ec","author":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"date":"2014-04-28 21:23:31.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"94a13293bef320f4036fa0ef3c7374ddf740a444","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-02 23:13:20.000000000","message":"Patch Set 7: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":7},{"id":"d0737777d98620326140369af0549727f1347d63","author":{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},"date":"2014-05-02 23:13:20.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)\n\nLike the idea, and the spec looks mostly good.\n\nJust one point I\u0027d like clarified.","accounts_in_message":[],"_revision_number":7},{"id":"01665af84a1bb60346486a12bc47b9e60fb2a7f8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-02 23:19:57.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/7/check/gate-nova-specs-python27/145ca5d : SUCCESS in 59s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/7/check/gate-nova-specs-docs/bea4fd2/doc/build/html/ : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":7},{"id":"216f19d60edf1bc1f1a616726a0f43b6749b860c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-12 10:17:42.000000000","message":"Patch Set 7: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":7},{"id":"da79e3a0601338326d2676e6ad12851ed090a3eb","author":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"date":"2014-05-12 10:17:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7aa49a858ac34ad25894e50faf32041e59a30b13","author":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"date":"2014-05-12 10:17:51.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"0e9475cbc6ab2a43db85a616092ebedee0147e84","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-12 10:17:52.000000000","message":"Patch Set 8:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":8},{"id":"0fede78581d0bd5b55edb0773c3ffd75f57ec47a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-12 10:18:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/8/check/gate-nova-specs-python27/551646d : SUCCESS in 38s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/8/check/gate-nova-specs-docs/5b4714e/doc/build/html/ : SUCCESS in 40s","accounts_in_message":[],"_revision_number":8},{"id":"cb2cf5727c75abb7f2e46981075901ddb6103ae9","author":{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},"date":"2014-05-13 02:21:54.000000000","message":"Patch Set 8: Code-Review+1\n\nThanks for the clarification","accounts_in_message":[],"_revision_number":8},{"id":"6c8a15ad0fd4249ea457b92dc99fd2300cf5fac6","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2014-06-04 01:29:20.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"57f099bd294895926eeb47891e37259c0b1546cd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-04 01:29:21.000000000","message":"Patch Set 8: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":8},{"id":"edad1811b17e68802b4a543621aec033f776ec1a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-04 02:42:55.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/8/check/gate-nova-specs-python27/8cec5d1 : SUCCESS in 38s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/8/check/gate-nova-specs-docs/23d9187/doc/build/html/ : SUCCESS in 53s","accounts_in_message":[],"_revision_number":8},{"id":"5817cc9995602bbdca62ac1b40740a91c452e885","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2014-06-04 12:12:39.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"d0922e571da36e1ca853046d712149d253ee212f","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-06-10 21:05:39.000000000","message":"Patch Set 8:\n\nI love the idea. I was looking to do something similar for Ironic, but we need extra information and possibly a slightly different layout. We\u0027re in a similar boat: no DHCP, no nova-agent, configuring networking with configdrive and cloud-init. The big two things we need support for are VLANs and bonded interfaces. The main change I see to make support for more advanced configurations is moving the MAC addresses into their own field outside of ports, and referencing them from the ports. Decoupling the links makes bonding, bridging, etc easy. Here\u0027s the layout I came up with: http://paste.openstack.org/show/83580/. \n\nThe main issue with bonds and VLANs is the lack of support in Neutron. We have a Neutron plugin we\u0027re working on upstreaming, and I\u0027ve seen others working on specs to do the same.","accounts_in_message":[],"_revision_number":8},{"id":"964b6895d5024b0b0342fc0c35a1ab4c86ae07c2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-10 21:05:40.000000000","message":"Patch Set 8: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":8},{"id":"7d23eaf02f11383c9ea3831d20c6cd9c253fff3b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-10 21:12:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/8/check/gate-nova-specs-python27/38ff92b : SUCCESS in 1m 01s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/8/check/gate-nova-specs-docs/379aeec/doc/build/html/ : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":8},{"id":"94fb0f90117004f89d61c075f584eec3c494bf8a","author":{"_account_id":4573,"name":"yunhong-jiang","email":"yunhong.jiang@intel.com","username":"yunhong-jiang"},"date":"2014-06-11 04:40:18.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"a023cdf4276b2219fec63a48a00c678ddbcad37d","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-06-24 20:46:42.000000000","message":"Patch Set 8: Code-Review-1\n\nSo I reworked my previous example to be a bit more clear and flexible, taking some inspiration from systemd-networkd. I think the best way to do this is separating the network portion (IPs, netmasks, routes, etc) from the links that give them connectivity, which include VIFs, NICs (either straight physical ones like you\u0027d have in Ironic or PCI passthrough NICs in VMs), VLANs, and future support for bonds/bridges/etc. I think if we\u0027re going to pass JSON network information out, we should make it as flexible as possible so we don\u0027t have to rework it in a year. I\u0027m more than willing to write the code to support something like this.\n\nhttp://paste.openstack.org/show/84822/","accounts_in_message":[],"_revision_number":8},{"id":"c0570df561fa562d3277aca6c9ac92227b77f275","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-24 20:46:43.000000000","message":"Patch Set 8: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":8},{"id":"ea5074d4c482ac8e46b012bc377e94fd20de0233","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-24 21:02:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/8/check/gate-nova-specs-python27/c89e30a : SUCCESS in 1m 07s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/8/check/gate-nova-specs-docs/34e2ece/doc/build/html/ : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":8},{"id":"b4863deb6e2f6c59a3229778fd89432de9e0b5cc","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-06-26 18:32:55.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"29dcff73bc81acf54c0175133f33bb107bbb3130","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-26 18:32:56.000000000","message":"Patch Set 9:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":9},{"id":"9c6eb89eb8cb64749f7e9c1c90cf6ddf57bbf85b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-26 18:50:30.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/9/check/gate-nova-specs-python27/25b2001 : SUCCESS in 32s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/9/check/gate-nova-specs-docs/65af152/doc/build/html/ : SUCCESS in 51s","accounts_in_message":[],"_revision_number":9},{"id":"db25efcba085685b9eaab0770c630bc3f2b97e20","author":{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},"date":"2014-06-27 13:02:20.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"39cb44eef1096dbe9ef88f2f82cd6f0666f100a1","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2014-07-01 17:44:53.000000000","message":"Patch Set 9: Code-Review+2\n\nOk, I think I like the link separation, so you can talk about bonds.\n\nI am a little worried about making this move before we see what the neutron API does to represent that.\n\nOn the other hand, I want to see this move forward, and I don\u0027t have a better idea.","accounts_in_message":[],"_revision_number":9},{"id":"f401c82d251e52929252767670d886f4bd034e0a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-01 17:44:54.000000000","message":"Patch Set 9: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":9},{"id":"93ab313508aaa79356225eb64886b57ca45aefcc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-01 17:52:57.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/9/check/gate-nova-specs-python27/e899806 : SUCCESS in 1m 15s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/9/check/gate-nova-specs-docs/588f124/doc/build/html/ : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":9},{"id":"7341175f759f01409517ac7370b71e54873b7f55","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-07-07 21:38:47.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6df833660703be83d30bcbde1423deb8fc5d0117","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-07 21:38:48.000000000","message":"Patch Set 10:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":10},{"id":"39ef052dceb96bc0d726b6137e486faf6124208f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-07 21:40:33.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/10/check/gate-nova-specs-python27/401dc09 : SUCCESS in 37s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/10/check/gate-nova-specs-docs/af518a5/doc/build/html/ : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":10},{"id":"ce009b9ad07ec66993eb1c74d2f04d1c90d60eb8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2014-07-07 21:43:06.000000000","message":"Patch Set 10: Code-Review+2\n\nThis was already +2d, I just nitted out on the commit message line length.","accounts_in_message":[],"_revision_number":10},{"id":"a64026c4f816ca49aa52de5895fd90100ff422da","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2014-07-07 21:53:49.000000000","message":"Patch Set 10: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"5ba8c2f51d2c4fe685498929b68a30b2cf4c74c6","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-07-07 22:37:48.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"67bb6930c537f350970ebfb8833ad5d9568292cd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-07 22:37:50.000000000","message":"Patch Set 11:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":11},{"id":"5df52f2da064fb6b8dda74af85189537d93e3ea4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-07 22:41:36.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/11/check/gate-nova-specs-python27/852ebb2 : SUCCESS in 1m 24s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/11/check/gate-nova-specs-docs/8df6494/doc/build/html/ : SUCCESS in 2m 13s","accounts_in_message":[],"_revision_number":11},{"id":"df7e1cbd38a84ccef27c257cefa83f1162b88679","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2014-07-07 23:28:22.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)\n\nlooking for some clarification on the meaning of id. I thought I knew before, but now I\u0027m not sure. Maybe it deserves an inline comment in the api example...","accounts_in_message":[],"_revision_number":11},{"id":"66635ce79a4f3f93eef8d7c658f8f83e91c46894","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-07-08 00:15:05.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"67b8eb9a94d6371bca163f7a6b62cc8972dc4978","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-07-08 00:15:22.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"22cf5339c0ce6db5451f37f084360f8f0f3472d3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-08 00:15:23.000000000","message":"Patch Set 12:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":12},{"id":"4a7c0814d3300c760d7aed646adbe18b223d031e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-08 00:16:35.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/12/check/gate-nova-specs-python27/94e92d8 : SUCCESS in 1m 10s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/12/check/gate-nova-specs-docs/61322c2/doc/build/html/ : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":12},{"id":"81dc57b55c0863ef5843895b170fde1d0d2ab621","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2014-07-08 03:04:41.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4d7b3c54ece445b68d8f54aac1ecd643d6367ccf","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2014-07-08 14:05:57.000000000","message":"Patch Set 12:\n\nHere\u0027s some additional feedback maybe we should try to make more clear:\n\nhttp://paste.openstack.org/show/85675/","accounts_in_message":[],"_revision_number":12},{"id":"61497e6039a0c84530f78e2c35fef500b3344c13","author":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"date":"2014-07-08 14:06:48.000000000","message":"Patch Set 12: Code-Review-1\n\n(3 comments)\n\nFrom a technical perspective:\n\nI\u0027m concerned that the \"id\" field in the example JSON appears to refer to Linux interface names, which would never be visible to Neutron.\n\nFrom a blueprint perspective:\n\nWhile I don\u0027t disagree that this is a useful feature, I think that the blueprint in its current form doesn\u0027t make a very good case for it.  The \"Problem description\" section, in particular, does not open with a clear description of why the current state of things is a problem.\n\nI also think the blueprint needs to be more explicit about what information it will be exposing and where that information will come from.","accounts_in_message":[],"_revision_number":12},{"id":"446f8af6670634115a3f21e1bf282bb178101d37","author":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"date":"2014-07-08 15:55:07.000000000","message":"Patch Set 12: Code-Review-1\n\nWe have a specific use case for this blueprint:\n- No DHCP\n- Injection of additional static routes\n- Improve support of non-Debian operating systems\n\nThe challenges are:\n- Injecting additional static routes in the final network configuration\n- Parsing a Debian-style network interfaces config file (as stated by comstud on IRC)\n\nThis blueprint would address those challenges by:\n- Providing more information about the network configuration (static routes)\n- Providing such information in an easily parseable format (JSON).\n\nTo give an idea of how challenging this is currently is:\n- we have to parse a Debian-style interfaces file on non-Debian platforms (CentOS and Windows) to get the actual IP address (this is obviously sub-optional)\n- we have to inject a custom script in our guest to retrieve (optimistically parse) and configure the static routes\n\nThis blueprint should reflect those challenges (especially benefits for non-Debian platforms) and how it plans to address them.","accounts_in_message":[],"_revision_number":12},{"id":"1aa80b8a24a2c4440d2a95acbdeb3feb4517fc04","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-07-10 01:12:49.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"af5f77d513cd51a5d236a3b973eb52fdac634955","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-10 01:12:50.000000000","message":"Patch Set 13:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":13},{"id":"328afa5f56ad161851189801ba11a475c0732496","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-10 01:13:55.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/13/check/gate-nova-specs-python27/90b52d6 : SUCCESS in 47s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/13/check/gate-nova-specs-docs/01944ec/doc/build/html/ : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":13},{"id":"a1ce43ba7e705956b02b3f2da47124ea19f9c3a9","author":{"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},"date":"2014-07-10 04:56:42.000000000","message":"Patch Set 13: Code-Review-1\n\n(12 comments)\n\nNice.  Go talk to freenode\u0027s #cloud-init people.  And you\u0027re talking only metadata but you should be talking \u0027this is a file that can be fetched from either metadata or on the config-drive\u0027 as they\u0027re basically equivalent at the moment (and for good reason).","accounts_in_message":[],"_revision_number":13},{"id":"8c4821d2830cffcd6a3cafec55640f748b0f8dfd","author":{"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},"date":"2014-07-10 05:03:25.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"b775a12c566fbe6fed87a875eba429cafdaa87d2","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-07-10 21:51:28.000000000","message":"Patch Set 13:\n\n(10 comments)\n\nThanks for the review! I\u0027ll work on updating these. This likely isn\u0027t going to make it in for the approval deadline today, but I\u0027d like to keep iterating on this, because I\u0027ll need to implement this (even if it isn\u0027t a permanent implementation) before K specs open up.","accounts_in_message":[],"_revision_number":13},{"id":"ce29f40db5e38bcb041fdbe4897cd212acde08e0","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2014-07-11 18:14:56.000000000","message":"Patch Set 13:\n\nI hope parsing a debian file isn\u0027t that hard.\n\nFeel free to use (or reference):\n\nhttp://bazaar.launchpad.net/~cloud-init-dev/cloud-init/trunk/view/head:/cloudinit/distros/net_util.py\n\nI\u0027m +2 on a better format that is agnostic to debian, rhel, ...","accounts_in_message":[],"_revision_number":13},{"id":"058585663514db91a294d38df7d3af0eb6241731","author":{"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},"date":"2014-07-11 18:29:24.000000000","message":"Patch Set 13:\n\nActually, there are many problems with interpreting an interfaces file, generally related to its semantic meaning after parsing:\n\n- the bonding configuration changed at some point (list-on-master to list-on-slave), but both sets of options sort-of-work and so you now find that some configs work and some don\u0027t.  Exporting that to other OSes seems evil\n- options can be added by extensions to the basic parser.  So bonding options are ignored if bonding packages are not present and respected if they are.  Meaning that any given interfaces file has a meaning on a given host, and a more abstract meaning that the author actually intended\n- there are any number of reasons why you have to resort to \u0027up\u0027 and \u0027down\u0027 commands\n- IPv4 behaviour is \u0027off till I turn it on\u0027 and IPv6 behaviour is \u0027off unless the interface is up, even if it\u0027s up for an IPv4 stanza\u0027\n\nI\u0027m being a little pedantic, but I think there\u0027s enough reason to justify using a decent format for the interface description, because the Debian format is often ambiguous.","accounts_in_message":[],"_revision_number":13},{"id":"de620158f5e5541ee4b283938324a386d09ffeda","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2014-07-13 05:59:23.000000000","message":"Patch Set 13:\n\nAgreed, better format, good :)","accounts_in_message":[],"_revision_number":13},{"id":"50d52fced2af13748a844091130d68cf0c2988ac","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-13 05:59:24.000000000","message":"Patch Set 13: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":13},{"id":"2afb2e1128f27cdc333203ea37b5f071b6dc46ad","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-13 06:00:47.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/13/check/gate-nova-specs-python27/f92ccec : SUCCESS in 1m 06s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/13/check/gate-nova-specs-docs/0087f4f/doc/build/html/ : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":13},{"id":"e4c257ceb810e7a261838570ec88d5eb8e6793fb","author":{"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},"date":"2014-07-13 06:04:32.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"0f6eb4c337deaa39c042262965759ecfc0074614","author":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"date":"2014-07-14 00:10:08.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)\n\n-1 for inline comments.\n\nOther comment:\n\nTBH, I was more or less thinking about a direct dump of the network model found in nova/network/model. (with id stripped if it is a concern)\n\nWhy reinventing yet an other model? We have plenty already. Lets use them.","accounts_in_message":[],"_revision_number":13},{"id":"bd56fb8671bde95bf3c170e3f257d88ebe718ffd","author":{"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},"date":"2014-07-14 00:42:19.000000000","message":"Patch Set 13:\n\nNova network model is missing bonding and MTUs, and can\u0027t express multiple v4 addresses (I think - may be wrong on that) but it might be a place to start.","accounts_in_message":[],"_revision_number":13},{"id":"754681ce368c377c2537a7307442a1333d50be56","author":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"date":"2014-07-14 05:28:36.000000000","message":"Patch Set 13:\n\nAFAIK, bonding is not part of Nova or Neutron models. We would have to implement it first in those services before introducing such information in what I consider the \"end of the chain\". IMO, it\u0027s not up to this blueprint to define such modeling.\n\nAbout MTU, I\u0027m not sure why it isn\u0027t part of the model already. Shouldn\u0027t be hard to add it.\n\nAbout multiple ips, the model supports it. ips is a list of ips: https://github.com/openstack/nova/blob/2014.2.b1/nova/network/model.py#L176 (linking to juno-1 so line number does not change over time)","accounts_in_message":[],"_revision_number":13},{"id":"e2560a51ce583e36a26c2e9f3a1a9de487898f8f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-16 11:05:54.000000000","message":"Patch Set 13: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":13},{"id":"feaeaa093e7d9177ce59d236d4fa6cb01eb8e947","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2014-07-16 11:05:54.000000000","message":"Patch Set 13: Code-Review-2\n\nSorry, this blueprint has missed the Juno spec approval freeze 10 July:\nhttp://lists.openstack.org/pipermail/openstack-dev/2014-July/039259.html\nPlease consider re-submitting when we open nova-specs for the K release.\nIf you are wanting an exception, please raise this in the next nova-meeting.","accounts_in_message":[],"_revision_number":13},{"id":"2488bf75cbf7ddcd5d16c5984d804d05bdcf7bf7","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2014-07-16 11:06:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":13},{"id":"8f06f9dc4f82a860881d3cdb21481a4ebdf0a28f","author":{"_account_id":1082,"name":"Scott Moser","email":"smoser@ubuntu.com","username":"smoser"},"date":"2014-07-18 14:08:38.000000000","message":"Patch Set 13:\n\n(3 comments)\n\nHi,\n I really like this idea, and think its basically a requirement\n A couple things I think we might need to consider that I didn\u0027t see mentioned.\na.) MTU on the network interfaces\nb.) multiple IP addresses per NIC\n\nJust in case you weren\u0027t aware, Amazon does this sort of thing also.  I\u0027m not suggesting it should be copied, but that we shoudl be aware of it as it might give light to some problems that we\u0027d see.\n\nLast major thing is that the api/format has to be versioned.\n\nthank you for taking the time to write this.","accounts_in_message":[],"_revision_number":13},{"id":"45896af44014aba83795c29350e6aa0df9b9d718","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2014-07-18 15:04:17.000000000","message":"Restored","accounts_in_message":[],"_revision_number":13},{"id":"63ef489b889d9bfc1da843ce36cde70017d8b9c1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-18 15:04:19.000000000","message":"Patch Set 13:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":13},{"id":"6398a9281810c72accb6da2e733a29e5a6a84fb4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-18 15:41:22.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded.\n\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/13/check/gate-nova-specs-python27/53b5529 : SUCCESS in 1m 19s\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/13/check/gate-nova-specs-docs/9ccdd46/doc/build/html/ : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":13},{"id":"825105d0dc6753d63d55817869a730398145256f","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-09-04 19:53:35.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"bfcd2e3b72db91436f8128d4ce0f077200f0a1ac","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-09-04 19:59:36.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"3759950744d01c60e2848065bcfcb2f62efa4b3e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-04 20:50:14.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded.\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/15/check/gate-nova-specs-docs/71f709c/doc/build/html/ : SUCCESS in 50s\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/15/check/gate-nova-specs-python27/1642960 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":15},{"id":"c4c411f2e4c06d7f5db9fb34807bdc8f3342c2d0","author":{"_account_id":1082,"name":"Scott Moser","email":"smoser@ubuntu.com","username":"smoser"},"date":"2014-10-01 20:05:49.000000000","message":"Patch Set 15:\n\nJust a comment here.\nI was thinking some about boot and cloud-init.\nI think it might be useful to declare that a interface is mandatory.  I could see this being necessary/useful in at least the following scenario:\\\n\n- config drive passes networking info to get a route to metadata service\n- the config drive data might be stale at any point.  so cloud-init should configure networking per config drive on boot, then bring up networking to ask the MD what *it* knows about, as config drive info could be stale.\n\nif the information on the config drive is wrong, and contains a NIC that has been un-hotplugged, then cloud-init cannot reasonably determine between \"hotplug hasnt occurred yet\" and \"hotplug never will\".\n\npossibly that just means that the network devices in config drive should never be removed. but that seems limiting.","accounts_in_message":[],"_revision_number":15},{"id":"53f110ab3af58c6fc9656fa47a8aa7f6507e4bbd","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-10-10 22:38:12.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"2eeda441a66d955345eac6bbae10dfbf70ae3343","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-10-10 22:40:16.000000000","message":"Patch Set 16:\n\nScott, I think that\u0027s reasonable and a good idea. I think it would require Neutron having support for required interfaces, otherwise Nova doesn\u0027t really have a way to mark one as required and one as not.","accounts_in_message":[],"_revision_number":16},{"id":"6678f28b085d21821356cc5e1aa6c7bcfec031a0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-10 22:44:47.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded.\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/16/check/gate-nova-specs-docs/53dcdc8/doc/build/html/ : SUCCESS in 1m 47s\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/16/check/gate-nova-specs-python27/dd694f0 : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":16},{"id":"313b3af7ef8a25dec6a2aaf6166d46a21a43dae9","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2014-10-13 14:10:57.000000000","message":"Patch Set 16: -Code-Review\n\nkilo is now open","accounts_in_message":[],"_revision_number":16},{"id":"7d41288799cddcc00cb45993d1818423f153f45d","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2014-10-21 17:37:49.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)\n\nDefinitely support this blueprint and idea, but I\u0027d like to see details on how you plan to introduce versioning to the network_data JSON document returned from the request to the metadata service. See inline for details.\n\nBest,\n\n-jay","accounts_in_message":[],"_revision_number":16},{"id":"f50c551f57484d182770bcc8c4a9f524710f3c43","author":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"date":"2014-11-04 13:59:34.000000000","message":"Patch Set 16:\n\nThis is worth examining too I think. https://github.com/openstack/tripleo-specs/blob/master/specs/juno/network_configuration.rst","accounts_in_message":[],"_revision_number":16},{"id":"48aed51df1135d85952fd1122509a99ec537a9e4","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-11-12 22:11:51.000000000","message":"Patch Set 16:\n\n(1 comment)\n\n@jaypipes thanks! just a question\n\n@lifeless the formats are remarkably similar for being developed independently. Let me look into how close they are and if there are any major differences.","accounts_in_message":[],"_revision_number":16},{"id":"11eae9f55d55389b4fc3e400966af280008304bb","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2014-11-18 17:12:08.000000000","message":"Patch Set 16: Code-Review+1\n\n(3 comments)\n\nComments inline. Nothing showstopping afaict.\n\nNotably; Rackspace OnMetal is running a version of this proposal downstream (with the information in vendor_data), and has a downstream patch to cloud-init to make it work (https://github.com/racker/cloud-init/pull/1). This has enabled us to use configurations that wouldn\u0027t otherwise have been possible with the existing exposed network information.\n\nI\u0027m working with upstream cloud-init, and we\u0027re eager to see these changes come in to make network interfaces file generation more simple.","accounts_in_message":[],"_revision_number":16},{"id":"578a23d5a448676e39992864993016f34e9810d5","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2014-12-09 22:47:49.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"338bbe580cf6679aac6475cccddd931fd19f1cc5","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-12-09 23:05:09.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"ff3f3c63cb8319700ec9a0dc75e32f1fa17834a5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-09 23:07:02.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/17/check/gate-nova-specs-docs/4ee9b0c/doc/build/html/ : SUCCESS in 1m 50s\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/17/check/gate-nova-specs-python27/e7407b7 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":17},{"id":"961e656410407633b5d13b43b7487c6073d643e9","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2014-12-10 00:33:21.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"d306e396d8a64779d963f7c87a4dcb3155ad3a58","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2014-12-12 13:38:11.000000000","message":"Patch Set 17: Code-Review+1\n\nSo metadata already has a versioning mechanism, which means I don\u0027t think there is really an API impact here (we know how to do versioning in that case).","accounts_in_message":[],"_revision_number":17},{"id":"30cbdba60686ae7ccc5fa17d2d91716adb74c424","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-12-12 21:36:30.000000000","message":"Patch Set 17:\n\n(2 comments)\n\ninline question","accounts_in_message":[],"_revision_number":17},{"id":"d57ed76114513d73f9c2408a5c879a63c6d1ddc1","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-12-12 22:15:52.000000000","message":"Patch Set 17:\n\n(2 comments)\n\nThanks! Joe: Responses inline.","accounts_in_message":[],"_revision_number":17},{"id":"dec6ce0b10c1b623b89441a1eee61e72b9bd2643","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-12-13 00:04:35.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)\n\nsee inline, overall looks good","accounts_in_message":[],"_revision_number":17},{"id":"6a472240f6ccdb120c3b228bd198c6af914eb0eb","author":{"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},"date":"2014-12-14 12:59:03.000000000","message":"Patch Set 17: Code-Review+1\n\nI don\u0027t think there\u0027s a need to tie it to the HA use cases, and what\u0027s there fixes any number of problems (e.g. Linux-specific, Debian-specific interfaces file template, MTUs) that we have at the moment.  What\u0027s here looks fine, and the format is sane; go for it.","accounts_in_message":[],"_revision_number":17},{"id":"5722b4fcfe1a6abda4bd73a3fd91079890c255f9","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-12-15 18:25:25.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"395dbff2a3fd0c7d7d3cffe4951c9fff98ecf5bd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-15 18:54:06.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/18/check//gate-nova-specs-docs/e9cfe0c//doc/build/html/ : SUCCESS in 2m 06s\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/18/check//gate-nova-specs-python27/64a84b9/ : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":18},{"id":"0f5fcd4a362efd2038206ca4cd16de2ae4460968","author":{"_account_id":11547,"name":"Andy Tylee","email":"andy.tylee@metaswitch.com"},"date":"2014-12-15 23:52:21.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"75fbc1836f1b1896eefc3a50827c1bbd8fd2e342","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2014-12-17 22:42:24.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"26acea53ad71f5cbd280b573ea361e8372ddeab8","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-12-17 23:06:17.000000000","message":"Patch Set 18: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"702ecf291063fbb309240a1a8a74c4ee2e3a2977","author":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"date":"2014-12-17 23:12:19.000000000","message":"Patch Set 18:\n\n(1 comment)\n\nThanks for the review, Joe, added the comment so anyone reviewing has context to why I chose this over the alternative.","accounts_in_message":[],"_revision_number":18},{"id":"373f6a41dca96e1449c42d6e2e614316712a4219","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2014-12-18 21:40:14.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1\n\nSeems good to me.\n\nFull disclosure: I worked with comstud on some plans for this, and did a design summit on this idea, on a previous blueprint I never got chance to finish.","accounts_in_message":[],"_revision_number":18},{"id":"b8d24c4341f0076cf3fbefea62d4d201df30eb90","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-18 21:40:22.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":18},{"id":"41799b29008bb991ce812e39f22b1025a7576419","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-18 21:43:25.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/73/85673/18/gate/gate-nova-specs-docs/f22805d//doc/build/html/ : SUCCESS in 1m 50s\n- gate-nova-specs-python27 http://logs.openstack.org/73/85673/18/gate/gate-nova-specs-python27/8c36afc/ : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":18},{"id":"27e834b8bb92c3d49388915c0c3f67c7807f3b29","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-18 21:43:27.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"c43781f4fe40dffeef31012397363d2a29c21028","revisions":{"131ded7f810a497a7c5a6918e19ad0cca11660df":{"kind":"REWORK","_number":1,"created":"2014-04-28 20:01:23.000000000","uploader":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"ref":"refs/changes/73/85673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/1"}}},"commit":{"parents":[{"commit":"efdbbb13d5bd2fc86d3e34ce908b5573a01a59c5","subject":"Add .testrepository to gitignore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/efdbbb13d5bd2fc86d3e34ce908b5573a01a59c5"}]}],"author":{"name":"Claxton","email":"claxton@gmail.com","date":"2014-04-07 09:25:56.000000000","tz":330},"committer":{"name":"Claxton","email":"claxton@gmail.com","date":"2014-04-07 09:28:28.000000000","tz":330},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network information.\nLike user_data and vendor_data, we should also provide network_data. Really we should\nprovide all the information we can (from neutron) in both config drive and metadata service.\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/131ded7f810a497a7c5a6918e19ad0cca11660df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/131ded7f810a497a7c5a6918e19ad0cca11660df"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"efdbbb13d5bd2fc86d3e34ce908b5573a01a59c5","is_merged_in_target_branch":true,"change_id":"Ifb1b5da6cfa708b1f6eb2ab3f12b1b580704b04f","change_number":85422,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c132f336ee29c54d27dc1021fc6b5cc13ab14f72":{"kind":"REWORK","_number":2,"created":"2014-04-28 20:01:23.000000000","uploader":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"ref":"refs/changes/73/85673/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/2"}}},"commit":{"parents":[{"commit":"efdbbb13d5bd2fc86d3e34ce908b5573a01a59c5","subject":"Add .testrepository to gitignore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/efdbbb13d5bd2fc86d3e34ce908b5573a01a59c5"}]}],"author":{"name":"Claxton","email":"claxton@gmail.com","date":"2014-04-07 09:25:56.000000000","tz":330},"committer":{"name":"Claxton","email":"claxton@gmail.com","date":"2014-04-07 09:36:35.000000000","tz":330},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network information.\nLike user_data and vendor_data, we should also provide network_data. Really we should\nprovide all the information we can (from neutron) in both config drive and metadata service.\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c132f336ee29c54d27dc1021fc6b5cc13ab14f72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c132f336ee29c54d27dc1021fc6b5cc13ab14f72"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"efdbbb13d5bd2fc86d3e34ce908b5573a01a59c5","is_merged_in_target_branch":true,"change_id":"Ifb1b5da6cfa708b1f6eb2ab3f12b1b580704b04f","change_number":85422,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d1eda4bcf2d02fdaea5cc5b3936233897199f814":{"kind":"REWORK","_number":3,"created":"2014-04-28 20:01:23.000000000","uploader":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"ref":"refs/changes/73/85673/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/3"}}},"commit":{"parents":[{"commit":"efdbbb13d5bd2fc86d3e34ce908b5573a01a59c5","subject":"Add .testrepository to gitignore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/efdbbb13d5bd2fc86d3e34ce908b5573a01a59c5"}]}],"author":{"name":"Claxton","email":"claxton@gmail.com","date":"2014-04-07 09:25:56.000000000","tz":330},"committer":{"name":"Claxton","email":"claxton@gmail.com","date":"2014-04-07 09:41:19.000000000","tz":330},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network information.\nLike user_data and vendor_data, we should also provide network_data. Really we should\nprovide all the information we can (from neutron) in both config drive and metadata service.\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d1eda4bcf2d02fdaea5cc5b3936233897199f814"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d1eda4bcf2d02fdaea5cc5b3936233897199f814"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"efdbbb13d5bd2fc86d3e34ce908b5573a01a59c5","is_merged_in_target_branch":true,"change_id":"Ifb1b5da6cfa708b1f6eb2ab3f12b1b580704b04f","change_number":85422,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ddde5e11ee67ef2cef8827a62f848012584ebe97":{"kind":"REWORK","_number":4,"created":"2014-04-28 20:01:23.000000000","uploader":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"ref":"refs/changes/73/85673/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/4"}}},"commit":{"parents":[{"commit":"f4d8e19065bab653717187560897e565680cfec8","subject":"Add standardize-nova-image blueprint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f4d8e19065bab653717187560897e565680cfec8"}]}],"author":{"name":"Claxton","email":"claxton@gmail.com","date":"2014-04-07 09:25:56.000000000","tz":330},"committer":{"name":"Claxton","email":"claxton@gmail.com","date":"2014-04-09 05:45:23.000000000","tz":330},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network information.\nLike user_data and vendor_data, we should also provide network_data. Really we should\nprovide all the information we can (from neutron) in both config drive and metadata service.\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ddde5e11ee67ef2cef8827a62f848012584ebe97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ddde5e11ee67ef2cef8827a62f848012584ebe97"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f4d8e19065bab653717187560897e565680cfec8","is_merged_in_target_branch":true,"change_id":"I17b3dc79c0a2286f019506e4869b23f03e15a0e7","change_number":85569,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8047020845b672243a96109f205f292002a9b522":{"kind":"REWORK","_number":5,"created":"2014-04-28 20:01:23.000000000","uploader":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"ref":"refs/changes/73/85673/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/5"}}},"commit":{"parents":[{"commit":"f4d8e19065bab653717187560897e565680cfec8","subject":"Add standardize-nova-image blueprint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f4d8e19065bab653717187560897e565680cfec8"}]}],"author":{"name":"Claxton","email":"claxton@gmail.com","date":"2014-04-07 09:25:56.000000000","tz":330},"committer":{"name":"Claxton","email":"claxton@gmail.com","date":"2014-04-09 05:49:45.000000000","tz":330},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network information.\nLike user_data and vendor_data, we should also provide network_data. Really we should\nprovide all the information we can (from neutron) in both config drive and metadata service.\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8047020845b672243a96109f205f292002a9b522"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8047020845b672243a96109f205f292002a9b522"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f4d8e19065bab653717187560897e565680cfec8","is_merged_in_target_branch":true,"change_id":"I17b3dc79c0a2286f019506e4869b23f03e15a0e7","change_number":85569,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b14d1fc41d34a5b81bdedebe7f03194a8b9612e0":{"kind":"REWORK","_number":6,"created":"2014-04-28 20:01:23.000000000","uploader":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"ref":"refs/changes/73/85673/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/6"}}},"commit":{"parents":[{"commit":"f4d8e19065bab653717187560897e565680cfec8","subject":"Add standardize-nova-image blueprint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f4d8e19065bab653717187560897e565680cfec8"}]}],"author":{"name":"Claxton","email":"claxton@gmail.com","date":"2014-04-07 09:25:56.000000000","tz":330},"committer":{"name":"Claxton","email":"claxton@gmail.com","date":"2014-04-09 05:54:11.000000000","tz":330},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network information.\nLike user_data and vendor_data, we should also provide network_data. Really we should\nprovide all the information we can (from neutron) in both config drive and metadata service.\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b14d1fc41d34a5b81bdedebe7f03194a8b9612e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b14d1fc41d34a5b81bdedebe7f03194a8b9612e0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f4d8e19065bab653717187560897e565680cfec8","is_merged_in_target_branch":true,"change_id":"I17b3dc79c0a2286f019506e4869b23f03e15a0e7","change_number":85569,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"80da582bb2725f907869513a60901f9f985d9d32":{"kind":"REWORK","_number":7,"created":"2014-04-28 20:01:23.000000000","uploader":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"ref":"refs/changes/73/85673/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/7"}}},"commit":{"parents":[{"commit":"48888062f39cdb8ea00606d5d435613b72f7843a","subject":"Merge \"Add a note about requiring ascii diagrams\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/48888062f39cdb8ea00606d5d435613b72f7843a"}]}],"author":{"name":"Claxton","email":"claxton@gmail.com","date":"2014-04-07 09:25:56.000000000","tz":330},"committer":{"name":"Claxton Correya","email":"claxton@gmail.com","date":"2014-04-24 06:52:04.000000000","tz":330},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network information.\nLike user_data and vendor_data, we should also provide network_data. Really we should\nprovide all the information we can (from neutron) in both config drive and metadata service.\n\nblueprint metadata-service-network-info\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/80da582bb2725f907869513a60901f9f985d9d32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/80da582bb2725f907869513a60901f9f985d9d32"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"48888062f39cdb8ea00606d5d435613b72f7843a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"5213c7cb02a1cc194654ab11ed10ee412a960958":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2014-05-12 10:17:51.000000000","uploader":{"_account_id":7602,"name":"Claxton Correya","email":"claxton@gmail.com","username":"claxton"},"ref":"refs/changes/73/85673/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/8"}}},"commit":{"parents":[{"commit":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","subject":"add test to ensure there are no \\r","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e71dedc6b45a3c6a5de1b360544f17806cad6ff2"}]}],"author":{"name":"Claxton","email":"claxton@gmail.com","date":"2014-04-07 09:25:56.000000000","tz":330},"committer":{"name":"Claxton Correya","email":"claxton@gmail.com","date":"2014-05-12 10:17:51.000000000","tz":0},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network information.\nLike user_data and vendor_data, we should also provide network_data. Really we should\nprovide all the information we can (from neutron) in both config drive and metadata service.\n\nblueprint metadata-service-network-info\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5213c7cb02a1cc194654ab11ed10ee412a960958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5213c7cb02a1cc194654ab11ed10ee412a960958"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","is_merged_in_target_branch":true,"change_id":"I59ed022e882cfde00c004a66659a43134fd2df5c","change_number":93024,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"82be8da5278e4e87f66262f339f4489a1dd7ce23":{"kind":"REWORK","_number":9,"created":"2014-06-26 18:32:55.000000000","uploader":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"ref":"refs/changes/73/85673/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/9"}}},"commit":{"parents":[{"commit":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","subject":"add test to ensure there are no \\r","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e71dedc6b45a3c6a5de1b360544f17806cad6ff2"}]}],"author":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-06-26 18:32:39.000000000","tz":-420},"committer":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-06-26 18:32:55.000000000","tz":-420},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network information.\nLike user_data and vendor_data, we should also provide network_data. Really we should\nprovide all the information we can (from neutron) in both config drive and metadata service.\n\nblueprint metadata-service-network-info\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/82be8da5278e4e87f66262f339f4489a1dd7ce23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/82be8da5278e4e87f66262f339f4489a1dd7ce23"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","is_merged_in_target_branch":true,"change_id":"I59ed022e882cfde00c004a66659a43134fd2df5c","change_number":93024,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"30b37f9f8392dd8e3121f503215202ac19150dcc":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2014-07-07 21:38:47.000000000","uploader":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"ref":"refs/changes/73/85673/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/10"}}},"commit":{"parents":[{"commit":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","subject":"add test to ensure there are no \\r","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e71dedc6b45a3c6a5de1b360544f17806cad6ff2"}]}],"author":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-06-26 18:32:39.000000000","tz":-420},"committer":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-07-07 21:38:07.000000000","tz":-420},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network\ninformation. Like user_data and vendor_data, we should also provide\nnetwork_data. Really we should provide all the information we can\n(from neutron) in both config drive and metadata service.\n\nblueprint metadata-service-network-info\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/30b37f9f8392dd8e3121f503215202ac19150dcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/30b37f9f8392dd8e3121f503215202ac19150dcc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","is_merged_in_target_branch":true,"change_id":"I59ed022e882cfde00c004a66659a43134fd2df5c","change_number":93024,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f25111388d2509a6220de77c2b8b775844f0cb44":{"kind":"REWORK","_number":11,"created":"2014-07-07 22:37:48.000000000","uploader":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"ref":"refs/changes/73/85673/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/11"}}},"commit":{"parents":[{"commit":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","subject":"add test to ensure there are no \\r","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e71dedc6b45a3c6a5de1b360544f17806cad6ff2"}]}],"author":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-06-26 18:32:39.000000000","tz":-420},"committer":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-07-07 22:37:42.000000000","tz":-420},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network\ninformation. Like user_data and vendor_data, we should also provide\nnetwork_data. Really we should provide all the information we can\n(from neutron) in both config drive and metadata service.\n\nblueprint metadata-service-network-info\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f25111388d2509a6220de77c2b8b775844f0cb44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f25111388d2509a6220de77c2b8b775844f0cb44"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","is_merged_in_target_branch":true,"change_id":"I59ed022e882cfde00c004a66659a43134fd2df5c","change_number":93024,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d97efa1b028d712c437323557442e34e23afac96":{"kind":"REWORK","_number":12,"created":"2014-07-08 00:15:22.000000000","uploader":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"ref":"refs/changes/73/85673/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/12"}}},"commit":{"parents":[{"commit":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","subject":"add test to ensure there are no \\r","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e71dedc6b45a3c6a5de1b360544f17806cad6ff2"}]}],"author":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-06-26 18:32:39.000000000","tz":-420},"committer":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-07-08 00:15:18.000000000","tz":-420},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network\ninformation. Like user_data and vendor_data, we should also provide\nnetwork_data. Really we should provide all the information we can\n(from neutron) in both config drive and metadata service.\n\nblueprint metadata-service-network-info\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d97efa1b028d712c437323557442e34e23afac96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d97efa1b028d712c437323557442e34e23afac96"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","is_merged_in_target_branch":true,"change_id":"I59ed022e882cfde00c004a66659a43134fd2df5c","change_number":93024,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"12a7f713831ebdb7d247cdd82e3d561d7ac11cec":{"kind":"REWORK","_number":13,"created":"2014-07-10 01:12:49.000000000","uploader":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"ref":"refs/changes/73/85673/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/13"}}},"commit":{"parents":[{"commit":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","subject":"add test to ensure there are no \\r","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e71dedc6b45a3c6a5de1b360544f17806cad6ff2"}]}],"author":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-06-26 18:32:39.000000000","tz":-420},"committer":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-07-10 01:12:46.000000000","tz":-420},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network\ninformation. Like user_data and vendor_data, we should also provide\nnetwork_data. Really we should provide all the information we can\n(from neutron) in both config drive and metadata service.\n\nblueprint metadata-service-network-info\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/12a7f713831ebdb7d247cdd82e3d561d7ac11cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/12a7f713831ebdb7d247cdd82e3d561d7ac11cec"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","is_merged_in_target_branch":true,"change_id":"I59ed022e882cfde00c004a66659a43134fd2df5c","change_number":93024,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"564a3cc90df733aa686bb87c59297d9a08ae89d2":{"kind":"REWORK","_number":14,"created":"2014-09-04 19:53:35.000000000","uploader":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"ref":"refs/changes/73/85673/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/14"}}},"commit":{"parents":[{"commit":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","subject":"add test to ensure there are no \\r","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e71dedc6b45a3c6a5de1b360544f17806cad6ff2"}]}],"author":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-06-26 18:32:39.000000000","tz":-420},"committer":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-09-04 19:48:48.000000000","tz":-420},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network\ninformation. Like user_data and vendor_data, we should also provide\nnetwork_data. Really we should provide all the information we can\n(from neutron) in both config drive and metadata service.\n\nblueprint metadata-service-network-info\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/564a3cc90df733aa686bb87c59297d9a08ae89d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/564a3cc90df733aa686bb87c59297d9a08ae89d2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","is_merged_in_target_branch":true,"change_id":"I59ed022e882cfde00c004a66659a43134fd2df5c","change_number":93024,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"68ab780a2e4fa7785aaf1ad748d5660ed58be029":{"kind":"REWORK","_number":15,"created":"2014-09-04 19:59:36.000000000","uploader":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"ref":"refs/changes/73/85673/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/15"}}},"commit":{"parents":[{"commit":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","subject":"add test to ensure there are no \\r","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e71dedc6b45a3c6a5de1b360544f17806cad6ff2"}]}],"author":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-06-26 18:32:39.000000000","tz":-420},"committer":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-09-04 19:57:19.000000000","tz":-420},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network\ninformation. Like user_data and vendor_data, we should also provide\nnetwork_data. Really we should provide all the information we can\n(from neutron) in both config drive and metadata service.\n\nblueprint metadata-service-network-info\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\nCo-Authored-By: Josh Gachnang \u003cjosh@servercobra.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/68ab780a2e4fa7785aaf1ad748d5660ed58be029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/68ab780a2e4fa7785aaf1ad748d5660ed58be029"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e71dedc6b45a3c6a5de1b360544f17806cad6ff2","is_merged_in_target_branch":true,"change_id":"I59ed022e882cfde00c004a66659a43134fd2df5c","change_number":93024,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3e0aa7247906503c1afa4752dcfe3799bc617aa8":{"kind":"REWORK","_number":16,"created":"2014-10-10 22:38:12.000000000","uploader":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"ref":"refs/changes/73/85673/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/16"}}},"commit":{"parents":[{"commit":"f727bebf78f704e68c280ab05dd6e1d92a6b09fc","subject":"Merge \"Create Nova Scheduler IO Ops Weighter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f727bebf78f704e68c280ab05dd6e1d92a6b09fc"}]}],"author":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-06-26 18:32:39.000000000","tz":-420},"committer":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-10-10 22:39:07.000000000","tz":-420},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network\ninformation. Like user_data and vendor_data, we should also provide\nnetwork_data. Really we should provide all the information we can\n(from neutron) in both config drive and metadata service.\n\nblueprint metadata-service-network-info\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\nCo-Authored-By: Josh Gachnang \u003cjosh@servercobra.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3e0aa7247906503c1afa4752dcfe3799bc617aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3e0aa7247906503c1afa4752dcfe3799bc617aa8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f727bebf78f704e68c280ab05dd6e1d92a6b09fc","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"bbf0cbc127e3dce1bb07f718b8cc15b8610eab04":{"kind":"REWORK","_number":17,"created":"2014-12-09 23:05:09.000000000","uploader":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"ref":"refs/changes/73/85673/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/17"}}},"commit":{"parents":[{"commit":"f727bebf78f704e68c280ab05dd6e1d92a6b09fc","subject":"Merge \"Create Nova Scheduler IO Ops Weighter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f727bebf78f704e68c280ab05dd6e1d92a6b09fc"}]}],"author":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-06-26 18:32:39.000000000","tz":-420},"committer":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-12-09 23:05:08.000000000","tz":-480},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network\ninformation. Like user_data and vendor_data, we should also provide\nnetwork_data. Really we should provide all the information we can\n(from neutron) in both config drive and metadata service.\n\nblueprint metadata-service-network-info\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\nCo-Authored-By: Josh Gachnang \u003cjosh@servercobra.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bbf0cbc127e3dce1bb07f718b8cc15b8610eab04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bbf0cbc127e3dce1bb07f718b8cc15b8610eab04"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f727bebf78f704e68c280ab05dd6e1d92a6b09fc","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c43781f4fe40dffeef31012397363d2a29c21028":{"kind":"REWORK","_number":18,"created":"2014-12-15 18:25:25.000000000","uploader":{"_account_id":10380,"name":"Josh Gachnang","email":"josh@pcsforeducation.com","username":"JoshNang"},"ref":"refs/changes/73/85673/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/73/85673/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/73/85673/18"}}},"commit":{"parents":[{"commit":"82e8407c855dae78a40cdfa8208ca8d950fb4155","subject":"Merge \"Relax resource name restrictions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/82e8407c855dae78a40cdfa8208ca8d950fb4155"}]}],"author":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-06-26 18:32:39.000000000","tz":-420},"committer":{"name":"Josh Gachnang","email":"josh@pcsforeducation.com","date":"2014-12-15 18:25:20.000000000","tz":-480},"subject":"API: Proxy neutron configuration to guest instance","message":"API: Proxy neutron configuration to guest instance\n\nCurrently the metadata service only contains minimal network\ninformation. Like user_data and vendor_data, we should also provide\nnetwork_data. Really we should provide all the information we can\n(from neutron) in both config drive and metadata service.\n\nblueprint metadata-service-network-info\n\nChange-Id: Ia6967af4b4147ad4dcf6ae3ae5f4b6d9c1fda246\nCo-Authored-By: Josh Gachnang \u003cjosh@servercobra.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c43781f4fe40dffeef31012397363d2a29c21028"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c43781f4fe40dffeef31012397363d2a29c21028"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"82e8407c855dae78a40cdfa8208ca8d950fb4155","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
