)]}'
{"id":"openstack%2Fnova-specs~858858","triplet_id":"openstack%2Fnova-specs~master~I122fd0b9b9630a59059062e2186ac5fdd238e5bc","project":"openstack/nova-specs","branch":"master","topic":"compute-customizable-inventory","attention_set":{"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2022-11-15 18:06:29.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"28542":{"account":{"_account_id":28542,"name":"Eigil Obrestad","email":"eigil@obrestad.org","username":"obrestad"},"last_update":"2022-11-15 13:38:54.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-11-15 18:06:29.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-11-16 12:23:04.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_9708\u003e by clicking the attention icon","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}},"hashtags":[],"change_id":"I122fd0b9b9630a59059062e2186ac5fdd238e5bc","subject":"Compute Inventory Customization","status":"NEW","created":"2022-09-22 06:51:56.000000000","updated":"2022-11-16 12:23:04.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":31,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"bae24ad912c8899aed3a9712af70d206a2b5b38a","_number":858858,"virtual_id_number":858858,"owner":{"_account_id":28542,"name":"Eigil Obrestad","email":"eigil@obrestad.org","username":"obrestad"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-30 12:25:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":-1,"date":"2022-11-15 13:38:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-11-15 17:34:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-20 08:35:59.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-22 07:07:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-22 07:54:11.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-09-22 08:27:58.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-10-20 08:35:48.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"21ba06f641586998071f49306983c7e16e21dca8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28542,"name":"Eigil Obrestad","email":"eigil@obrestad.org","username":"obrestad"},"date":"2022-09-22 06:51:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b308049345ab5e575dbedbb0502f34d9154d612d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 07:07:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c6930bc753947aba2d78d0dea5aacf9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a8689f502414e40b65f3b488a9a3a86 : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b76ecff46c0743669cbcec3f320103d5 : FAILURE in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"100e9e267ed326e3a7cfc52f54f0538f34fc4a8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28542,"name":"Eigil Obrestad","email":"eigil@obrestad.org","username":"obrestad"},"date":"2022-09-22 07:41:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5afa1d364426afa411e1915b565c0a00123fddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 07:51:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50a41cd64f6a451a859679d740374bab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce4914373edd413ab8e131cf9576a9d2 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e56fa3741344e7c9ccaf4b7cc6ed70a : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"d4e713af1fa6161eaec718f0c161fc618e427829","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-09-22 07:54:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0d3f690b4410b941d30814969e3f732cd46dd4ba","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-09-22 08:27:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fe433521eb7dd6ffb38312c3d2582db91bad15e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28542,"name":"Eigil Obrestad","email":"eigil@obrestad.org","username":"obrestad"},"date":"2022-09-30 11:05:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8befc00567407aac76ee34cde8dd059850895365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 11:10:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76492dccc0ac40d7bcad138a923565a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc7302a5bf8448b193f75231cb5531f5 : FAILURE in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87726fcf00e345cc8771aea640c176b1 : FAILURE in 3m 08s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/2023.1/approved/compute-customizable-inventory.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/2023.1/approved/compute-customizable-inventory.rst","accounts_in_message":[],"_revision_number":3},{"id":"4134ee854c0c229d13f997407aa32b870db2ff08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28542,"name":"Eigil Obrestad","email":"eigil@obrestad.org","username":"obrestad"},"date":"2022-09-30 11:18:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ffe57bbdb284c135271605d8d33c2327777a3f75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 11:26:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b44ea6f7839432fa3399d29422d3256\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/572679d07af849b3a49ce2c8abd74807 : FAILURE in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b2b666edba1431faafe94c4d42c62bd : SUCCESS in 6m 02s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/2023.1/approved/compute-customizable-inventory.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/2023.1/approved/compute-customizable-inventory.rst","accounts_in_message":[],"_revision_number":4},{"id":"05a59e953eb7a41007c538f48c6f56639759be63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28542,"name":"Eigil Obrestad","email":"eigil@obrestad.org","username":"obrestad"},"date":"2022-09-30 12:17:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"54e1df41d8a1c2cbda0ed16763696bef1aec8e12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 12:25:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/115353a777774d208dbaefeeb7b8e40d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab155d7e8d24e5f8466e269bc9eba07 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64148449b20b46239a66e9ac6f058d16 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":5},{"id":"892dc72e1d13ebf28065c759a5d6b87320292d52","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-10-20 08:35:59.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"301207934196dd3ea474110b34181fe1a0d775ce","author":{"_account_id":28542,"name":"Eigil Obrestad","email":"eigil@obrestad.org","username":"obrestad"},"date":"2022-11-07 11:15:41.000000000","message":"Patch Set 5:\n\n(9 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3210b0e93f5644df70993bfd8eb1a82c2ff9adc8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-11-15 13:38:54.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"dd16b5b60805e04af4452fbc937e06878b3758a7","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-11-15 17:34:46.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"380592edc47a5a5e34cd967954a68e182176ffce","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-15 18:06:29.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"44adb7a1414dd58930ab94ec3d0824ff7618fdd0","revisions":{"4381cc9fe3e9bfc82038b9577f53c2d1a415381a":{"kind":"REWORK","_number":1,"created":"2022-09-22 06:51:56.000000000","uploader":{"_account_id":28542,"name":"Eigil Obrestad","email":"eigil@obrestad.org","username":"obrestad"},"ref":"refs/changes/58/858858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/58/858858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/1"}}},"commit":{"parents":[{"commit":"0d0000f0939e8220b9d580d46e19fc8f022dea51","subject":"Create specs directory for 2023.1 Antelope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d0000f0939e8220b9d580d46e19fc8f022dea51"}]}],"author":{"name":"Eigil Obrestad","email":"eigil-git@obrestad.org","date":"2022-09-22 06:51:20.000000000","tz":120},"committer":{"name":"Eigil Obrestad","email":"eigil-git@obrestad.org","date":"2022-09-22 06:51:20.000000000","tz":120},"subject":"Compute Inventory Customization","message":"Compute Inventory Customization\n\nhttps://blueprints.launchpad.net/nova/+spec/compute-customizable-inventory\n\nThe compute-nodes register their inventory in placement automaticly. The\ntypical inventory are VCPU, MEMORY_MB and DISK_GB with values for total amount,\nminimal reservations, maximal reservations, allocation_ratio and step_size.\nThis blueprints proposes to use provider.yaml and the placement API to set\nthese values.\n\nChange-Id: I122fd0b9b9630a59059062e2186ac5fdd238e5bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4381cc9fe3e9bfc82038b9577f53c2d1a415381a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4381cc9fe3e9bfc82038b9577f53c2d1a415381a"}]},"branch":"refs/heads/master"},"1275a41bbea3988e5f80bddfe7c1f9ab21195841":{"kind":"REWORK","_number":2,"created":"2022-09-22 07:41:38.000000000","uploader":{"_account_id":28542,"name":"Eigil Obrestad","email":"eigil@obrestad.org","username":"obrestad"},"ref":"refs/changes/58/858858/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/58/858858/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/2"}}},"commit":{"parents":[{"commit":"0d0000f0939e8220b9d580d46e19fc8f022dea51","subject":"Create specs directory for 2023.1 Antelope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d0000f0939e8220b9d580d46e19fc8f022dea51"}]}],"author":{"name":"Eigil Obrestad","email":"eigil-git@obrestad.org","date":"2022-09-22 06:51:20.000000000","tz":120},"committer":{"name":"Eigil Obrestad","email":"eigil-git@obrestad.org","date":"2022-09-22 07:41:29.000000000","tz":120},"subject":"Compute Inventory Customization","message":"Compute Inventory Customization\n\nhttps://blueprints.launchpad.net/nova/+spec/compute-customizable-inventory\n\nThe compute-nodes register their inventory in placement automaticly. The\ntypical inventory are VCPU, MEMORY_MB and DISK_GB with values for total amount,\nminimal reservations, maximal reservations, allocation_ratio and step_size.\nThis blueprints proposes to use provider.yaml and the placement API to set\nthese values.\n\nChange-Id: I122fd0b9b9630a59059062e2186ac5fdd238e5bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1275a41bbea3988e5f80bddfe7c1f9ab21195841"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1275a41bbea3988e5f80bddfe7c1f9ab21195841"}]},"branch":"refs/heads/master"},"392c8c35d6340d0bcdc85d170f1fb1b7edfba1fe":{"kind":"REWORK","_number":3,"created":"2022-09-30 11:05:37.000000000","uploader":{"_account_id":28542,"name":"Eigil Obrestad","email":"eigil@obrestad.org","username":"obrestad"},"ref":"refs/changes/58/858858/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/58/858858/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/3"}}},"commit":{"parents":[{"commit":"0d0000f0939e8220b9d580d46e19fc8f022dea51","subject":"Create specs directory for 2023.1 Antelope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d0000f0939e8220b9d580d46e19fc8f022dea51"}]}],"author":{"name":"Eigil Obrestad","email":"eigil-git@obrestad.org","date":"2022-09-22 06:51:20.000000000","tz":120},"committer":{"name":"Eigil Obrestad","email":"eigil-git@obrestad.org","date":"2022-09-30 11:05:30.000000000","tz":120},"subject":"Compute Inventory Customization","message":"Compute Inventory Customization\n\nhttps://blueprints.launchpad.net/nova/+spec/compute-customizable-inventory\n\nThe compute-nodes register their inventory in placement automaticly. The\ntypical inventory are VCPU, MEMORY_MB and DISK_GB with values for total amount,\nminimal reservations, maximal reservations, allocation_ratio and step_size.\nThis blueprints proposes to use provider.yaml and the placement API to set\nthese values.\n\nChange-Id: I122fd0b9b9630a59059062e2186ac5fdd238e5bc\nblueprint: compute-customizable-inventory\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/392c8c35d6340d0bcdc85d170f1fb1b7edfba1fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/392c8c35d6340d0bcdc85d170f1fb1b7edfba1fe"}]},"branch":"refs/heads/master"},"e65c58767d3080d8351454befd0deafdf96c6303":{"kind":"REWORK","_number":4,"created":"2022-09-30 11:18:51.000000000","uploader":{"_account_id":28542,"name":"Eigil Obrestad","email":"eigil@obrestad.org","username":"obrestad"},"ref":"refs/changes/58/858858/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/58/858858/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/4"}}},"commit":{"parents":[{"commit":"0d0000f0939e8220b9d580d46e19fc8f022dea51","subject":"Create specs directory for 2023.1 Antelope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d0000f0939e8220b9d580d46e19fc8f022dea51"}]}],"author":{"name":"Eigil Obrestad","email":"eigil-git@obrestad.org","date":"2022-09-22 06:51:20.000000000","tz":120},"committer":{"name":"Eigil Obrestad","email":"eigil-git@obrestad.org","date":"2022-09-30 11:18:43.000000000","tz":120},"subject":"Compute Inventory Customization","message":"Compute Inventory Customization\n\nhttps://blueprints.launchpad.net/nova/+spec/compute-customizable-inventory\n\nThe compute-nodes register their inventory in placement automaticly. The\ntypical inventory are VCPU, MEMORY_MB and DISK_GB with values for total amount,\nminimal reservations, maximal reservations, allocation_ratio and step_size.\nThis blueprints proposes to use provider.yaml and the placement API to set\nthese values.\n\nChange-Id: I122fd0b9b9630a59059062e2186ac5fdd238e5bc\nblueprint: compute-customizable-inventory\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e65c58767d3080d8351454befd0deafdf96c6303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e65c58767d3080d8351454befd0deafdf96c6303"}]},"branch":"refs/heads/master"},"44adb7a1414dd58930ab94ec3d0824ff7618fdd0":{"kind":"REWORK","_number":5,"created":"2022-09-30 12:17:06.000000000","uploader":{"_account_id":28542,"name":"Eigil Obrestad","email":"eigil@obrestad.org","username":"obrestad"},"ref":"refs/changes/58/858858/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/58/858858/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/58/858858/5"}}},"commit":{"parents":[{"commit":"0d0000f0939e8220b9d580d46e19fc8f022dea51","subject":"Create specs directory for 2023.1 Antelope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d0000f0939e8220b9d580d46e19fc8f022dea51"}]}],"author":{"name":"Eigil Obrestad","email":"eigil-git@obrestad.org","date":"2022-09-22 06:51:20.000000000","tz":120},"committer":{"name":"Eigil Obrestad","email":"eigil-git@obrestad.org","date":"2022-09-30 12:16:59.000000000","tz":120},"subject":"Compute Inventory Customization","message":"Compute Inventory Customization\n\nhttps://blueprints.launchpad.net/nova/+spec/compute-customizable-inventory\n\nThe compute-nodes register their inventory in placement automaticly. The\ntypical inventory are VCPU, MEMORY_MB and DISK_GB with values for total amount,\nminimal reservations, maximal reservations, allocation_ratio and step_size.\nThis blueprints proposes to use provider.yaml and the placement API to set\nthese values.\n\nChange-Id: I122fd0b9b9630a59059062e2186ac5fdd238e5bc\nblueprint: compute-customizable-inventory\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/44adb7a1414dd58930ab94ec3d0824ff7618fdd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/44adb7a1414dd58930ab94ec3d0824ff7618fdd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
