)]}'
{"id":"openstack%2Fnova-specs~861591","triplet_id":"openstack%2Fnova-specs~master~I529100701b4e37838894913613d485fa5ecc9d63","project":"openstack/nova-specs","branch":"master","topic":"external-cpu-management","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-11-28 11:17:53.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-11-28 10:43:46.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2022-11-28 11:27:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I529100701b4e37838894913613d485fa5ecc9d63","subject":"Proposes cpu power managment in libvirt","status":"MERGED","created":"2022-10-17 09:17:19.000000000","updated":"2022-11-28 11:28:38.000000000","submitted":"2022-11-28 11:27:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":48,"unresolved_comment_count":10,"has_review_started":true,"submission_id":"861591-external-cpu-management","meta_rev_id":"d105e69357c567b50d1b0263c9193c4432b33947","_number":861591,"virtual_id_number":861591,"owner":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-28 11:27:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2022-11-28 11:17:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2022-11-28 10:43:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2022-11-28 11:17:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-17 09:40:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-17 11:21:16.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2022-11-15 11:10:43.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2022-11-15 11:10:46.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-11-15 16:13:44.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"7ddbbe91879b17305f85bd1db008443d40ec6417","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-10-17 09:17:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c6bccfd21b695acc20796ab8b6df453afe8a9ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-10-17 09:19:10.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"60d99913578efe4a2c8355d2099c18fb62071ae2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-10-17 09:30:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e1cacb7fe425a6797edd039972590db6ed4237d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 09:40:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d253d8fa67a4c5f91f057e4065f35a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e22bdc01776e42feaca47ca0a3a2d4f6 : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43fa103059c64cfc8bddc63fb759e95e : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":3},{"id":"ca4fd0c80070e9238214d9b93dd2fe4c3ed5cb11","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-10-17 11:21:16.000000000","message":"Patch Set 3:\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ee8a830a945ce4b6b0b16c36c0d6c36f7328479c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-11-15 11:10:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"97e63d6db6423cac24e34ffa4c5cc3ded7edc3df","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-11-15 11:10:46.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"1b0500af2f796bfd14aa26eaec0e2f96f844babb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-15 12:04:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"70cbdb47087343469ab913f983006b8833053b6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-11-15 13:28:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0afa537e5c33d0e370dda3f75858dc948f44e418","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 13:40:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e2dd2af53384f2c868da90c968fb721\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f486a1b11214313aec1f041d1343041 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfbbb6d972274217b362f056df98aadd : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":4},{"id":"90c305346b2b48fe3c348d33972b3e88a2dbee4e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-11-15 16:05:59.000000000","message":"Patch Set 4: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"71ee025c9c3e590e824eeef8bee2564bb14fb610","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-15 16:13:44.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a0637a6a640c6689924a690bb5d319649ffdfe73","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-15 16:23:38.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"56cbcac3f5c1c779b7787776f037ebf690484332","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-11-15 16:35:13.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a95f5077819ab7867a72608c8bd41b9c8b2c2138","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-11-16 16:43:27.000000000","message":"Patch Set 4: Workflow-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5943a055152b24dd9bd0d8f01d76580473e9a6e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-11-16 16:43:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c8d2558245eb58423c44def6054ebb153b96779a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 16:53:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/646e4c697f9b4225aa0cbd5c1fccdc32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea7ba779fb924b1fb44fef37ee9546e5 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27bdc6114d1447019831d5e763498299 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":5},{"id":"c6776a4762eeec4d6f92da914cf4df913c0a0ce9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-11-16 17:19:59.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3a55b8e33e215cffd532bccd32cb0eff30735418","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-16 17:41:47.000000000","message":"Patch Set 5: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2636209bd7988571357561730c5f845e67523d31","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-11-25 15:12:13.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9acd912ce734df146485f4943e0d3748ecc8034d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-11-25 15:13:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"15ebff2c9b2de3fe76ebfeb15f2c69e7b3555ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 15:28:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c72d134838144e249f9a2a6263891f39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fd4c4da722a43ef80b511adf3d391c3 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3aff15cbf0e480db3eb56a9bde35dcf : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":6},{"id":"00580c38105848c0661ce36a75cd6992c958ace4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-11-28 10:43:46.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0ad706c0240c8a3ca3724a4b43f15584bbf4c2da","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-28 11:17:53.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d75d9b769b6da8cf43dce1a4bb0dab8041e09b6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 11:18:13.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f34542cac1707bd243073285b71230ac8a01889b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 11:27:08.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9447c12524fa461dbd44d2d467675829\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/580563f9c2c44ad19efa1277b0f1b181 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ad426f65a174ff09d5aa931a6945232 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":6},{"id":"8445dd00f7f05cc596d554484c0dc8e0c15fa6f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 11:27:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"d105e69357c567b50d1b0263c9193c4432b33947","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 11:28:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60a180c6892f444cbf200dabc697bbeb\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/e42529e526794886a5276c1ce8b002c3 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bba19dc7939198528f9b3583472691d997860e03","revisions":{"ed3310a372d9d865944eba2777e43f706aca1764":{"kind":"REWORK","_number":1,"created":"2022-10-17 09:17:19.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/91/861591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/861591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/1"}}},"commit":{"parents":[{"commit":"883ce45d73c3a2fb9260d3469fca012593a91381","subject":"Move specs to implemented","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/883ce45d73c3a2fb9260d3469fca012593a91381"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2021-12-10 15:51:16.000000000","tz":0},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2022-10-17 09:16:47.000000000","tz":120},"subject":"Add spec for external cpu power managment","message":"Add spec for external cpu power managment\n\nChange-Id: I529100701b4e37838894913613d485fa5ecc9d63\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ed3310a372d9d865944eba2777e43f706aca1764"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ed3310a372d9d865944eba2777e43f706aca1764"}]},"branch":"refs/heads/master"},"f25893955fe6fe02100ce285347249cd13118737":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-10-17 09:19:10.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/91/861591/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/861591/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/2"}}},"commit":{"parents":[{"commit":"883ce45d73c3a2fb9260d3469fca012593a91381","subject":"Move specs to implemented","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/883ce45d73c3a2fb9260d3469fca012593a91381"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2021-12-10 15:51:16.000000000","tz":0},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2022-10-17 09:18:00.000000000","tz":120},"subject":"Proposes external cpu power managment","message":"Proposes external cpu power managment\n\nThis spec describes blueprint external-cpu-management\n\nChange-Id: I529100701b4e37838894913613d485fa5ecc9d63\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f25893955fe6fe02100ce285347249cd13118737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f25893955fe6fe02100ce285347249cd13118737"}]},"branch":"refs/heads/master"},"da67beac37b0dfe7ee263284e9de3b03f9f16299":{"kind":"REWORK","_number":3,"created":"2022-10-17 09:30:32.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/91/861591/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/861591/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/3"}}},"commit":{"parents":[{"commit":"883ce45d73c3a2fb9260d3469fca012593a91381","subject":"Move specs to implemented","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/883ce45d73c3a2fb9260d3469fca012593a91381"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2021-12-10 15:51:16.000000000","tz":0},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2022-10-17 09:30:05.000000000","tz":120},"subject":"Proposes external cpu power managment","message":"Proposes external cpu power managment\n\nThis spec describes blueprint external-cpu-management\n\nChange-Id: I529100701b4e37838894913613d485fa5ecc9d63\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/da67beac37b0dfe7ee263284e9de3b03f9f16299"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/da67beac37b0dfe7ee263284e9de3b03f9f16299"}]},"branch":"refs/heads/master"},"4edf34adcc30edf1f8d95ea237f930a76f3b8e8c":{"kind":"REWORK","_number":4,"created":"2022-11-15 13:28:50.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/91/861591/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/861591/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/4"}}},"commit":{"parents":[{"commit":"883ce45d73c3a2fb9260d3469fca012593a91381","subject":"Move specs to implemented","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/883ce45d73c3a2fb9260d3469fca012593a91381"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2021-12-10 15:51:16.000000000","tz":0},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2022-11-15 13:28:12.000000000","tz":60},"subject":"Proposes cpu power managment in libvirt","message":"Proposes cpu power managment in libvirt\n\nThis spec describes blueprint libvirt-cpu-state-mgmt\n\nChange-Id: I529100701b4e37838894913613d485fa5ecc9d63\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4edf34adcc30edf1f8d95ea237f930a76f3b8e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4edf34adcc30edf1f8d95ea237f930a76f3b8e8c"}]},"branch":"refs/heads/master"},"4d8daf7db868af8e3c9d8e8e6a20c4c869f492a5":{"kind":"REWORK","_number":5,"created":"2022-11-16 16:43:51.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/91/861591/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/861591/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/5"}}},"commit":{"parents":[{"commit":"883ce45d73c3a2fb9260d3469fca012593a91381","subject":"Move specs to implemented","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/883ce45d73c3a2fb9260d3469fca012593a91381"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2021-12-10 15:51:16.000000000","tz":0},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2022-11-16 16:43:35.000000000","tz":60},"subject":"Proposes cpu power managment in libvirt","message":"Proposes cpu power managment in libvirt\n\nThis spec describes blueprint libvirt-cpu-state-mgmt\n\nChange-Id: I529100701b4e37838894913613d485fa5ecc9d63\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4d8daf7db868af8e3c9d8e8e6a20c4c869f492a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4d8daf7db868af8e3c9d8e8e6a20c4c869f492a5"}]},"branch":"refs/heads/master"},"bba19dc7939198528f9b3583472691d997860e03":{"kind":"REWORK","_number":6,"created":"2022-11-25 15:13:06.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/91/861591/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/861591/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/861591/6"}}},"commit":{"parents":[{"commit":"883ce45d73c3a2fb9260d3469fca012593a91381","subject":"Move specs to implemented","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/883ce45d73c3a2fb9260d3469fca012593a91381"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2021-12-10 15:51:16.000000000","tz":0},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2022-11-25 15:12:58.000000000","tz":60},"subject":"Proposes cpu power managment in libvirt","message":"Proposes cpu power managment in libvirt\n\nThis spec describes blueprint libvirt-cpu-state-mgmt\n\nChange-Id: I529100701b4e37838894913613d485fa5ecc9d63\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bba19dc7939198528f9b3583472691d997860e03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bba19dc7939198528f9b3583472691d997860e03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
