)]}'
{"id":"openstack%2Fnova-specs~861881","triplet_id":"openstack%2Fnova-specs~master~Ic27781ca5c5e1dc4020524592d0ff9e712ae0c6b","project":"openstack/nova-specs","branch":"master","topic":"bp/virtiofs-scaphandre","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-01-11 13:27:12.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2023-01-12 12:49:36.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-01-12 12:59:27.000000000","reason":"Change was submitted"},"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2023-01-12 12:59:27.000000000","reason":"Change was submitted"},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2023-01-12 12:59:27.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-01-12 12:59:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic27781ca5c5e1dc4020524592d0ff9e712ae0c6b","subject":"Allow local scaphandre directory to be mapped to an instance using virtiofs","status":"MERGED","created":"2022-10-19 12:27:48.000000000","updated":"2023-01-12 13:00:52.000000000","submitted":"2023-01-12 12:59:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":66,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"861881-bp/virtiofs-scaphandre","meta_rev_id":"cdcf73f85d302861971892b6eb92257d5013d53d","_number":861881,"virtual_id_number":861881,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-12 12:59:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2023-01-12 12:26:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2023-01-12 12:49:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-12 10:37:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2023-01-12 12:49:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2023-01-12 12:26:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"CC":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-19 12:33:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-19 15:10:49.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2022-11-15 14:16:02.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-11-15 14:44:45.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-11-15 17:32:48.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2023-01-10 11:04:44.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"}],"messages":[{"id":"6b7ce9811c724511cb2944f017336f1f78dd9777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-10-19 12:27:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3a16653b1bf53b42810363d13673a4d32110908","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 12:33:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f80412ab1d8548f996b534be891f4567\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08c572ae6f8045a781ae85c0d4b31fa2 : FAILURE in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f017c7b6d1034decb54862c461b249c0 : SUCCESS in 2m 35s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/zed/approved/virtiofs_scaphandre.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/zed/approved/virtiofs_scaphandre.rst","accounts_in_message":[],"_revision_number":1},{"id":"f77b94fd1be16ecee8d26437a8fb4e1d05e85d4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-10-19 12:38:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d03a5c41eac468293a90356e1897079fc1bfc7ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 12:42:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c27fb7d8f3bc403d910c39ed4531e86a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95d396d14227447a8b29dc67d582af7e : FAILURE in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77af4d2c971c4778ad0a27a2548d2345 : SUCCESS in 2m 54s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/zed/approved/virtiofs_scaphandre.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/zed/approved/virtiofs_scaphandre.rst","accounts_in_message":[],"_revision_number":2},{"id":"b060ccc9a2d726aa90b7d2ab4be63dfd7f832c86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-10-20 12:33:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"02d3a13f520aada7af6c23f04b56e62c44932204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-10-20 12:37:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c1d08843289cee20157ee5025c435af530b37d46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 12:46:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57b63d73ccb04003be4a18c04cf56c97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c506ce1e530d485da410d8fd5dbdfbb7 : FAILURE in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36b9d49518e54423a54afe8a1a6d2dee : FAILURE in 2m 39s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/2023.1/approved/virtiofs_scaphandre.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/2023.1/approved/virtiofs_scaphandre.rst","accounts_in_message":[],"_revision_number":4},{"id":"48abc0ae217d1b55d0107abcd28fb98079545eb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-11-14 15:38:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"558d7d9d0224a3e4f48eafdaf0fff6e546df4a01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 15:46:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae90a6ccbb2a4d4685448e18b2a53121\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/571aca75c5384da6aa11ead3536c6a6d : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa3236d797d7428ca3b1f131eac5cea4 : FAILURE in 2m 30s","accounts_in_message":[],"_revision_number":5},{"id":"c208ac339b46109e8959cca471d8bd17ed4cbf85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2022-11-14 16:14:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e71eb82e97b03fbdbf7df878eea48e3b52effec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 16:22:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9db327a9c43b41f1896ce97b8132f8a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b48b886c52f0489893a0286614c76252 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb20d0b091ee44e7b6cf6a8685f974e1 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":6},{"id":"68f62aab9dada4e18aa471389f0c9cbca797f2ce","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-11-15 14:16:02.000000000","message":"Patch Set 6: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2eb2c2f05073811b79b188b46ae741d742050933","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-15 14:44:45.000000000","message":"Patch Set 6: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2bbcfde25853090f6713dbbff56fa6e62cd61bac","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-11-15 16:08:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fa4da16e7faba4b26969a761372eabcf53b59339","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-11-15 17:32:48.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e2c4daa9d6f39fbb746d8e9c2f4891e73b573586","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-11-16 09:46:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"38d35fdf28048549c66d0cc99569bebea941d1b1","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-01-03 13:45:15.000000000","message":"Patch Set 6:\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2372a8c2eae631482e1f42703efbcad5c8ea1652","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-01-03 13:54:42.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"517af9fe82759d110c85e70ec67652947e18dd2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 14:05:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6d1c04f7fb54172b5a34c4a34d577b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f8e79d102a4ec6b8c94115b00f96db : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/971ef6218797495aa5952c2b8995885a : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":7},{"id":"5166bd5604ec3dcd33bf216d8a4c0ba443f689f1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-09 21:16:46.000000000","message":"Patch Set 7: Review-Priority+1","accounts_in_message":[],"_revision_number":7},{"id":"64f87b2e3122f6ecc74668fcacfca7fd3c90ccdc","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-09 21:32:10.000000000","message":"Patch Set 7: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":7},{"id":"0dfde5c533777e9aeba54536f2e424a83421463a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-01-10 10:38:12.000000000","message":"Patch Set 7: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"61b8496c4f73729be2afb7ff35e5e4f10297253f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-01-10 11:04:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ef0e5c5c00353c5329d5e9a6c36847a8fa45007a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-10 11:34:32.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"bdf1c3fbbe141c6b118168c6221e902499f23d8c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-10 11:35:30.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"af2c5a9aeae7fcca978e12fb7c8fd63d649ceacc","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-01-10 11:42:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d95c6fdb252ead8297ba71afd3892ec4c8e48e98","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-10 13:26:19.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"9f2b4a09edfeaa2ea2679d20c6f28df97b3ccd43","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-01-10 13:47:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"17ccbb31f482fbd0361a35c222de24c6f64306ac","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-01-10 14:04:35.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"30763dc7f83302b7e86fb404c54d9d57adb404cf","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-01-11 10:42:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c0c0541685cd79109c035e0f57159d8532e418b1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-11 12:02:28.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"786cba5a7e5c1725d80ccefaefccf918b670bb3f","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-01-11 13:12:40.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"488f09deb53fbfa3619708aa1506acc11641f325","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-11 13:27:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7348bb3ce73ee7969599b68d2f889b7a96a44def","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-01-11 15:21:33.000000000","message":"Patch Set 7:\n\n(10 comments)","accounts_in_message":[],"_revision_number":7},{"id":"43888ecc96a01c6f2b9061e71c007fec744307fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-01-11 15:25:35.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"16c24033a523fc0d635534dfa0fd1b8046631a4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-01-11 15:31:15.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8abdd8c781a19cf3e9f17a2281d6512c76c6cfae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 15:39:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d91e4f3c83e14c7286c57c8a999b2d0c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88d3a3d44cd74731938f0bb1b70d7a84 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd10c7e6d94a4cd8a7f3d61f05fe84f5 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":9},{"id":"b39f9e31a8040f7db5a0d04c78020680b04978cb","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-01-12 10:37:04.000000000","message":"Patch Set 9: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"bd7d9cc82cb0564be3226b3f6b72af128c1cc765","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-12 12:26:46.000000000","message":"Patch Set 9: Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"8c633650c757e86d415ddd913bed99b30c2596a2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-01-12 12:49:36.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"0a8ae5c7ee0c770e349ed9f9e9eae2b858b02228","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:49:57.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"b6c8c15e2e728400df208f3df2d4b363553db858","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:59:27.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bce38ba2fb046c8a00949eb5716cc4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1762420de6e14221826ccc28876e0b05 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06d589561a9a4bbd83c5e5b6f5b9f3a7 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":9},{"id":"fd3cd63b8725af8d2caa3f2d4f44bb0af354d927","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:59:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"cdcf73f85d302861971892b6eb92257d5013d53d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 13:00:52.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f0d92a7410f4f188c8099783126bd9f\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/136311c6b94340fb86215a6df656af09 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"84f2d76e9df5207ae05bbdaf7861c654b339be31","revisions":{"edc84bbd79d704a6656ca29e23d8030f6e9ac3d4":{"kind":"REWORK","_number":1,"created":"2022-10-19 12:27:48.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/81/861881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/861881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/1"}}},"commit":{"parents":[{"commit":"883ce45d73c3a2fb9260d3469fca012593a91381","subject":"Move specs to implemented","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/883ce45d73c3a2fb9260d3469fca012593a91381"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-10-19 12:25:59.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-10-19 12:25:59.000000000","tz":120},"subject":"Draft spec to support scaphandre","message":"Draft spec to support scaphandre\n\nSimplifi usage of scaphandre agent. Allowing users to reports VM\nconsumption and VM processes consumption.\n\nChange-Id: Ic27781ca5c5e1dc4020524592d0ff9e712ae0c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/edc84bbd79d704a6656ca29e23d8030f6e9ac3d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/edc84bbd79d704a6656ca29e23d8030f6e9ac3d4"}]},"branch":"refs/heads/master"},"67e3b3c389f7121a9d17f0443079a7997eccf986":{"kind":"REWORK","_number":2,"created":"2022-10-19 12:38:45.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/81/861881/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/861881/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/2"}}},"commit":{"parents":[{"commit":"883ce45d73c3a2fb9260d3469fca012593a91381","subject":"Move specs to implemented","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/883ce45d73c3a2fb9260d3469fca012593a91381"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-10-19 12:25:59.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-10-19 12:38:06.000000000","tz":120},"subject":"Draft spec to support scaphandre","message":"Draft spec to support scaphandre\n\nSimplify usage of scaphandre agent. Allowing users to reports VM\nconsumption and VM processes consumption.\n\nChange-Id: Ic27781ca5c5e1dc4020524592d0ff9e712ae0c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/67e3b3c389f7121a9d17f0443079a7997eccf986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/67e3b3c389f7121a9d17f0443079a7997eccf986"}]},"branch":"refs/heads/master"},"44c8a7668fb49132f99d2a247719ea0a75c3778a":{"kind":"REWORK","_number":3,"created":"2022-10-20 12:33:02.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/81/861881/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/861881/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/3"}}},"commit":{"parents":[{"commit":"883ce45d73c3a2fb9260d3469fca012593a91381","subject":"Move specs to implemented","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/883ce45d73c3a2fb9260d3469fca012593a91381"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-10-19 12:25:59.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-10-20 12:32:56.000000000","tz":120},"subject":"Draft spec to support scaphandre","message":"Draft spec to support scaphandre\n\nSimplify usage of scaphandre agent. Allowing users to reports VM\nconsumption and VM processes consumption.\n\nChange-Id: Ic27781ca5c5e1dc4020524592d0ff9e712ae0c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/44c8a7668fb49132f99d2a247719ea0a75c3778a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/44c8a7668fb49132f99d2a247719ea0a75c3778a"}]},"branch":"refs/heads/master"},"894cded4bc2715653a21ed1e67e4c8cf1e5c5508":{"kind":"REWORK","_number":4,"created":"2022-10-20 12:37:29.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/81/861881/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/861881/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/4"}}},"commit":{"parents":[{"commit":"883ce45d73c3a2fb9260d3469fca012593a91381","subject":"Move specs to implemented","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/883ce45d73c3a2fb9260d3469fca012593a91381"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-10-19 12:25:59.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-10-20 12:37:23.000000000","tz":120},"subject":"Draft spec to support scaphandre","message":"Draft spec to support scaphandre\n\nSimplify usage of scaphandre agent. Allowing users to reports VM\nconsumption and VM processes consumption.\n\nChange-Id: Ic27781ca5c5e1dc4020524592d0ff9e712ae0c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/894cded4bc2715653a21ed1e67e4c8cf1e5c5508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/894cded4bc2715653a21ed1e67e4c8cf1e5c5508"}]},"branch":"refs/heads/master"},"634cd9c481c7ff229722df507ff6b172c8925407":{"kind":"REWORK","_number":5,"created":"2022-11-14 15:38:47.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/81/861881/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/861881/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/5"}}},"commit":{"parents":[{"commit":"7ad4e88db2648b070db8505f2e47f6991a1b3c8b","subject":"Merge \"Re-propose PCI Device Tracking In Placement for A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7ad4e88db2648b070db8505f2e47f6991a1b3c8b"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-10-19 12:25:59.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-11-14 15:34:15.000000000","tz":60},"subject":"Allow local scaphandre directory to be mapped to an instance using virtiofs","message":"Allow local scaphandre directory to be mapped to an instance using virtiofs\n\nSimplify usage of scaphandre agent. Allowing users to reports VM\nconsumption and VM processes consumption.\n\nImplements: blueprint virtiofs-scaphandre\n\nChange-Id: Ic27781ca5c5e1dc4020524592d0ff9e712ae0c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/634cd9c481c7ff229722df507ff6b172c8925407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/634cd9c481c7ff229722df507ff6b172c8925407"}]},"branch":"refs/heads/master"},"c4b8d0e46a5feaf9d358157fee5b02fa457200aa":{"kind":"REWORK","_number":6,"created":"2022-11-14 16:14:54.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/81/861881/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/861881/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/6"}}},"commit":{"parents":[{"commit":"7ad4e88db2648b070db8505f2e47f6991a1b3c8b","subject":"Merge \"Re-propose PCI Device Tracking In Placement for A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7ad4e88db2648b070db8505f2e47f6991a1b3c8b"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-10-19 12:25:59.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-11-14 16:14:13.000000000","tz":60},"subject":"Allow local scaphandre directory to be mapped to an instance using virtiofs","message":"Allow local scaphandre directory to be mapped to an instance using virtiofs\n\nSimplify usage of scaphandre agent. Allowing users to reports VM\nconsumption and VM processes consumption.\n\nImplements: blueprint virtiofs-scaphandre\n\nChange-Id: Ic27781ca5c5e1dc4020524592d0ff9e712ae0c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c4b8d0e46a5feaf9d358157fee5b02fa457200aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c4b8d0e46a5feaf9d358157fee5b02fa457200aa"}]},"branch":"refs/heads/master"},"8f45e29dd6502ecf16b53ea5e0bd6a7e0e1e0675":{"kind":"REWORK","_number":7,"created":"2023-01-03 13:54:42.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/81/861881/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/861881/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/7"}}},"commit":{"parents":[{"commit":"7ad4e88db2648b070db8505f2e47f6991a1b3c8b","subject":"Merge \"Re-propose PCI Device Tracking In Placement for A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7ad4e88db2648b070db8505f2e47f6991a1b3c8b"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-10-19 12:25:59.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2023-01-03 13:53:27.000000000","tz":60},"subject":"Allow local scaphandre directory to be mapped to an instance using virtiofs","message":"Allow local scaphandre directory to be mapped to an instance using virtiofs\n\nSimplify usage of scaphandre agent. Allowing users to reports VM\nconsumption and VM processes consumption.\n\nImplements: blueprint virtiofs-scaphandre\n\nChange-Id: Ic27781ca5c5e1dc4020524592d0ff9e712ae0c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8f45e29dd6502ecf16b53ea5e0bd6a7e0e1e0675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8f45e29dd6502ecf16b53ea5e0bd6a7e0e1e0675"}]},"branch":"refs/heads/master"},"98f15740484fbe1eeed58cedd353da85e87d0e79":{"kind":"REWORK","_number":8,"created":"2023-01-11 15:25:35.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/81/861881/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/861881/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/8"}}},"commit":{"parents":[{"commit":"cfc9837aa86d75ab48303cd4903d79f4da84d5c0","subject":"Merge \"Ironic shard_key to replace peer_list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cfc9837aa86d75ab48303cd4903d79f4da84d5c0"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-10-19 12:25:59.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2023-01-11 15:25:00.000000000","tz":60},"subject":"Allow local scaphandre directory to be mapped to an instance using virtiofs","message":"Allow local scaphandre directory to be mapped to an instance using virtiofs\n\nSimplify usage of scaphandre agent. Allowing users to reports VM\nconsumption and VM processes consumption.\n\nImplements: blueprint virtiofs-scaphandre\n\nChange-Id: Ic27781ca5c5e1dc4020524592d0ff9e712ae0c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/98f15740484fbe1eeed58cedd353da85e87d0e79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/98f15740484fbe1eeed58cedd353da85e87d0e79"}]},"branch":"refs/heads/master"},"84f2d76e9df5207ae05bbdaf7861c654b339be31":{"kind":"REWORK","_number":9,"created":"2023-01-11 15:31:15.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/81/861881/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/861881/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/861881/9"}}},"commit":{"parents":[{"commit":"cfc9837aa86d75ab48303cd4903d79f4da84d5c0","subject":"Merge \"Ironic shard_key to replace peer_list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cfc9837aa86d75ab48303cd4903d79f4da84d5c0"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2022-10-19 12:25:59.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2023-01-11 15:31:02.000000000","tz":60},"subject":"Allow local scaphandre directory to be mapped to an instance using virtiofs","message":"Allow local scaphandre directory to be mapped to an instance using virtiofs\n\nSimplify usage of scaphandre agent. Allowing users to reports VM\nconsumption and VM processes consumption.\n\nImplements: blueprint virtiofs-scaphandre\n\nChange-Id: Ic27781ca5c5e1dc4020524592d0ff9e712ae0c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/84f2d76e9df5207ae05bbdaf7861c654b339be31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/84f2d76e9df5207ae05bbdaf7861c654b339be31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[]}
