)]}'
{"id":"openstack%2Fnova-specs~868377","triplet_id":"openstack%2Fnova-specs~master~Icc532cd8ad17fd2cb48de1b9821a6b1d0e346096","project":"openstack/nova-specs","branch":"master","topic":"bp/proposes","attention_set":{},"removed_from_attention_set":{"35489":{"account":{"_account_id":35489,"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","username":"dvo-plv"},"last_update":"2023-05-03 12:58:50.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-05-03 12:39:34.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"35239":{"account":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"last_update":"2023-12-01 07:56:35.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_35489\u003e by clicking the attention icon","reason_account":{"_account_id":35489,"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","username":"dvo-plv"}},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2023-05-03 12:58:50.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2023-05-03 12:42:56.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}},"hashtags":[],"change_id":"Icc532cd8ad17fd2cb48de1b9821a6b1d0e346096","subject":"VirtIO PackedRing Configuration support","status":"MERGED","created":"2022-12-22 07:57:32.000000000","updated":"2023-12-01 07:56:35.000000000","submitted":"2023-05-03 12:58:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"868377-bp/proposes","meta_rev_id":"11c733297132c5d9bec70c97076876df0e3ccddb","_number":868377,"virtual_id_number":868377,"owner":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-03 12:58:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2023-05-03 12:39:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-03 12:42:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-03 12:42:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2023-05-03 12:39:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":35489,"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","username":"dvo-plv"}],"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-22 08:10:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-02 05:35:50.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-03-16 09:31:12.000000000","updated_by":{"_account_id":35489,"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","username":"dvo-plv"},"reviewer":{"_account_id":35489,"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","username":"dvo-plv"},"state":"CC"},{"updated":"2023-04-21 10:53:28.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2023-04-26 09:15:47.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-05-30 12:38:02.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"CC"}],"messages":[{"id":"deb265dbf7f3ae522e7d557a9e6d437463521121","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2022-12-22 07:57:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eea113a4dfced6b55f02eb4a75ff237889dadf17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 08:10:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b0f7530368341f1b18156a94878d87e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16419a0b94874ab8b4d6b2534233ee21 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df3c9899184c4394bc52f124e4359c34 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"85d9c48d794a7b2d23cbf424b9a6a41f8c48690e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-02 05:35:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3fb6c0d0c5031b1c32bfeed7a046de5d83ce8890","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2023-01-02 08:32:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab7060fb8c0e85d3c0758f25602f4794f7de3f7e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-06 11:59:51.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c752bf7bf1985666cab38404f0dbb009885fb5b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2023-02-24 12:26:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9aee1a81d0ff9ccb6b9e44a343a54a7ce41828b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 12:34:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d654548a270c4ff383cfda33e2eb0bbe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b14546e537c547febd1f0a98d1f76aea : FAILURE in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a42a17e80845188325061aada5308a : FAILURE in 2m 38s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/2023.1/approved/virtio_packedring_configuration_support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/2023.1/approved/virtio_packedring_configuration_support.rst","accounts_in_message":[],"_revision_number":2},{"id":"75120784148563fed33ea4e2e49cdb5c2008231c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2023-03-02 14:20:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"73811b70b2bceaac016256b73ffd45ad6ec4febe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2023-03-02 14:24:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"93865de09b323a44db9e594a6c0a69dc9cbf350d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 14:35:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/46bc59d48ef54597b0e4ce800a585bd1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b08d81e919ad4d058b32d7e1e406ccb3 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8638852e7ad460685dc4e6a6547bbf0 : FAILURE in 4m 21s","accounts_in_message":[],"_revision_number":4},{"id":"9c30f8bcd419672ed374c5349d905351d00b594e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2023-03-02 14:37:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ee6c19e6a62d3f7f095d5bdabbd37656c72638f1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14510,"name":"Oleksii Butenko","email":"apbutenko@gmail.com","username":"obutenko"},"date":"2023-03-02 14:49:14.000000000","message":"Topic set to VirtIO_PackedRing","accounts_in_message":[],"_revision_number":5},{"id":"737e45bd13c08ea9c3b025bed585ee377cc7f6db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 14:50:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b8d1205a4964145b4811cd8b786f14f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1198b4339adb405cafb0be6a72e95ff1 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db007eacc0f84b3793e1797ab745d36a : FAILURE in 3m 08s","accounts_in_message":[],"_revision_number":5},{"id":"8a69aa56ff7ef5c7c14e57fa1a625aa4d45c16c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2023-03-02 14:52:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f37c731a40645de3c5f45de3d4085d11d85b7429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 15:02:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30b0a15490b04c809c00ef2dc29a4ee8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5707788a1b7949adbe77fbed157cf67f : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55e289b2236948f0b453ef7625e946f3 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":6},{"id":"4f55e16913bd3fcc15650423c5e418ab907411bc","author":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"date":"2023-03-03 08:17:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"56de7d0e5f4ebe04ab89b3cefc231de2eb059f81","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-03-15 13:55:27.000000000","message":"Patch Set 6: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"988065093c37f1677cc4d15a40fb6455705ac3cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35489,"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","username":"dvo-plv"},"date":"2023-03-16 09:30:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b4694f5d226e2a1611b960f0c90bdc09a433bc95","author":{"_account_id":35489,"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","username":"dvo-plv"},"date":"2023-03-16 09:31:12.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b994f21e4208d2f84fdad52cbd21066a6978aafe","author":{"_account_id":35489,"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","username":"dvo-plv"},"date":"2023-03-16 09:31:30.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"21eb3c7d7dee802c2d2b2495118b0e3efea4a783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 09:45:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5252da08e6fd4e43829b2153d7ffff77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a552fb4945ea447ba29445fc51ac229b : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b4016e35c1340d5b89235d34c00edf8 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":7},{"id":"c3db552c561edf66d3045f734bc7bed538662901","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-04-21 09:51:18.000000000","message":"Patch Set 7: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":7},{"id":"08cb09a6119fb48d7f3289258ca692e4aba97a3f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-04-21 09:52:35.000000000","message":"Topic bp/proposes removed","accounts_in_message":[],"_revision_number":7},{"id":"ba81fc5d91c4ffe5120bd2b692a0497d0a5dc4e5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-04-21 09:53:10.000000000","message":"Topic set to bp/virtio-packedring-configuration-support","accounts_in_message":[],"_revision_number":7},{"id":"bea635664ecad2f8315a76f814684c59c0050ada","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-04-21 10:53:28.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a325c6d24714022a16c32b0606f7a11899358ec7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-04-21 11:44:51.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ec7c1dffe98d705aa52e269637e26f2c947b70ce","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-04-26 09:15:47.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"fe1dfb5984f146a3e6535f1210244d9ddc89f30e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35489,"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","username":"dvo-plv"},"date":"2023-05-03 11:54:10.000000000","message":"Uploaded patch set 8.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"a521d990b6c2c728715bb8cd4b6a4592ec4dbb7e","author":{"_account_id":35489,"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","username":"dvo-plv"},"date":"2023-05-03 12:02:49.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"52f17b53f85f72178c79d44f4141e4b915c24566","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 12:06:06.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0437eceaa5ff4542a2d2af290aaf6b42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cd5cdc7ba8646be91d03d62a9830c82 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7fda3b09a194e188a5e450677d1180b : FAILURE in 3m 25s","accounts_in_message":[],"_revision_number":8},{"id":"43e992aa8c2a91fa13dc51261564294deddb147f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-05-03 12:18:13.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"730227d75918a6f187d97ae9c314a671e4e2213f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35489,"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","username":"dvo-plv"},"date":"2023-05-03 12:23:14.000000000","message":"Uploaded patch set 9.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"f416a8cebd5d92fd254426cd1474316fb9881e83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 12:31:36.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02624e7fa49841b895bb04e7f4556e79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8772064ea664a77b665aef1c53e4276 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6d0916b8c54dd68cb8b9ea3efb2f1b : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":9},{"id":"1b4698c3587275117bba58aba3017f4f2ed76b0a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-05-03 12:39:34.000000000","message":"Patch Set 9: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":9},{"id":"f987db1ccf239553f98b329dd8441ef167ff3c9f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-05-03 12:42:56.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a8fa84163ef55f9d7edf2628921875f85ae0284b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 12:43:16.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"de3dc8d7a40fad3f6649a2a71cea7894bf1d2428","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 12:58:50.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/291d6284b58b4d1c81eee3aecfc05a2b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1c35a429a5a47069564e361b307809e : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e850f8666a749f69fc0ceb761477385 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":9},{"id":"8eb48592217312268951c56094027d08666f0ce3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 12:58:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"d4facbe5343e4c1c42333b1258c6d817841eb675","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 13:00:07.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a94441d9f4044ac886364a393e30ba44\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/1cc35767b03c404fbc3f2353345e731a : SUCCESS in 57s","accounts_in_message":[],"_revision_number":9},{"id":"d74dc766abfe70f17562cfc76fe9c7f7a4404f17","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-05-30 12:38:02.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"b41d265f275e8e0ca70e8082357eab74db507460","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-05-30 13:03:02.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b455dc9787f2e0f75f08286a578fb383d5ed31b4","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-05-30 13:17:40.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"3a80f8a42fa8b976600885ffe8b91db1f2a988c3","revisions":{"46a9922419b4ebb5886f5847fa9039d888194bad":{"kind":"REWORK","_number":1,"created":"2022-12-22 07:57:32.000000000","uploader":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"ref":"refs/changes/77/868377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/77/868377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/1"}}},"commit":{"parents":[{"commit":"cfc9837aa86d75ab48303cd4903d79f4da84d5c0","subject":"Merge \"Ironic shard_key to replace peer_list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cfc9837aa86d75ab48303cd4903d79f4da84d5c0"}]}],"author":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-12-22 07:56:55.000000000","tz":60},"committer":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-12-22 07:57:28.000000000","tz":60},"subject":"VirtIO PackedRing Configuration support","message":"VirtIO PackedRing Configuration support\n\nThis blueprint proposes to expose the LibVirt `packed` option that allows a\nguest to negotiate support for the VirtIO packed-ring feature. This\nblueprint is used to solicit community\u0027s input.\n\nChange-Id: Icc532cd8ad17fd2cb48de1b9821a6b1d0e346096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/46a9922419b4ebb5886f5847fa9039d888194bad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/46a9922419b4ebb5886f5847fa9039d888194bad"}]},"branch":"refs/heads/master"},"359c8eb3d4c2ed4a06ef03846267b0e79b48f49e":{"kind":"REWORK","_number":2,"created":"2023-02-24 12:26:47.000000000","uploader":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"ref":"refs/changes/77/868377/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/77/868377/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/2"}}},"commit":{"parents":[{"commit":"cfc9837aa86d75ab48303cd4903d79f4da84d5c0","subject":"Merge \"Ironic shard_key to replace peer_list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cfc9837aa86d75ab48303cd4903d79f4da84d5c0"}]}],"author":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-12-22 07:56:55.000000000","tz":60},"committer":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2023-02-24 12:25:58.000000000","tz":60},"subject":"VirtIO PackedRing Configuration support","message":"VirtIO PackedRing Configuration support\n\nThis blueprint proposes to expose the LibVirt `packed` option that allows a\nguest to negotiate support for the VirtIO packed-ring feature. This\nblueprint is used to solicit community\u0027s input.\n\nChange-Id: Icc532cd8ad17fd2cb48de1b9821a6b1d0e346096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/359c8eb3d4c2ed4a06ef03846267b0e79b48f49e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/359c8eb3d4c2ed4a06ef03846267b0e79b48f49e"}]},"branch":"refs/heads/master"},"abda1103d9334a500ab0af0405991e6dc7bb8f41":{"kind":"REWORK","_number":3,"created":"2023-03-02 14:20:26.000000000","uploader":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"ref":"refs/changes/77/868377/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/77/868377/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/3"}}},"commit":{"parents":[{"commit":"b745c5e33a1e2b7f7f1d32f03797e36deb755661","subject":"Merge \"Create specs directory for 2023.2 Bobcat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b745c5e33a1e2b7f7f1d32f03797e36deb755661"}]}],"author":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-12-22 07:56:55.000000000","tz":60},"committer":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2023-03-02 14:20:12.000000000","tz":60},"subject":"VirtIO PackedRing Configuration support","message":"VirtIO PackedRing Configuration support\n\nThis blueprint proposes to expose the LibVirt `packed` option that allows a\nguest to negotiate support for the VirtIO packed-ring feature. This\nblueprint is used to solicit community\u0027s input.\n\nChange-Id: Icc532cd8ad17fd2cb48de1b9821a6b1d0e346096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/abda1103d9334a500ab0af0405991e6dc7bb8f41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/abda1103d9334a500ab0af0405991e6dc7bb8f41"}]},"branch":"refs/heads/master"},"9fdcf08cee86afe67097584857887dc2eadd3e9a":{"kind":"REWORK","_number":4,"created":"2023-03-02 14:24:37.000000000","uploader":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"ref":"refs/changes/77/868377/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/77/868377/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/4"}}},"commit":{"parents":[{"commit":"b745c5e33a1e2b7f7f1d32f03797e36deb755661","subject":"Merge \"Create specs directory for 2023.2 Bobcat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b745c5e33a1e2b7f7f1d32f03797e36deb755661"}]}],"author":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-12-22 07:56:55.000000000","tz":60},"committer":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2023-03-02 14:24:26.000000000","tz":60},"subject":"VirtIO PackedRing Configuration support","message":"VirtIO PackedRing Configuration support\n\nThis blueprint proposes to expose the LibVirt `packed` option that allows a\nguest to negotiate support for the VirtIO packed-ring feature. This\nblueprint is used to solicit community\u0027s input.\n\nChange-Id: Icc532cd8ad17fd2cb48de1b9821a6b1d0e346096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9fdcf08cee86afe67097584857887dc2eadd3e9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9fdcf08cee86afe67097584857887dc2eadd3e9a"}]},"branch":"refs/heads/master"},"5c5d5887744a7d12b16f8181e0eebd827cb4d5b0":{"kind":"REWORK","_number":5,"created":"2023-03-02 14:37:32.000000000","uploader":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"ref":"refs/changes/77/868377/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/77/868377/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/5"}}},"commit":{"parents":[{"commit":"b745c5e33a1e2b7f7f1d32f03797e36deb755661","subject":"Merge \"Create specs directory for 2023.2 Bobcat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b745c5e33a1e2b7f7f1d32f03797e36deb755661"}]}],"author":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-12-22 07:56:55.000000000","tz":60},"committer":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2023-03-02 14:37:22.000000000","tz":60},"subject":"VirtIO PackedRing Configuration support","message":"VirtIO PackedRing Configuration support\n\nThis blueprint proposes to expose the LibVirt `packed` option that allows a\nguest to negotiate support for the VirtIO packed-ring feature. This\nblueprint is used to solicit community\u0027s input.\n\nChange-Id: Icc532cd8ad17fd2cb48de1b9821a6b1d0e346096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5c5d5887744a7d12b16f8181e0eebd827cb4d5b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5c5d5887744a7d12b16f8181e0eebd827cb4d5b0"}]},"branch":"refs/heads/master"},"ea6635da3b93e80a40162ed277b8fc530595ab70":{"kind":"REWORK","_number":6,"created":"2023-03-02 14:52:34.000000000","uploader":{"_account_id":35239,"name":"Justas Poderys","email":"jpo@napatech.com","username":"justas_napa"},"ref":"refs/changes/77/868377/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/77/868377/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/6"}}},"commit":{"parents":[{"commit":"b745c5e33a1e2b7f7f1d32f03797e36deb755661","subject":"Merge \"Create specs directory for 2023.2 Bobcat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b745c5e33a1e2b7f7f1d32f03797e36deb755661"}]}],"author":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-12-22 07:56:55.000000000","tz":60},"committer":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2023-03-02 14:52:24.000000000","tz":60},"subject":"VirtIO PackedRing Configuration support","message":"VirtIO PackedRing Configuration support\n\nThis blueprint proposes to expose the LibVirt `packed` option that allows a\nguest to negotiate support for the VirtIO packed-ring feature. This\nblueprint is used to solicit community\u0027s input.\n\nChange-Id: Icc532cd8ad17fd2cb48de1b9821a6b1d0e346096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ea6635da3b93e80a40162ed277b8fc530595ab70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ea6635da3b93e80a40162ed277b8fc530595ab70"}]},"branch":"refs/heads/master"},"d382551d4bdf91e618453fe0b7716b03f655fbd6":{"kind":"REWORK","_number":7,"created":"2023-03-16 09:30:22.000000000","uploader":{"_account_id":35489,"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","username":"dvo-plv"},"ref":"refs/changes/77/868377/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/77/868377/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/7"}}},"commit":{"parents":[{"commit":"b745c5e33a1e2b7f7f1d32f03797e36deb755661","subject":"Merge \"Create specs directory for 2023.2 Bobcat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b745c5e33a1e2b7f7f1d32f03797e36deb755661"}]}],"author":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-12-22 07:56:55.000000000","tz":60},"committer":{"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","date":"2023-03-16 09:30:19.000000000","tz":120},"subject":"VirtIO PackedRing Configuration support","message":"VirtIO PackedRing Configuration support\n\nThis blueprint proposes to expose the LibVirt `packed` option that allows a\nguest to negotiate support for the VirtIO packed-ring feature. This\nblueprint is used to solicit community\u0027s input.\n\nChange-Id: Icc532cd8ad17fd2cb48de1b9821a6b1d0e346096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d382551d4bdf91e618453fe0b7716b03f655fbd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d382551d4bdf91e618453fe0b7716b03f655fbd6"}]},"branch":"refs/heads/master"},"4787148e4d68282b5bfbed277cc46087a3d8d220":{"kind":"REWORK","_number":8,"created":"2023-05-03 11:54:10.000000000","uploader":{"_account_id":35489,"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","username":"dvo-plv"},"ref":"refs/changes/77/868377/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/77/868377/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/8"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-12-22 07:56:55.000000000","tz":60},"committer":{"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","date":"2023-05-03 11:54:08.000000000","tz":180},"subject":"VirtIO PackedRing Configuration support","message":"VirtIO PackedRing Configuration support\n\nThis blueprint proposes to expose the LibVirt `packed` option that allows a\nguest to negotiate support for the VirtIO packed-ring feature. This\nblueprint is used to solicit community\u0027s input.\n\nChange-Id: Icc532cd8ad17fd2cb48de1b9821a6b1d0e346096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4787148e4d68282b5bfbed277cc46087a3d8d220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4787148e4d68282b5bfbed277cc46087a3d8d220"}]},"branch":"refs/heads/master"},"3a80f8a42fa8b976600885ffe8b91db1f2a988c3":{"kind":"REWORK","_number":9,"created":"2023-05-03 12:23:14.000000000","uploader":{"_account_id":35489,"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","username":"dvo-plv"},"ref":"refs/changes/77/868377/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/77/868377/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/77/868377/9"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Justas Poderys","email":"jpo@napatech.com","date":"2022-12-22 07:56:55.000000000","tz":60},"committer":{"name":"Danylo Vodopianov","email":"dvo-plv@napatech.com","date":"2023-05-03 12:23:04.000000000","tz":180},"subject":"VirtIO PackedRing Configuration support","message":"VirtIO PackedRing Configuration support\n\nThis blueprint proposes to expose the LibVirt `packed` option that allows a\nguest to negotiate support for the VirtIO packed-ring feature. This\nblueprint is used to solicit community\u0027s input.\n\nChange-Id: Icc532cd8ad17fd2cb48de1b9821a6b1d0e346096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3a80f8a42fa8b976600885ffe8b91db1f2a988c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3a80f8a42fa8b976600885ffe8b91db1f2a988c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
