)]}'
{"id":"openstack%2Fnova-specs~876707","triplet_id":"openstack%2Fnova-specs~master~I6ba8f5ef0c32126f3ea38087a4d671bed6032782","project":"openstack/nova-specs","branch":"master","topic":"bp/virtiofs_scaphandre","attention_set":{},"removed_from_attention_set":{"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-03-07 15:06:08.000000000","reason":"\u003cGERRIT_ACCOUNT_7166\u003e replied on the change","reason_account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2023-03-07 15:18:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ba8f5ef0c32126f3ea38087a4d671bed6032782","subject":"Re-propose \"Allow local scaphandre directory to be mapped to an instance using virtiofs\" for Bobcat","status":"MERGED","created":"2023-03-07 09:45:13.000000000","updated":"2023-03-07 15:20:02.000000000","submitted":"2023-03-07 15:18:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876707-bp/virtiofs_scaphandre","meta_rev_id":"acc91d75957346a734fd81ae26dea43311692b6a","_number":876707,"virtual_id_number":876707,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-07 15:18:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-07 15:06:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-07 15:06:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-07 09:55:22.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2023-03-07 10:21:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b83858254ad6ae9364ae0af6eda1fdd7e9d2d449","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-03-07 09:45:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f735737dd3a4039bbba2239a6bd2f563164ee2a2","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-03-07 09:55:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06bce7997807f93a32973bc2d037afa43aabcca8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-03-07 10:07:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e6daf0d28122f17597e848fb9ce97bd4d099645b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 10:21:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d9684f2473746199263454bdfb7a081\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4927597f4331409ab998775609c01ce0 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2828d230c4054e96aef6cf07b71f24b3 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2},{"id":"21af6824907296d67612644c3456b99e18d145e4","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-03-07 14:43:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b7ae78b94d0a463b982435df538353f3d4a4834e","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-03-07 15:06:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d3320caf751519202c139d0907c600ae8f012de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 15:06:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2063f5e797fd4552989039454e290310f0591515","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 15:18:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7da1eebf59164189bb6e7c307da5e934\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f6cf198df1c46db9f477ed382e9703c : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc39cb1f1236480da7e139e2c0d4cf77 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":2},{"id":"e737093a078d99fbb33819a0e4b2512727f0c56d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 15:18:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"acc91d75957346a734fd81ae26dea43311692b6a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 15:20:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1b352c193004368af9a547e1f5e9763\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/8a60f6250062441da198191aa119e098 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0a23bb20c2d7b5c72920bc8f0ae9440984e54092","revisions":{"63237226a62dc57d9eb92bfa35bf6631f58d5da0":{"kind":"REWORK","_number":1,"created":"2023-03-07 09:45:13.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/07/876707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/07/876707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/876707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/876707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/876707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/07/876707/1"}}},"commit":{"parents":[{"commit":"b745c5e33a1e2b7f7f1d32f03797e36deb755661","subject":"Merge \"Create specs directory for 2023.2 Bobcat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b745c5e33a1e2b7f7f1d32f03797e36deb755661"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2023-03-07 09:43:09.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2023-03-07 09:43:09.000000000","tz":60},"subject":"Re-propose \"Allow local scaphandre directory to be mapped to an instance using virtiofs\" for Bobcat","message":"Re-propose \"Allow local scaphandre directory to be mapped to an instance using virtiofs\"\nfor Bobcat\n\nImplements: blueprint virtiofs-scaphandre\nChange-Id: I6ba8f5ef0c32126f3ea38087a4d671bed6032782\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/63237226a62dc57d9eb92bfa35bf6631f58d5da0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/63237226a62dc57d9eb92bfa35bf6631f58d5da0"}]},"branch":"refs/heads/master"},"0a23bb20c2d7b5c72920bc8f0ae9440984e54092":{"kind":"REWORK","_number":2,"created":"2023-03-07 10:07:21.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/07/876707/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/07/876707/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/876707/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/876707/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/876707/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/07/876707/2"}}},"commit":{"parents":[{"commit":"b745c5e33a1e2b7f7f1d32f03797e36deb755661","subject":"Merge \"Create specs directory for 2023.2 Bobcat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b745c5e33a1e2b7f7f1d32f03797e36deb755661"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2023-03-07 10:06:27.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2023-03-07 10:06:27.000000000","tz":60},"subject":"Re-propose \"Allow local scaphandre directory to be mapped to an instance using virtiofs\" for Bobcat","message":"Re-propose \"Allow local scaphandre directory to be mapped to an instance using virtiofs\"\nfor Bobcat\n\nImplements: blueprint virtiofs-scaphandre\nChange-Id: I6ba8f5ef0c32126f3ea38087a4d671bed6032782\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0a23bb20c2d7b5c72920bc8f0ae9440984e54092"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0a23bb20c2d7b5c72920bc8f0ae9440984e54092"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
