)]}'
{"id":"openstack%2Fnova-specs~878066","triplet_id":"openstack%2Fnova-specs~master~Iade013228d5795311b5124285699cd706a9cf874","project":"openstack/nova-specs","branch":"master","topic":"virtio-scsi-multiqueue","attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2023-03-21 15:34:34.000000000","reason":"\u003cGERRIT_ACCOUNT_34443\u003e replied on the change","reason_account":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-03-21 14:24:44.000000000","reason":"\u003cGERRIT_ACCOUNT_34443\u003e replied on the change","reason_account":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"}}},"removed_from_attention_set":{"34443":{"account":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"last_update":"2023-03-21 15:34:34.000000000","reason":"\u003cGERRIT_ACCOUNT_34443\u003e replied on the change","reason_account":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"}}},"hashtags":[],"change_id":"Iade013228d5795311b5124285699cd706a9cf874","subject":"Support virtio-scsi multiqueue for improved I/O performance on nova\u0027s guests","status":"NEW","created":"2023-03-21 11:32:34.000000000","updated":"2023-03-22 09:26:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":10,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"c7d51dc0219d73c9815afc4654a4fb181a3783ee","_number":878066,"virtual_id_number":878066,"owner":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-22 09:26:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-21 11:47:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-21 14:24:44.000000000","updated_by":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-03-21 14:24:44.000000000","updated_by":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"}],"messages":[{"id":"43918aa60b54ad969027f3c5528f6cd70b7bbb27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"date":"2023-03-21 11:32:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bf1e8eb78cb1c329dc8582eb08c254717a83b90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"date":"2023-03-21 11:34:46.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"13e96770668921c01441dda924c7fb54739d425c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 11:47:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad48258370844b16adb9ae63ff9480e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6174542160a4a5495d7c3510bf708b0 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89f06b737b0d4237a38a5ea4e95b7964 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":2},{"id":"29adf54ca5f1857d80f4b67114ef8cc6d0890add","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"date":"2023-03-21 12:02:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"625c61fd571bf69d0bda46692b5e58cb81f0aa80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 12:14:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5bb0033041d4959a466cd63d1a5f701\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4479e963d2214958a1be152213729c26 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa1dd4492e5741d1bd15b934c201d8ac : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":3},{"id":"f390625d5c2799631910a94b46d044e2bc77861c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"date":"2023-03-21 14:23:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d23ca349b9e98879cd876a99d4d4262d6416effe","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"date":"2023-03-21 14:25:00.000000000","message":"Topic set to virtio-scsi-multiqueue","accounts_in_message":[],"_revision_number":4},{"id":"d74f94346c0d54fdb806de932f98eeca0401beb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 14:34:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be9678b93e484e6d8ca7be32683a3d25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f92ebee94d4084a2a791ac233d0991 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6e89f52b19f43c381a590408c2a90f3 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":4},{"id":"5c9d487a63763ec81c44367f173d3cfc680b2b3f","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2023-03-21 15:14:25.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8597daba8a6e0ff52ccecc2dce894c8d66831f72","author":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"date":"2023-03-21 15:34:34.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3a3897ee7f860d4901db10c7f8d83548b32f5c5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"date":"2023-03-22 09:16:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"96c0a3a52a0f3b3e04bfc9d0df698112c4c89c2c","author":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"date":"2023-03-22 09:17:52.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c7d51dc0219d73c9815afc4654a4fb181a3783ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 09:26:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c26ec53b1070429ea92f8b5a5de6043b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/524a8f73726d4302b6aad9755fdf4497 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e9079e24a9f4a158928caf034c82765 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"74b923e94d90fb170903c0aa1753d491b81c2994","revisions":{"7dedf72f84a6101861e99e2040b66d6bd802b0ba":{"kind":"REWORK","_number":1,"created":"2023-03-21 11:32:34.000000000","uploader":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"ref":"refs/changes/66/878066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/66/878066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/1"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Jorge San Emeterio","email":"jsanemet@redhat.com","date":"2023-03-21 11:30:39.000000000","tz":60},"committer":{"name":"Jorge San Emeterio","email":"jsanemet@redhat.com","date":"2023-03-21 11:32:36.000000000","tz":60},"subject":"WIP: Support virtio-scsi multiqueue for improved I/O performance on nova\u0027s guests.","message":"WIP: Support virtio-scsi multiqueue for improved I/O performance on nova\u0027s\nguests.\n\nChange-Id: Iade013228d5795311b5124285699cd706a9cf874\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7dedf72f84a6101861e99e2040b66d6bd802b0ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7dedf72f84a6101861e99e2040b66d6bd802b0ba"}]},"branch":"refs/heads/master"},"c8a4753483cce96ae0be512c18dfcaf07ec37056":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-21 11:34:46.000000000","uploader":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"ref":"refs/changes/66/878066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/66/878066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/2"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Jorge San Emeterio","email":"jsanemet@redhat.com","date":"2023-03-21 11:30:39.000000000","tz":60},"committer":{"name":"Jorge San Emeterio","email":"jsanemet@redhat.com","date":"2023-03-21 11:34:33.000000000","tz":60},"subject":"WIP: Support virtio-scsi multiqueue for improved I/O performance on nova\u0027s guests.","message":"WIP: Support virtio-scsi multiqueue for improved I/O performance on nova\u0027s guests.\n\nChange-Id: Iade013228d5795311b5124285699cd706a9cf874\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c8a4753483cce96ae0be512c18dfcaf07ec37056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c8a4753483cce96ae0be512c18dfcaf07ec37056"}]},"branch":"refs/heads/master"},"db2dd7ef792a2a62d9c076c67b307da250d237de":{"kind":"REWORK","_number":3,"created":"2023-03-21 12:02:53.000000000","uploader":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"ref":"refs/changes/66/878066/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/66/878066/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/3"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Jorge San Emeterio","email":"jsanemet@redhat.com","date":"2023-03-21 11:30:39.000000000","tz":60},"committer":{"name":"Jorge San Emeterio","email":"jsanemet@redhat.com","date":"2023-03-21 12:02:52.000000000","tz":60},"subject":"WIP: Support virtio-scsi multiqueue for improved I/O performance on nova\u0027s guests.","message":"WIP: Support virtio-scsi multiqueue for improved I/O performance on nova\u0027s guests.\n\nChange-Id: Iade013228d5795311b5124285699cd706a9cf874\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/db2dd7ef792a2a62d9c076c67b307da250d237de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/db2dd7ef792a2a62d9c076c67b307da250d237de"}]},"branch":"refs/heads/master"},"05fcb49eab6c1e9e56dd4616d8c787581d90ec05":{"kind":"REWORK","_number":4,"created":"2023-03-21 14:23:28.000000000","uploader":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"ref":"refs/changes/66/878066/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/66/878066/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/4"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Jorge San Emeterio","email":"jsanemet@redhat.com","date":"2023-03-21 11:30:39.000000000","tz":60},"committer":{"name":"Jorge San Emeterio","email":"jsanemet@redhat.com","date":"2023-03-21 14:22:40.000000000","tz":60},"subject":"Support virtio-scsi multiqueue for improved I/O performance on nova\u0027s guests","message":"Support virtio-scsi multiqueue for improved I/O performance on nova\u0027s guests\n\nThe blueprint aims to improve the libvirt driver by enabling the virtio-scsi\nmultiqueue feature. This will allow guest instances to have multiple I/O queues\nrunning in parallel, thus providing better performance.\n\nChange-Id: Iade013228d5795311b5124285699cd706a9cf874\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/05fcb49eab6c1e9e56dd4616d8c787581d90ec05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/05fcb49eab6c1e9e56dd4616d8c787581d90ec05"}]},"branch":"refs/heads/master"},"74b923e94d90fb170903c0aa1753d491b81c2994":{"kind":"REWORK","_number":5,"created":"2023-03-22 09:16:52.000000000","uploader":{"_account_id":34443,"name":"Jorge San Emeterio","display_name":"jsanemet","email":"jsanemet@redhat.com","username":"jsanemet"},"ref":"refs/changes/66/878066/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/66/878066/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/66/878066/5"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Jorge San Emeterio","email":"jsanemet@redhat.com","date":"2023-03-21 11:30:39.000000000","tz":60},"committer":{"name":"Jorge San Emeterio","email":"jsanemet@redhat.com","date":"2023-03-22 09:12:52.000000000","tz":60},"subject":"Support virtio-scsi multiqueue for improved I/O performance on nova\u0027s guests","message":"Support virtio-scsi multiqueue for improved I/O performance on nova\u0027s guests\n\nThe blueprint aims to improve the libvirt driver by enabling the virtio-scsi\nmultiqueue feature. This will allow guest instances to have multiple I/O queues\nrunning in parallel, thus providing better performance.\n\nChange-Id: Iade013228d5795311b5124285699cd706a9cf874\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/74b923e94d90fb170903c0aa1753d491b81c2994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/74b923e94d90fb170903c0aa1753d491b81c2994"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
