)]}'
{"id":"openstack%2Fnova-specs~878757","triplet_id":"openstack%2Fnova-specs~master~Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902","project":"openstack/nova-specs","branch":"master","topic":"dangling-volumes","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2023-07-06 15:32:33.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-07-06 15:32:33.000000000","reason":"Change was submitted"},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2023-07-06 15:32:33.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-07-06 15:32:33.000000000","reason":"Change was submitted"},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2023-07-06 15:32:33.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-07-05 16:34:17.000000000","reason":"\u003cGERRIT_ACCOUNT_7166\u003e replied on the change","reason_account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2023-07-06 15:32:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902","subject":"Adds cleanup to remove dangling volumes","status":"MERGED","created":"2023-03-28 10:21:28.000000000","updated":"2023-07-06 15:33:49.000000000","submitted":"2023-07-06 15:32:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":99,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"878757-dangling-volumes","meta_rev_id":"93c0347400b7a9f99120b8ab24f8e847b6d76b83","_number":878757,"virtual_id_number":878757,"owner":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-06 15:32:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-06 15:23:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2023-07-05 16:34:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":1,"date":"2023-07-05 08:50:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-06 15:23:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2023-07-05 16:34:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-28 10:29:20.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2023-03-28 10:29:20.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2023-03-28 10:29:20.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2023-03-28 10:29:20.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-03-28 10:32:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-31 15:44:05.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2023-05-25 17:22:54.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2023-07-05 08:50:01.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2023-07-06 15:23:45.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"5e59ec1ca631d0ff4ba818573caa00236466b631","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-03-28 10:21:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2419bf859178d8e71adc3001bdb4926bb087e3e0","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-03-28 10:25:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd120a455b335717873e37edaecf0512cf501e6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 10:32:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38cb72dfe459431db6f14b8b59abe587\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a640bbfcc7ca42a0b7d5d326308749a5 : FAILURE in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c47818c9002b4ea48426afdf2ec38685 : SUCCESS in 2m 50s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/2023.2/approved/nova-manage-cleanup-dangling-volume-attachments.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/2023.2/approved/nova-manage-cleanup-dangling-volume-attachments.rst","accounts_in_message":[],"_revision_number":1},{"id":"a07a9486a6c18acd6f479154ad722b285e66f884","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-03-28 10:45:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"716fcf38f3ba0c5d4abce54692ac9d1e4b899bed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 10:57:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e40f07b024f464b9bd2c672a3c65d15\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92e86ce22d7a44808eeb925b0ead4b2a : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69368cae469848f482f618aa0cd6a6e8 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"ef74a2649388a956ba1057d4e2737dae7f6f886d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-03-29 06:38:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7cd680da96765f438724d7fba4959d3d57058d5c","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-03-29 06:45:26.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"795d88e8733c29abf01b737cfc0c74f70d7dbee0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 06:46:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9fd9a7e734b4fd7b50b06481f41f819\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/203790b4c4d1448abb798c2c383a172d : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8ea2c1ef0bf45dc91e900be6d7443e4 : FAILURE in 2m 46s","accounts_in_message":[],"_revision_number":3},{"id":"ef49994631afd58e87938adbff583f7b8abfdf01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-03-29 06:52:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"003c2f3ed8e1e4292b598b92321bfd59f5690d2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-03-29 06:55:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3a19b6e460f9fcb17b5d16a00477bd50b0c3b668","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 07:27:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acd19ec3a8dc47f3817fc0a03802723e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d44988e584af48da869315cc64e437be : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73dcbadc82eb48e18c8541d99f6ab0d8 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":5},{"id":"6e2a5ec0ec41851cb2d41b876835bf57eb2330fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-03-31 10:55:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e6fbc41947439dcb82ba9ba15d5b93eb6beb20f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 11:05:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df84b22f26f34850948492fbe0502722\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b300e49d2fa74cab838d6c30073b0232 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c45fd3a942a474d9f203938ee074604 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":6},{"id":"e584a0a0311773b867dc9f86a18c62ba4f65cb3d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-03-31 11:23:30.000000000","message":"Patch Set 6: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c71ae81751d953b2ce3db980c959fc70bc005bb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-03-31 11:39:03.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4797669fc307efd183f7d65e9bf21c982f9a7c1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 11:47:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d655a1ea5d84bafb092ba020e869606\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1058fb3429d64125b785b8fbddd9a357 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5224c7462eab486b9aa7d6eb2d100d00 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":7},{"id":"fb7edf7261ace22ef03e98b0385a519e26a903ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-03-31 13:08:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1f3dbe57b58d866c8b885b0459eff1354538e145","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-03-31 13:10:24.000000000","message":"Patch Set 8:\n\n(8 comments)","accounts_in_message":[],"_revision_number":8},{"id":"5eebe4cfd5744e98cf5a3dc8e7544b7f629edf58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 13:20:35.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1ff07cf4a5d44bdbe1570f6e8c0c2ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e973807d17e4623a4df7701b6545f96 : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad5fe5e553174234bd4cf0072ee523e2 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":8},{"id":"dde97bf25b2848e8f1253463824817f66a028cb6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-03-31 17:45:17.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"611a269909b84eb388d788f7d403479f0f7fc75d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-04-03 06:20:47.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ebd9bf1670791e609e563479881836e7e58624ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 06:27:19.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcb3a73d0f55429e96212847e4ef8d6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bf9cafee4974511ac354f31e77f849f : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce0c61bc2fcc45c9ab4e68358b4e6865 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":9},{"id":"ebc26d4804fa4122224edc53270f121b5037b164","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-04-03 06:32:12.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"0b548012a458a87e45784541643672694eea3317","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-04-03 13:57:07.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f504b68fc67592450a14f565c9aac6ce858a0458","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-04-06 05:10:46.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"cbe1f89132089c1e43cc85e2e302aa2ea756106e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-04-06 14:00:26.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6c25081cc973e8e11251aca31cdaeb89514f4e6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-04-10 06:37:01.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"48f04f2af3f4b43573042792fc7868ead0364093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 06:51:12.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e091fb8bcbb4de7942411c96789adce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f902bbbde3c1493caef54e572756030e : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be2480413d6440498a225a2a174bc139 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":10},{"id":"4dbe5aecfb947ce81c1a7ef0fa7c917ba807efc5","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-04-10 07:02:26.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"1a09c50211641f70e36c9a1e9006b417070c222f","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-04-18 16:21:16.000000000","message":"Patch Set 10: Review-Priority+1","accounts_in_message":[],"_revision_number":10},{"id":"44fac48ec2129a55daa05a8c24abf62b946fb8a6","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-05-12 11:06:29.000000000","message":"Patch Set 10: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"55eb8bab3701469eb7df33743bfd113b2e5d7bef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-05-15 11:18:49.000000000","message":"Uploaded patch set 11.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"7824168faa2df09834fbc3b08008422d9dd53d52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 11:27:00.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83f68cab64a74e2995374b059ddb3f9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a6ac5579ee14bdda4a3778b98359bf8 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c913b582def4f6580f6fb71fce89087 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":11},{"id":"96b0dc2ea9784f365c04a1988a28056ce91d3fac","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-05-15 11:33:47.000000000","message":"Patch Set 11:\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"66009f5f46f7f43f7ffbdb6a7c9d8793c9609634","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-05-16 04:24:58.000000000","message":"Topic bp/https removed","accounts_in_message":[],"_revision_number":11},{"id":"6b8653d09567b8c590ad02eee58b1f78d5ad2326","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-05-16 04:25:16.000000000","message":"Topic set to dangling-volumes","accounts_in_message":[],"_revision_number":11},{"id":"488644d90feb9a99f727226cb280ce5bd8aadb01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-05-16 09:27:30.000000000","message":"Uploaded patch set 12.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"79c0cab5ecdf1b0f9d8205b331f11ac62a989a87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 10:06:32.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/725010cdeb6d4c86a108003b10248b25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5facbd85c1e45c292b919b91dad83f3 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ca5665f441544feb85f7f62fc5ac924 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":12},{"id":"8c5cc5f3c932d00ce41312a9ef7a7d2d4a442b8d","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-05-22 15:09:01.000000000","message":"Patch Set 12: Code-Review-1 Review-Priority+1\n\n(9 comments)","accounts_in_message":[],"_revision_number":12},{"id":"cc78a0a684de908a9fa1a5e6618a3faac7e18646","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-05-24 16:41:01.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"60fe6eb289f95477250e0aacceedf5cc73b2f550","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-05-24 17:02:30.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"ebd084304ded2929608a55861bd0fb547fb53834","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-05-24 17:10:45.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"2ba7922a481bf4194b94c0682e77b4754f191242","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-05-24 18:25:17.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b4eee7d10205f957a34b1900a9d35ba357ae8ba3","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-05-25 17:22:54.000000000","message":"Patch Set 12: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"ab6cf6019ead201001c68c094454c187ae910762","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-05-26 06:21:29.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"3af0bc07cb0ac6016aab127bbc8444c9b9560235","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-06-07 08:14:01.000000000","message":"Uploaded patch set 13.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"64e6eff3e99cafee89c13c2c39a58c00b23cfbb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 08:21:41.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6de9ae7e6ad4fe2be40899484872fb9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41d00f53b5148e48160d585471c2e06 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b554ff9f6b95436389a79991ef09627d : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":13},{"id":"335eeda9796ce29cfa281d34959987bd2327b7fe","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-06-07 08:43:47.000000000","message":"Patch Set 13:\n\n(5 comments)","accounts_in_message":[],"_revision_number":13},{"id":"80b0c6aa16ce732978dbf99c4fd2777318953ba7","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-06-08 07:57:41.000000000","message":"Patch Set 13: Code-Review-1 Review-Priority+1\n\n(8 comments)","accounts_in_message":[],"_revision_number":13},{"id":"726aac93bcac6b34f187ecdf0611e868eb390597","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-06-21 11:37:11.000000000","message":"Uploaded patch set 14.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"149cc545c8044fcb60d1d15d6a04ae8f6663dd41","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-06-21 11:39:00.000000000","message":"Patch Set 14:\n\n(10 comments)","accounts_in_message":[],"_revision_number":14},{"id":"c53260e00a8cc8a0afd87a770228566a935b95c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 11:54:52.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c889a4bd45c4ae387c6dd3c67ccfdf5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed2c85decc7e45fdb4ac9d390dd52862 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce02c4b7866d4ef4821056c105ae7d35 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":14},{"id":"40d1d591d8c3acb314cc0e4a0c04ae80a204e5af","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-06-21 12:33:21.000000000","message":"Patch Set 14:\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"e0d25d2b8ca7ee88b461b76a88c4ae96f73a2777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-06-30 09:45:31.000000000","message":"Uploaded patch set 15.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"f9bd5bc40f238d070d8db4cacc5bf93f84e6f90c","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-06-30 09:50:46.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"5466fb1adb58de0d99cc451ec4645a3c999d9d74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 09:52:05.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d607d0b82364c048e03cb5a087138b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2234f9c1074043309ecc75f8a71822a8 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ac0632ecfeb47dfad1d049a872d1153 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":15},{"id":"b749e3b9b3eb3fd816afb19887ad7316cb3faf5d","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-07-05 08:50:01.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"047054206aaf941c4586add6a88e6a300110d0e0","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-07-05 16:34:17.000000000","message":"Patch Set 15: Code-Review+2 Review-Priority+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":15},{"id":"4e8f5608d44475bb737143992b6c5bb43651f9d8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-07-06 15:23:45.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"9318a311bd9ceaaf983e2aab5439291e4031c575","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 15:24:03.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"bbb2a32e21bf498577105208cae334979185cb4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 15:32:33.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afbe4c5674244be989134b472ddd4bce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b025c903b0ea4496ae4ff6f2f48fa3ba : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3913557cf6e48fcac3f0352fd9e6570 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":15},{"id":"7c21ce4b801026ad15e8574afeb0eae4aaf94c5b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 15:32:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"93c0347400b7a9f99120b8ab24f8e847b6d76b83","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 15:33:49.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bed9896c0b04dc692d2394a08bb29b5\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/df4bacf271394967975f403912a7c3af : SUCCESS in 54s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"871aa3e4779be7deca9cde9d2a9bf40a5c56d010","revisions":{"aa9691f34752ca6f25879f0513efbdc1823ea796":{"kind":"REWORK","_number":1,"created":"2023-03-28 10:21:28.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/57/878757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/878757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/1"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:16:07.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:19:18.000000000","tz":0},"subject":"Add cleanup flag spec to remove dangling volumes","message":"Add cleanup flag spec to remove dangling volumes\n\nChange-Id: Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aa9691f34752ca6f25879f0513efbdc1823ea796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aa9691f34752ca6f25879f0513efbdc1823ea796"}]},"branch":"refs/heads/master"},"5195e23c43fa781847dff06b294ffb4f2e27b820":{"kind":"REWORK","_number":2,"created":"2023-03-28 10:45:50.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/57/878757/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/878757/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/2"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:16:07.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:42:33.000000000","tz":0},"subject":"Add cleanup flag spec to remove dangling volumes","message":"Add cleanup flag spec to remove dangling volumes\n\nChange-Id: Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5195e23c43fa781847dff06b294ffb4f2e27b820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5195e23c43fa781847dff06b294ffb4f2e27b820"}]},"branch":"refs/heads/master"},"c117b9c579185f154bf2a84f14a5b2220e0c78cf":{"kind":"REWORK","_number":3,"created":"2023-03-29 06:38:39.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/57/878757/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/878757/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/3"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:16:07.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-29 06:35:29.000000000","tz":0},"subject":"Add cleanup flag to remove dangling volumes","message":"Add cleanup flag to remove dangling volumes\n\nChange-Id: Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c117b9c579185f154bf2a84f14a5b2220e0c78cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c117b9c579185f154bf2a84f14a5b2220e0c78cf"}]},"branch":"refs/heads/master"},"ca3bc5e30e8a29e2fe327f67345fc91c8e1df9dd":{"kind":"REWORK","_number":4,"created":"2023-03-29 06:52:51.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/57/878757/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/878757/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/4"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:16:07.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-29 06:50:15.000000000","tz":0},"subject":"Add cleanup flag to remove dangling volumes","message":"Add cleanup flag to remove dangling volumes\n\nChange-Id: Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ca3bc5e30e8a29e2fe327f67345fc91c8e1df9dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ca3bc5e30e8a29e2fe327f67345fc91c8e1df9dd"}]},"branch":"refs/heads/master"},"d6eb12da3104a40afc6fc5ad17028dbf6bba37e2":{"kind":"REWORK","_number":5,"created":"2023-03-29 06:55:58.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/57/878757/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/878757/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/5"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:16:07.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-29 06:53:28.000000000","tz":0},"subject":"Add cleanup flag to remove dangling volumes","message":"Add cleanup flag to remove dangling volumes\n\nChange-Id: Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d6eb12da3104a40afc6fc5ad17028dbf6bba37e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d6eb12da3104a40afc6fc5ad17028dbf6bba37e2"}]},"branch":"refs/heads/master"},"f39b630c356135e619562205ebcd3dc1118fa3cb":{"kind":"REWORK","_number":6,"created":"2023-03-31 10:55:11.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/57/878757/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/878757/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/6"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:16:07.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-31 10:51:57.000000000","tz":0},"subject":"Add cleanup flag to remove dangling volumes","message":"Add cleanup flag to remove dangling volumes\n\nBlueprint: https://blueprints.launchpad.net/nova/+spec/nova-manage-cleanup-dangling-volume-attachments\n\nChange-Id: Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f39b630c356135e619562205ebcd3dc1118fa3cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f39b630c356135e619562205ebcd3dc1118fa3cb"}]},"branch":"refs/heads/master"},"d446e080a4199c3105ba39f764998993b98e0644":{"kind":"REWORK","_number":7,"created":"2023-03-31 11:39:03.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/57/878757/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/878757/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/7"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:16:07.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-31 11:36:20.000000000","tz":0},"subject":"Add cleanup flag to remove dangling volumes","message":"Add cleanup flag to remove dangling volumes\n\nBlueprint: https://blueprints.launchpad.net/nova/+spec/nova-manage-cleanup-dangling-volume-attachments\n\nChange-Id: Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d446e080a4199c3105ba39f764998993b98e0644"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d446e080a4199c3105ba39f764998993b98e0644"}]},"branch":"refs/heads/master"},"377947a6c998a366c57d8d6fc132d210a63e69cb":{"kind":"REWORK","_number":8,"created":"2023-03-31 13:08:30.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/57/878757/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/878757/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/8"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:16:07.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-31 13:05:46.000000000","tz":0},"subject":"Add cleanup flag to remove dangling volumes","message":"Add cleanup flag to remove dangling volumes\n\nBlueprint: https://blueprints.launchpad.net/nova/+spec/nova-manage-cleanup-dangling-volume-attachments\n\nChange-Id: Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/377947a6c998a366c57d8d6fc132d210a63e69cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/377947a6c998a366c57d8d6fc132d210a63e69cb"}]},"branch":"refs/heads/master"},"cc020388757c6b6a09fd723520b6b8cef051461b":{"kind":"REWORK","_number":9,"created":"2023-04-03 06:20:47.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/57/878757/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/878757/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/9"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:16:07.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-04-03 06:17:53.000000000","tz":0},"subject":"Add cleanup flag to remove dangling volumes","message":"Add cleanup flag to remove dangling volumes\n\nBlueprint: https://blueprints.launchpad.net/nova/+spec/nova-manage-cleanup-dangling-volume-attachments\n\nChange-Id: Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cc020388757c6b6a09fd723520b6b8cef051461b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cc020388757c6b6a09fd723520b6b8cef051461b"}]},"branch":"refs/heads/master"},"228f40f2fa188f94c5e43b2fccda623c42eea74d":{"kind":"REWORK","_number":10,"created":"2023-04-10 06:37:01.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/57/878757/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/878757/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/10"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:16:07.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-04-10 06:33:43.000000000","tz":0},"subject":"Add cleanup flag to remove dangling volumes","message":"Add cleanup flag to remove dangling volumes\n\nBlueprint: https://blueprints.launchpad.net/nova/+spec/nova-manage-cleanup-dangling-volume-attachments\n\nChange-Id: Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/228f40f2fa188f94c5e43b2fccda623c42eea74d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/228f40f2fa188f94c5e43b2fccda623c42eea74d"}]},"branch":"refs/heads/master"},"2e1c161885a91de6eeb3eaadece591794a540d55":{"kind":"REWORK","_number":11,"created":"2023-05-15 11:18:49.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/57/878757/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/878757/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/11"}}},"commit":{"parents":[{"commit":"522972445084053b447990ee049a903a1b87719b","subject":"Move Antelope implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/522972445084053b447990ee049a903a1b87719b"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:16:07.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-05-15 11:18:28.000000000","tz":0},"subject":"Adds cleanup to remove dangling volumes","message":"Adds cleanup to remove dangling volumes\n\nChange-Id: Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2e1c161885a91de6eeb3eaadece591794a540d55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2e1c161885a91de6eeb3eaadece591794a540d55"}]},"branch":"refs/heads/master"},"befeb2b945cd58663433afb9b0cf50ba54c64c30":{"kind":"REWORK","_number":12,"created":"2023-05-16 09:27:30.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/57/878757/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/878757/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/12"}}},"commit":{"parents":[{"commit":"e0ef37d5d049b5d70c088568fddb608ba90c6b3e","subject":"Merge \"Re-Propose \"Ironic Shards\" for Bobcat/2023.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e0ef37d5d049b5d70c088568fddb608ba90c6b3e"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:16:07.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-05-16 09:26:37.000000000","tz":0},"subject":"Adds cleanup to remove dangling volumes","message":"Adds cleanup to remove dangling volumes\n\nChange-Id: Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/befeb2b945cd58663433afb9b0cf50ba54c64c30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/befeb2b945cd58663433afb9b0cf50ba54c64c30"}]},"branch":"refs/heads/master"},"a50d1a21b081a7abfa719a34567902747d053fcd":{"kind":"REWORK","_number":13,"created":"2023-06-07 08:14:01.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/57/878757/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/878757/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/13"}}},"commit":{"parents":[{"commit":"e0ef37d5d049b5d70c088568fddb608ba90c6b3e","subject":"Merge \"Re-Propose \"Ironic Shards\" for Bobcat/2023.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e0ef37d5d049b5d70c088568fddb608ba90c6b3e"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:16:07.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-06-07 08:12:27.000000000","tz":0},"subject":"Adds cleanup to remove dangling volumes","message":"Adds cleanup to remove dangling volumes\n\nChange-Id: Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a50d1a21b081a7abfa719a34567902747d053fcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a50d1a21b081a7abfa719a34567902747d053fcd"}]},"branch":"refs/heads/master"},"1e1a3a926291df99de93cc7f73e6c8a3f4a583d3":{"kind":"REWORK","_number":14,"created":"2023-06-21 11:37:11.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/57/878757/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/878757/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/14"}}},"commit":{"parents":[{"commit":"e0ef37d5d049b5d70c088568fddb608ba90c6b3e","subject":"Merge \"Re-Propose \"Ironic Shards\" for Bobcat/2023.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e0ef37d5d049b5d70c088568fddb608ba90c6b3e"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:16:07.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-06-21 11:34:34.000000000","tz":0},"subject":"Adds cleanup to remove dangling volumes","message":"Adds cleanup to remove dangling volumes\n\nChange-Id: Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1e1a3a926291df99de93cc7f73e6c8a3f4a583d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1e1a3a926291df99de93cc7f73e6c8a3f4a583d3"}]},"branch":"refs/heads/master"},"871aa3e4779be7deca9cde9d2a9bf40a5c56d010":{"kind":"REWORK","_number":15,"created":"2023-06-30 09:45:31.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/57/878757/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/878757/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/878757/15"}}},"commit":{"parents":[{"commit":"e0ef37d5d049b5d70c088568fddb608ba90c6b3e","subject":"Merge \"Re-Propose \"Ironic Shards\" for Bobcat/2023.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e0ef37d5d049b5d70c088568fddb608ba90c6b3e"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-03-28 10:16:07.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-06-30 09:44:40.000000000","tz":0},"subject":"Adds cleanup to remove dangling volumes","message":"Adds cleanup to remove dangling volumes\n\nChange-Id: Ic7a456ecb59dd4498444f953a6bcb7f63ee3c902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/871aa3e4779be7deca9cde9d2a9bf40a5c56d010"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/871aa3e4779be7deca9cde9d2a9bf40a5c56d010"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
