)]}'
{"id":"openstack%2Fnova-specs~888101","triplet_id":"openstack%2Fnova-specs~master~I07615df5d313d76f4e17b3b7499819aed0671271","project":"openstack/nova-specs","branch":"master","attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2023-11-07 11:58:12.000000000","reason":"\u003cGERRIT_ACCOUNT_30623\u003e replied on the change","reason_account":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-12-06 11:21:11.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"30623":{"account":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"last_update":"2023-11-07 13:42:39.000000000","reason":"\u003cGERRIT_ACCOUNT_7166\u003e replied on the change","reason_account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I07615df5d313d76f4e17b3b7499819aed0671271","subject":"spec: evacuate api return migration","status":"NEW","created":"2023-07-11 02:23:23.000000000","updated":"2023-12-06 11:21:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":10,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"4befa6fc3b38fb340d25e6492fb7a1781fc31854","_number":888101,"virtual_id_number":888101,"owner":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-07-17 06:39:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":-1,"date":"2023-11-07 09:20:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-11-07 13:42:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}],"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-11 02:34:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-07 09:20:00.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2023-11-07 13:42:39.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2023-12-06 11:21:11.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"}],"messages":[{"id":"35e9520e07890939b4f79a61d65747bf8c21f6db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2023-07-11 02:23:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84c7c25b70f66863237db2d927fb98f1776ed6e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 02:34:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ba51026ab5545909aa8f5680b75072f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2190e81fd484445585ca44fce905994b : FAILURE in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9823bde89c0444a964d33e35eac3ab2 : FAILURE in 3m 14s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/2023.2/approved/evacuate-api-return-migration.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/2023.2/approved/evacuate-api-return-migration.rst","accounts_in_message":[],"_revision_number":1},{"id":"808911666d7aa321fca87a56dc560e8636927b33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2023-07-14 10:38:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"70c6fa4ad523179497e87ab692e059c49bf9dd63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 10:43:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/979a7585eb034ba695acb34146a5aef1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80c3bff2d9964956868b155929c668b6 : FAILURE in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/382d58252f1d4242b73554ea383e91a8 : FAILURE in 3m 52s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/2023.2/approved/evacuate-api-return-migration.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/2023.2/approved/evacuate-api-return-migration.rst","accounts_in_message":[],"_revision_number":2},{"id":"3b0494e0af4fda2cf49131eab87c5a350987cd0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2023-07-17 06:27:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"12b4b386623e4e31b8e0d1b87b9812797a395490","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 06:39:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fbb9ab916d34e37996346ad1e9a3553\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dcb0f7b3e7249cb812de1b2c6fad630 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f82ee61a9c34a7092c6ef036cd04c24 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":3},{"id":"e1d5cd5615c8b6c1c115832adf63fa5d093aa121","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-11-07 09:20:00.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8fedf88b993416a581869f3c4a7dacdaf83970e8","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2023-11-07 11:58:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e09d22726a0b02393d6b417e63076ad8ff1280d3","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-11-07 13:42:39.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4befa6fc3b38fb340d25e6492fb7a1781fc31854","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-12-06 11:21:11.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c31366ab3825986065f288908836d9d20758778d","revisions":{"86e7b77d05ec9cf5e4ca11370b055a8e71434462":{"kind":"REWORK","_number":1,"created":"2023-07-11 02:23:23.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/01/888101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/01/888101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/888101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/888101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/888101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/01/888101/1"}}},"commit":{"parents":[{"commit":"47845a267d0e655e8f2bd9ab5e469782e767e8be","subject":"Merge \"Add support for Napatech LinkVirt SmartNICs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/47845a267d0e655e8f2bd9ab5e469782e767e8be"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2023-07-11 02:14:22.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2023-07-11 02:14:22.000000000","tz":480},"subject":"spec: evacuate api return migration","message":"spec: evacuate api return migration\n\nIf Nova returns migration in the evacuate api, it could help Masakari to\nconfirm the evacuation in a simpler way.\n\nImplement: bp/evacuate-api-return-migration\nChange-Id: I07615df5d313d76f4e17b3b7499819aed0671271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/86e7b77d05ec9cf5e4ca11370b055a8e71434462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/86e7b77d05ec9cf5e4ca11370b055a8e71434462"}]},"branch":"refs/heads/master"},"eb6a4b7acd79f7d8e0352c20ef5554b1c2763a84":{"kind":"REWORK","_number":2,"created":"2023-07-14 10:38:35.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/01/888101/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/01/888101/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/888101/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/888101/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/888101/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/01/888101/2"}}},"commit":{"parents":[{"commit":"47845a267d0e655e8f2bd9ab5e469782e767e8be","subject":"Merge \"Add support for Napatech LinkVirt SmartNICs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/47845a267d0e655e8f2bd9ab5e469782e767e8be"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2023-07-11 02:14:22.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2023-07-14 08:01:07.000000000","tz":480},"subject":"spec: evacuate api return migration","message":"spec: evacuate api return migration\n\nIf Nova returns migration in the evacuate api, it could help Masakari to\nconfirm the evacuation in a simpler way.\n\nImplement: bp/evacuate-api-return-migration\nChange-Id: I07615df5d313d76f4e17b3b7499819aed0671271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eb6a4b7acd79f7d8e0352c20ef5554b1c2763a84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eb6a4b7acd79f7d8e0352c20ef5554b1c2763a84"}]},"branch":"refs/heads/master"},"c31366ab3825986065f288908836d9d20758778d":{"kind":"REWORK","_number":3,"created":"2023-07-17 06:27:54.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/01/888101/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/01/888101/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/888101/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/888101/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/888101/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/01/888101/3"}}},"commit":{"parents":[{"commit":"47845a267d0e655e8f2bd9ab5e469782e767e8be","subject":"Merge \"Add support for Napatech LinkVirt SmartNICs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/47845a267d0e655e8f2bd9ab5e469782e767e8be"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2023-07-11 02:14:22.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2023-07-17 06:27:20.000000000","tz":480},"subject":"spec: evacuate api return migration","message":"spec: evacuate api return migration\n\nIf Nova returns migration in the evacuate api, it could help Masakari to\nconfirm the evacuation in a simpler way.\n\nImplement: bp/evacuate-api-return-migration\nChange-Id: I07615df5d313d76f4e17b3b7499819aed0671271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c31366ab3825986065f288908836d9d20758778d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c31366ab3825986065f288908836d9d20758778d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
