)]}'
{"id":"openstack%2Fnova-specs~889844","triplet_id":"openstack%2Fnova-specs~master~I2c0ef44f4ae5185d40e117fabdd40472260c52c6","project":"openstack/nova-specs","branch":"master","topic":"bp/mks-proxy","attention_set":{},"removed_from_attention_set":{"33952":{"account":{"_account_id":33952,"name":"alecorps","display_name":"Alban Lecorps","email":"alban.lecorps@gmail.com","username":"alecorps7"},"last_update":"2023-07-27 12:57:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2c0ef44f4ae5185d40e117fabdd40472260c52c6","subject":"new spec: support of vnc console for mks","status":"ABANDONED","created":"2023-07-27 09:02:49.000000000","updated":"2023-07-27 12:57:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9c50137ae44fd5980bc106c104182912bf66f027","_number":889844,"virtual_id_number":889844,"owner":{"_account_id":33952,"name":"alecorps","display_name":"Alban Lecorps","email":"alban.lecorps@gmail.com","username":"alecorps7"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-07-27 09:17:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-27 09:17:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4afceaf2cc979914678d83e89862ac6aea52885d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33952,"name":"alecorps","display_name":"Alban Lecorps","email":"alban.lecorps@gmail.com","username":"alecorps7"},"date":"2023-07-27 09:02:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d000e7b8082a751b27555eedbb731a05da20d05e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 09:17:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dad5612103e944fb93f4eb02c4e3c97d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21c8621e40fc412ca3df56ea32a496de : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6052ab9717ac465ab55625f0585cc345 : FAILURE in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9c50137ae44fd5980bc106c104182912bf66f027","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33952,"name":"alecorps","display_name":"Alban Lecorps","email":"alban.lecorps@gmail.com","username":"alecorps7"},"date":"2023-07-27 12:57:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0156d5b58ee6a402ef2163a5706b23f5898ed5f","revisions":{"d0156d5b58ee6a402ef2163a5706b23f5898ed5f":{"kind":"REWORK","_number":1,"created":"2023-07-27 09:02:49.000000000","uploader":{"_account_id":33952,"name":"alecorps","display_name":"Alban Lecorps","email":"alban.lecorps@gmail.com","username":"alecorps7"},"ref":"refs/changes/44/889844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/889844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/889844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/889844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/889844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/889844/1"}}},"commit":{"parents":[{"commit":"47845a267d0e655e8f2bd9ab5e469782e767e8be","subject":"Merge \"Add support for Napatech LinkVirt SmartNICs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/47845a267d0e655e8f2bd9ab5e469782e767e8be"}]}],"author":{"name":"alecorps7","email":"alban.lecorps@gmail.com","date":"2023-07-27 09:02:15.000000000","tz":120},"committer":{"name":"alecorps7","email":"alban.lecorps@gmail.com","date":"2023-07-27 09:02:15.000000000","tz":120},"subject":"new spec: support of vnc console for mks","message":"new spec: support of vnc console for mks\n\nChange-Id: I2c0ef44f4ae5185d40e117fabdd40472260c52c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d0156d5b58ee6a402ef2163a5706b23f5898ed5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d0156d5b58ee6a402ef2163a5706b23f5898ed5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
