)]}'
{"id":"openstack%2Fnova-specs~895135","triplet_id":"openstack%2Fnova-specs~master~I5ceb4cfd357187a6be4a62dfbd3bc11a294ec703","project":"openstack/nova-specs","branch":"master","topic":"libvirt-maxphysaddr-support-caracal","attention_set":{},"removed_from_attention_set":{"31652":{"account":{"_account_id":31652,"name":"Nobuhiro MIKI","email":"nmiki@lycorp.co.jp","username":"nmiki"},"last_update":"2023-12-06 14:45:31.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-12-06 11:28:56.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-12-06 14:35:31.000000000","reason":"\u003cGERRIT_ACCOUNT_7166\u003e replied on the change","reason_account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}}},"hashtags":[],"change_id":"I5ceb4cfd357187a6be4a62dfbd3bc11a294ec703","subject":"Re-propose \"Add maxphysaddr support for Libvirt\" for 2024.1 Caracal","status":"MERGED","created":"2023-09-14 05:53:41.000000000","updated":"2023-12-06 14:47:06.000000000","submitted":"2023-12-06 14:45:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895135-libvirt-maxphysaddr-support-caracal","meta_rev_id":"006a47579d0717d4f63171ef1b25961fac4b6bb4","_number":895135,"virtual_id_number":895135,"owner":{"_account_id":31652,"name":"Nobuhiro MIKI","email":"nmiki@lycorp.co.jp","username":"nmiki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-06 14:45:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2023-12-06 11:28:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-06 14:35:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-06 14:35:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-14 06:08:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-07 13:32:32.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2023-11-08 10:26:27.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"838b6175ecfc51e51ea31c602530ce8af95f2eed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31652,"name":"Nobuhiro MIKI","email":"nmiki@lycorp.co.jp","username":"nmiki"},"date":"2023-09-14 05:53:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"646d00664367f249c6886753d8f88ac74af5e4ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 06:08:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/625f1f14aef84eef9670219c253c4943\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/334071e0730749d9a7f77f9a4031f449 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34ff68d66fb048f5b4f353c8e1afca1d : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2077b1a7d3800acd9dc6a05f8be850ff768317f9","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-11-07 13:32:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5963e27bed121dadab425d0fe2f301eb67f4df0f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-11-08 10:26:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"440461c610776afa77ed07e4c53125e36fc996cd","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-11-08 14:26:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7617330a20bfac6567f6c17ff3081b25d187cdd6","author":{"_account_id":31652,"name":"Nobuhiro MIKI","email":"nmiki@lycorp.co.jp","username":"nmiki"},"date":"2023-11-09 10:06:00.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b7ada0975981af79f685f89bfd27c8e0d4d987cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31652,"name":"Nobuhiro MIKI","email":"nmiki@lycorp.co.jp","username":"nmiki"},"date":"2023-11-10 02:30:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7e244f7a46d932e5ca4d586da6edfab4e3d5df97","author":{"_account_id":31652,"name":"Nobuhiro MIKI","email":"nmiki@lycorp.co.jp","username":"nmiki"},"date":"2023-11-10 02:33:05.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5fcd30f1fc1f553094c62b934b535fb720a48e92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 02:54:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/868bd5a1341a43d89badf736186f4159\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d4a8c6ffbda4f75b5962eeb86dd2cb2 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23daed6ec56644a7b2b83b3602052215 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":2},{"id":"94b086bae84ffaf81443f2c06fc845fe38e09d95","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-12-06 11:28:56.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be949fe606ccd1eba4a0d4ce98fa0a114c125eb6","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-12-06 14:35:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a9378285eafbe3feea21dd33f40976cb18bb1b1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 14:35:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"522848ba9d44a7872b5d80d052ab5b14d9e22b08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 14:45:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b32f990ade143a6b01108cb7789065a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9b23c023af84652986ab99733259bc7 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0c2bc1291734e59b1d175693f555243 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"58de2112fae0310da44e2b16f0c4fb78fd93bd2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 14:45:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"006a47579d0717d4f63171ef1b25961fac4b6bb4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 14:47:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/929cf3e3e1fc4ad9abb5c9f7c9175f30\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/e989a9cf83d54488ac44ca083e43ef39 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"346d77f7c6f9586e73a94c4332e5ac922c027bc8","revisions":{"4bac406db031d014399a76138641c5b21920e751":{"kind":"REWORK","_number":1,"created":"2023-09-14 05:53:41.000000000","uploader":{"_account_id":31652,"name":"Nobuhiro MIKI","email":"nmiki@lycorp.co.jp","username":"nmiki"},"ref":"refs/changes/35/895135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/35/895135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/895135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/895135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/895135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/35/895135/1"}}},"commit":{"parents":[{"commit":"d5ce36557922a89f3488ffd53e0132e7564b9ff4","subject":"Merge \"Create specs directory for 2024.1 Caracal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d5ce36557922a89f3488ffd53e0132e7564b9ff4"}]}],"author":{"name":"Nobuhiro MIKI","email":"nmiki@yahoo-corp.jp","date":"2023-09-14 05:36:05.000000000","tz":540},"committer":{"name":"Nobuhiro MIKI","email":"nmiki@yahoo-corp.jp","date":"2023-09-14 05:52:48.000000000","tz":540},"subject":"Re-propose \"Add maxphysaddr support for Libvirt\" for 2024.1 Caracal","message":"Re-propose \"Add maxphysaddr support for Libvirt\" for 2024.1 Caracal\n\nblueprint: libvirt-maxphysaddr-support-caracal\nSigned-off-by: Nobuhiro MIKI \u003cnmiki@yahoo-corp.jp\u003e\nChange-Id: I5ceb4cfd357187a6be4a62dfbd3bc11a294ec703\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4bac406db031d014399a76138641c5b21920e751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4bac406db031d014399a76138641c5b21920e751"}]},"branch":"refs/heads/master"},"346d77f7c6f9586e73a94c4332e5ac922c027bc8":{"kind":"REWORK","_number":2,"created":"2023-11-10 02:30:17.000000000","uploader":{"_account_id":31652,"name":"Nobuhiro MIKI","email":"nmiki@lycorp.co.jp","username":"nmiki"},"ref":"refs/changes/35/895135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/35/895135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/895135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/895135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/895135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/35/895135/2"}}},"commit":{"parents":[{"commit":"f1ec291df71cf155b7cbf700c30be75006e4b204","subject":"Merge \"Add libvirt-dev-alias spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f1ec291df71cf155b7cbf700c30be75006e4b204"}]}],"author":{"name":"Nobuhiro MIKI","email":"nmiki@yahoo-corp.jp","date":"2023-09-14 05:36:05.000000000","tz":540},"committer":{"name":"Nobuhiro MIKI","email":"nmiki@yahoo-corp.jp","date":"2023-11-10 02:28:30.000000000","tz":540},"subject":"Re-propose \"Add maxphysaddr support for Libvirt\" for 2024.1 Caracal","message":"Re-propose \"Add maxphysaddr support for Libvirt\" for 2024.1 Caracal\n\nblueprint: libvirt-maxphysaddr-support-caracal\nSigned-off-by: Nobuhiro MIKI \u003cnmiki@yahoo-corp.jp\u003e\nChange-Id: I5ceb4cfd357187a6be4a62dfbd3bc11a294ec703\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/346d77f7c6f9586e73a94c4332e5ac922c027bc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/346d77f7c6f9586e73a94c4332e5ac922c027bc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
