)]}'
{"id":"openstack%2Fnova-specs~898319","triplet_id":"openstack%2Fnova-specs~master~Ied43ce8fad0ff8daf3853b01a387700b6d931053","project":"openstack/nova-specs","branch":"master","topic":"bp/manila_shares_attachments","attention_set":{},"removed_from_attention_set":{"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-11-07 10:36:00.000000000","reason":"\u003cGERRIT_ACCOUNT_7166\u003e replied on the change","reason_account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2023-11-08 10:42:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied43ce8fad0ff8daf3853b01a387700b6d931053","subject":"Re-propose \"Allow Manila shares to be directly attached to an instance when using libvirt\" for Caracal","status":"MERGED","created":"2023-10-16 13:17:56.000000000","updated":"2023-11-08 10:43:46.000000000","submitted":"2023-11-08 10:42:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898319-bp/manila_shares_attachments","meta_rev_id":"43e4b600ca77ca26a50245234904b1257423aef7","_number":898319,"virtual_id_number":898319,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-08 10:42:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2023-11-08 10:29:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-07 10:36:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2023-11-08 10:29:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-16 13:19:14.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2023-10-16 13:27:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-08 10:29:45.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"cd7a22f705ef161d39dba7f8d3947d73454dee31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-10-16 13:17:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8dcb2695147d22e4648c0045631741da67559c82","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-10-16 13:18:40.000000000","message":"Topic bp/manila_shares_attachments_v2 removed","accounts_in_message":[],"_revision_number":1},{"id":"0c8bd801cf9c898a25fc31ea7d637d2b6e21a893","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-10-16 13:18:47.000000000","message":"Topic set to bp/manila_shares_attachments_v3","accounts_in_message":[],"_revision_number":1},{"id":"461c9d519398cddd1d95f69063d8e833e93e629a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-10-16 13:18:50.000000000","message":"Topic bp/manila_shares_attachments_v3 removed","accounts_in_message":[],"_revision_number":1},{"id":"b68e21d56af9d8f894ba8d8ac040d1d0e352f7e8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-10-16 13:18:58.000000000","message":"Topic set to bp/manila_shares_attachments","accounts_in_message":[],"_revision_number":1},{"id":"23548564dd33a455147d5461197d9246108929b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2023-10-16 13:20:11.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"91e9dccd4a44d92f1b5cd603de62e0a0ff1cbf4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 13:27:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3005dfc4a4924e27a70f84f988376811\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36ea5899ffd44adbf02c3bdcbefc60c : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85fec910d96c4f6d952410b9b882102c : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":2},{"id":"b8c14ceab5643af62457095ae8779a401b783900","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-11-07 10:36:00.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba7c8ca76e09a1a0d8e589e2097ffd875b60dafe","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-11-08 10:29:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3a47503616c909c390aca90166690b7fdb553c2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 10:30:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3f32fde649a318ead0fe85e156770c415bb8f21b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 10:42:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1afdb66fefb4f84888f8d9c6abc8a73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab92db1bbdb148139497a27c7051a360 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8267e86328484eaeb3a5c7732b27abe7 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"4b5acbff8350d611b6a32a56024781fac2df5f3c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 10:42:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"43e4b600ca77ca26a50245234904b1257423aef7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 10:43:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c92832f59c7b480db6adbe72cfc78bfb\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/43d6a3735bfa47d8a1173bc79b92a3d7 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f88dd19bfff719c9cdd495a19c83319610139e64","revisions":{"3d808b423f945d074f09411051dc9289621f2226":{"kind":"REWORK","_number":1,"created":"2023-10-16 13:17:56.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/19/898319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/19/898319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/19/898319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/19/898319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/19/898319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/19/898319/1"}}},"commit":{"parents":[{"commit":"c6b6eab6304203f6fca32dd3ce074b071b433171","subject":"Merge \"Re-propose using extend volume completion action for 2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c6b6eab6304203f6fca32dd3ce074b071b433171"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2023-10-16 13:14:48.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2023-10-16 13:14:48.000000000","tz":120},"subject":"Re-propose \"Allow Manila shares to be directly attached to an instance when using libvirt\" for Caracal","message":"Re-propose \"Allow Manila shares to be directly attached to an instance when\nusing libvirt\" for Caracal\n\n    Implements: blueprint libvirt-virtiofs-attach-manila-shares\n\nChange-Id: Ied43ce8fad0ff8daf3853b01a387700b6d931053\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3d808b423f945d074f09411051dc9289621f2226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3d808b423f945d074f09411051dc9289621f2226"}]},"branch":"refs/heads/master"},"f88dd19bfff719c9cdd495a19c83319610139e64":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-16 13:20:11.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/19/898319/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/19/898319/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/19/898319/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/19/898319/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/19/898319/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/19/898319/2"}}},"commit":{"parents":[{"commit":"c6b6eab6304203f6fca32dd3ce074b071b433171","subject":"Merge \"Re-propose using extend volume completion action for 2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c6b6eab6304203f6fca32dd3ce074b071b433171"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2023-10-16 13:14:48.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2023-10-16 13:19:53.000000000","tz":120},"subject":"Re-propose \"Allow Manila shares to be directly attached to an instance when using libvirt\" for Caracal","message":"Re-propose \"Allow Manila shares to be directly attached to an instance\nwhen using libvirt\" for Caracal\n\n    Implements: blueprint libvirt-virtiofs-attach-manila-shares\n\nChange-Id: Ied43ce8fad0ff8daf3853b01a387700b6d931053\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f88dd19bfff719c9cdd495a19c83319610139e64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f88dd19bfff719c9cdd495a19c83319610139e64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
