)]}'
{"id":"openstack%2Fnova-specs~898553","triplet_id":"openstack%2Fnova-specs~master~Idcc929a4e6488bbb3db5a5294b553582a3882875","project":"openstack/nova-specs","branch":"master","topic":"enforce-console-timeout","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2023-12-20 00:54:09.000000000","reason":"Change was submitted"},"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2023-12-20 00:54:09.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-12-20 00:46:50.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2023-12-20 00:54:09.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2023-12-19 16:15:02.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-12-20 00:54:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idcc929a4e6488bbb3db5a5294b553582a3882875","subject":"Enforce console session timeout","status":"MERGED","created":"2023-10-17 14:27:08.000000000","updated":"2024-01-29 04:43:28.000000000","submitted":"2023-12-20 00:54:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":72,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"898553-enforce-console-timeout","meta_rev_id":"43584d829b6fcab078cd01df630645e510b5a1e0","_number":898553,"virtual_id_number":898553,"owner":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-20 00:54:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2023-12-20 00:46:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-19 16:15:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2023-12-20 00:46:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-17 14:39:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-01 08:00:07.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2023-11-01 08:00:07.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-11-07 08:48:47.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2023-11-07 14:11:32.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2023-11-13 19:31:44.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"CC"}],"messages":[{"id":"393ce1f2f582d6f02ef36d83909c8808728e1320","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-10-17 14:27:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b52b77bed7763c202be2c9e94ee109ae895eb27d","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-10-17 14:32:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9587633aac65e9bbecb9526675857ab8dbb3d75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 14:39:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e7a43da0d0c44ab9e581edee22e3a02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39cb5a1e634543279e826f2cbad546a9 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aeecb9fd40f4ce8994947ab2873555b : FAILURE in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"1b152fdf84220f3948c3196a3628a2f57ee4cc63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-10-26 12:05:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6c5ae201106d7bb58a6ba16811c93b227f33f472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 12:16:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/609b852bf63e4220bf8b029ba7586722\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08474fded973434cb4d8eb847ab211e7 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fc0107eec414846b0b5f0fe25b0fa10 : FAILURE in 2m 40s","accounts_in_message":[],"_revision_number":2},{"id":"85d5f93a62a7354d9dbd5f69d50eef7e1fad2e32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-10-26 12:51:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"09de2d3eb5fbfbf6fd0a3a9c4bb95ecdcc10b41f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 13:06:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb01ab39942a450f85af4c5851a02219\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a46b272b4624c49a03fc48713d7ad5d : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f2639d55915460f968726191ca09044 : FAILURE in 2m 44s","accounts_in_message":[],"_revision_number":3},{"id":"9611c9e44ee5f78d73c00a54b189812953b0cd88","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-10-26 14:10:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e16b824070f75600bf0569f3687b3e6a348ebd86","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-10-27 08:24:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ad987a15b7c684f86422d4eb90a4ebd6204558c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-11-01 07:57:14.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1c1ce6c4e3eb249716ca1ec4d2b5a417840e43e5","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-11-01 08:00:07.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"944047e75349121e96727933496a1076444a6a7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 08:09:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f082f8562c4443a09bdfb7f0ce256d13\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/353ee9f385334fd7a22ae20b954036e6 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14084615f5df443a83beeed68ee09213 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":4},{"id":"7d08a09e4e7d55bdb60ee62ed08e4d1480c10061","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-11-07 08:48:47.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fbb9ad2c9452c76e208449bfd3e7b3cf6623e90d","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-11-07 11:27:16.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ebb16e0e8ada4acc4667094dbcbc47bf3ae96386","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-11-07 14:11:32.000000000","message":"Patch Set 4: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"24a85c734fb494c6cc31da5493d49393cd10be4b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-11-07 14:54:32.000000000","message":"Patch Set 4: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f4d74c1042a09350cdc49ff7844a2de9588909df","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2023-11-13 19:31:44.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"398ebbb902223d449d244c79fc6e3495c3077057","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-11-14 09:27:08.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"efe0718fb078428f34e8caa24e78737998700571","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-11-14 09:32:01.000000000","message":"Patch Set 5:\n\n(10 comments)","accounts_in_message":[],"_revision_number":5},{"id":"80a04d2515d5b6d968b13003585cd476144f32f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 09:39:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed59dce71131454ba7d8b174e29a7d2b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94f69498d1b246a6ab8740367a021e96 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3645ac8e1abf41da8dccd82c23618a49 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":5},{"id":"a8c6ee5d0ededb59e6e5b8cf18fd6638f41b2cf0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-12-06 11:51:49.000000000","message":"Patch Set 5: Code-Review-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":5},{"id":"cf012fae3ae4b790a06c7d31c44f5f5e2b69e33b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-12-07 07:25:56.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"fa18d6988aa2092e3ebd28f3be8a490686acc60a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 07:55:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52ac93642a8f4fa4ac15ec93f515d24c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac19da0bcd04f0aab29abaf0963599b : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5392f16dee32450da0a9dbea2e4916a7 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":6},{"id":"e5aa83d9e88e6439ee53c5f2b407cf967ee13c26","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-12-07 08:08:53.000000000","message":"Patch Set 6:\n\n(10 comments)","accounts_in_message":[],"_revision_number":6},{"id":"efb5298a8c44daa520b481f8a3a8c6800d11c782","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-12-19 16:15:02.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8e35c0d0bf1ccdcd5690b7ead65f8ad31eb92af9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-12-20 00:46:50.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"87613451d2cb16a8fb780de83c847f6a1078b030","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 00:47:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0f5b4557e5bbc57bbdd30c8162d59896612c9afa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 00:54:09.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8654331423764874bfa53e8e77da0e05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6da8b59dd5b74047935f7c496fe1f676 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79591e828024446f9124a0c7915369bb : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":6},{"id":"94edf4610e5bab9be819525e038a60a0c2cba771","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 00:54:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"ffaee100b0bca16cbe87ce10cfdee7ef7e67a778","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 00:55:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15da9c2820bb47c29959e599b5e26aa0\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/43f7350472da4b24983857feee37173c : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":6},{"id":"43584d829b6fcab078cd01df630645e510b5a1e0","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-01-29 04:43:28.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2a59cf0998deda81e8be7aa29a63f03bb5341462","revisions":{"579b0d9f467c3529214673717d0b0cc657f2c3ff":{"kind":"REWORK","_number":1,"created":"2023-10-17 14:27:08.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/53/898553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/53/898553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/1"}}},"commit":{"parents":[{"commit":"c6b6eab6304203f6fca32dd3ce074b071b433171","subject":"Merge \"Re-propose using extend volume completion action for 2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c6b6eab6304203f6fca32dd3ce074b071b433171"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-10-17 14:16:29.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-10-17 14:26:50.000000000","tz":0},"subject":"WIP: Enforce console session timeout","message":"WIP: Enforce console session timeout\n\nChange-Id: Idcc929a4e6488bbb3db5a5294b553582a3882875\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/579b0d9f467c3529214673717d0b0cc657f2c3ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/579b0d9f467c3529214673717d0b0cc657f2c3ff"}]},"branch":"refs/heads/master"},"68224b3ccbe82b24527078e22d843278a59e3b04":{"kind":"REWORK","_number":2,"created":"2023-10-26 12:05:01.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/53/898553/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/53/898553/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/2"}}},"commit":{"parents":[{"commit":"c6b6eab6304203f6fca32dd3ce074b071b433171","subject":"Merge \"Re-propose using extend volume completion action for 2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c6b6eab6304203f6fca32dd3ce074b071b433171"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-10-17 14:16:29.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-10-26 12:04:57.000000000","tz":0},"subject":"WIP: Enforce console session timeout","message":"WIP: Enforce console session timeout\n\nChange-Id: Idcc929a4e6488bbb3db5a5294b553582a3882875\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/68224b3ccbe82b24527078e22d843278a59e3b04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/68224b3ccbe82b24527078e22d843278a59e3b04"}]},"branch":"refs/heads/master"},"6ba23839d8ca699bf693dc8e2aabe0b81d51356b":{"kind":"REWORK","_number":3,"created":"2023-10-26 12:51:04.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/53/898553/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/53/898553/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/3"}}},"commit":{"parents":[{"commit":"c6b6eab6304203f6fca32dd3ce074b071b433171","subject":"Merge \"Re-propose using extend volume completion action for 2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c6b6eab6304203f6fca32dd3ce074b071b433171"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-10-17 14:16:29.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-10-26 12:50:47.000000000","tz":0},"subject":"WIP: Enforce console session timeout","message":"WIP: Enforce console session timeout\n\nChange-Id: Idcc929a4e6488bbb3db5a5294b553582a3882875\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ba23839d8ca699bf693dc8e2aabe0b81d51356b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ba23839d8ca699bf693dc8e2aabe0b81d51356b"}]},"branch":"refs/heads/master"},"3474c1140b7c8a8485258fc17bfc797d0ff240f1":{"kind":"REWORK","_number":4,"created":"2023-11-01 07:57:14.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/53/898553/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/53/898553/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/4"}}},"commit":{"parents":[{"commit":"c6b6eab6304203f6fca32dd3ce074b071b433171","subject":"Merge \"Re-propose using extend volume completion action for 2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c6b6eab6304203f6fca32dd3ce074b071b433171"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-10-17 14:16:29.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-11-01 07:56:50.000000000","tz":0},"subject":"Enforce console session timeout","message":"Enforce console session timeout\n\nChange-Id: Idcc929a4e6488bbb3db5a5294b553582a3882875\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3474c1140b7c8a8485258fc17bfc797d0ff240f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3474c1140b7c8a8485258fc17bfc797d0ff240f1"}]},"branch":"refs/heads/master"},"3f6266f16365ffe76622c449f14b57c7df3c9fbb":{"kind":"REWORK","_number":5,"created":"2023-11-14 09:27:08.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/53/898553/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/53/898553/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/5"}}},"commit":{"parents":[{"commit":"c6b6eab6304203f6fca32dd3ce074b071b433171","subject":"Merge \"Re-propose using extend volume completion action for 2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c6b6eab6304203f6fca32dd3ce074b071b433171"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-10-17 14:16:29.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-11-14 09:26:52.000000000","tz":0},"subject":"Enforce console session timeout","message":"Enforce console session timeout\n\nChange-Id: Idcc929a4e6488bbb3db5a5294b553582a3882875\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3f6266f16365ffe76622c449f14b57c7df3c9fbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3f6266f16365ffe76622c449f14b57c7df3c9fbb"}]},"branch":"refs/heads/master"},"2a59cf0998deda81e8be7aa29a63f03bb5341462":{"kind":"REWORK","_number":6,"created":"2023-12-07 07:25:56.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/53/898553/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/53/898553/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/53/898553/6"}}},"commit":{"parents":[{"commit":"c6b6eab6304203f6fca32dd3ce074b071b433171","subject":"Merge \"Re-propose using extend volume completion action for 2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c6b6eab6304203f6fca32dd3ce074b071b433171"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-10-17 14:16:29.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2023-12-07 07:19:39.000000000","tz":0},"subject":"Enforce console session timeout","message":"Enforce console session timeout\n\nChange-Id: Idcc929a4e6488bbb3db5a5294b553582a3882875\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2a59cf0998deda81e8be7aa29a63f03bb5341462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2a59cf0998deda81e8be7aa29a63f03bb5341462"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
