)]}'
{"id":"openstack%2Fnova-specs~900636","triplet_id":"openstack%2Fnova-specs~master~Iee16cb71d31e1308226adc0a39caef5d9970e00c","project":"openstack/nova-specs","branch":"master","topic":"bp/libvirt-mdev-live-migrate","attention_set":{},"removed_from_attention_set":{"6962":{"account":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"last_update":"2023-12-20 10:37:17.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-12-19 18:39:30.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2023-12-20 10:28:48.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-12-20 10:37:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iee16cb71d31e1308226adc0a39caef5d9970e00c","subject":"Proposes mdev live-migration support in libvirt","status":"MERGED","created":"2023-11-10 16:27:32.000000000","updated":"2023-12-20 10:38:48.000000000","submitted":"2023-12-20 10:37:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":55,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900636-bp/libvirt-mdev-live-migrate","meta_rev_id":"6cfb32d48e2589d8a4fab17ac013601230674981","_number":900636,"virtual_id_number":900636,"owner":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-20 10:37:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2023-12-19 18:39:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2023-12-20 10:28:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2023-12-20 10:28:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-10 16:41:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-06 12:29:29.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-12-14 13:31:18.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2023-12-14 16:30:04.000000000","updated_by":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"reviewer":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"state":"CC"}],"messages":[{"id":"8674fb517637182090bd756aaed14423dd8b783b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-11-10 16:27:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c05bbfa7c2a5bbb0c21823d0b162fe47d97bc4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 16:41:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5914759f6f04212ac17db2c2ef09274\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d6d8116771941ce99bb183b4418472f : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62e64996428d4143b1397f4776b27de9 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"bf9bd9ac0ba82887c28b832af92bb772b9ade352","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-11-14 16:34:06.000000000","message":"Patch Set 1: Workflow-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1c926bdfaf000fde8f20d1b8318fc02317b3fe18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-11-16 15:12:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a9a61b43d12dcffbfe1f03685e1f1b4a72fb27ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 15:20:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b86ea5d03ca14d2aab7a3d680af66904\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37099094b72f4eb0bc23f49f840cfe46 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cbe2e4758014497b4dbeca91009bca5 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2},{"id":"709175b7fb52794e96e2245e812c26ad076bd683","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-11-16 16:20:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7c26b656ce8beb3b236b5039e164849f31fe2b3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 16:29:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1402b50e0c8d4152bd358008971c2082\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840a3e971e2147559b84a34809902188 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed12bf86d14f40c69ad096909e8f95c1 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"74f341ca509f5a2d12802a18f892a0de08f2f9f6","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-11-17 10:04:46.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"334e99245462249a1daecc8b7c7a3a7bf92413cc","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-12-06 12:29:29.000000000","message":"Patch Set 3: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"51e2fa358fb48fcce0e674e76f3e37e6d0c4da4d","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-12-06 16:08:22.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"865f26e99880afedbed108c1dfae14a547bce9ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-12-08 12:31:22.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"0cd12f5379174d682d6a11056d55b7f5b8106d84","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-12-08 12:38:39.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0b164866d0aa4a75854505c1d97be03cf36aa12c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-12-08 12:38:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c74858a58484a214d58fdfe4dedc52cc6afa6cf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 12:47:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3452866ccb9848e5badfdd756f0ac2a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5d3979e5c864a999adeae4968a65493 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7bce66253ba45b1b596e6be805cec93 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":5},{"id":"62ffae834f8e244f1c6b6e57dcec11da02071dad","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-12-12 15:19:37.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"36943ef3f4f75f5da97588367402190c8fcccab4","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-12-12 15:42:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4d632d782ce2d9dfd0b221180271aa9f2e484db6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-12-14 13:31:18.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5bf8960fa9d572ab57eb670a455fd4615995c967","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-12-14 13:42:11.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5c9a6bd77be3f77f144334b18ee31b0be0cad0ab","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-12-14 15:12:34.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ebb1ed2f393da1e41d632f0f520f70d7849fff57","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-12-14 15:58:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"31eec1d359c5cc2880099283dab9882164ca589b","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-12-14 17:59:52.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e1220a398ca0d6067edaa048aefcf3c1db025f44","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-12-15 16:33:49.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b238bb60f9be7279b1c98b50a726050c3b69f297","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-12-15 16:34:33.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"0208728ba68bb16068cb8d85014ad139468974b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 17:08:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1258731d12ac49219abef10502acdda8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a30588630eab4fe18aa0b85066eb938e : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12d2ed06668d49f49fa318704d3ebf15 : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":6},{"id":"2d3e2842ccf8c5d03e308dbce9c4a33cb0b22b0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-12-18 08:37:33.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"d07d3730cb95cbf9300848f58fe4de9af9ea5374","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 08:49:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be7e4cb1b788430d837d26cb4b099d76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46e336d8304f46b9b6c04b42e49d05d9 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3524f6219c4e42f499b9e5b29e8d2cdf : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":7},{"id":"282511c49aa0ad5d5aac628c557582c030804c1a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-12-19 16:12:05.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"746af502792d1ab4247aab78572cce7a677ff3a2","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-12-19 17:03:11.000000000","message":"Patch Set 7: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"bcc3324aca00c8392c13e04c07ae9cd609b82094","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-12-19 17:34:23.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"43d5c5830a5bc6fadac9b219b73a876cd129a0cc","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-12-19 17:35:19.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"cf6b7f457c38f97a31eda199d9c61d70dbbbb9f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 17:44:44.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c60d6980c88942a191541bac719d5246\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e18090927f1c4ae0bff8033860a763f8 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f54439242cfb428bb8caca532875f0a0 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":8},{"id":"8f349edb6bc23a6348934c9d855a5bca05196193","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-12-19 18:39:30.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"a56a0413e4324e3fa9fb0bc64e3c7ac3ef5b2fc9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-12-20 10:28:48.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"2a2b8b4cc9b202d7413ee8e4723b6551418c5ca3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 10:29:11.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"eb06e8e13dfada871d0206b7a0cb0b440c936a4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 10:37:17.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6140f21bf7a84f65b54b0c515213dbc2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bbf9f1875ae4ea9b9962801bea198d7 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37cdd460651c49fb9d3e127e5ca7cefe : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":8},{"id":"b362b93544205e9e1d70a7ff5c58cbc980f5c0f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 10:37:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"6cfb32d48e2589d8a4fab17ac013601230674981","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 10:38:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/658e85a12d0e4bfda9d00c73df8c6afa\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/37b866effd864e8c995f5a6efab0e78b : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"24ac23166673e5169f3f396382c0bb7a472d978d","revisions":{"dc17ac82cb48d43a9152a6bcdf9e25bb872e28b9":{"kind":"REWORK","_number":1,"created":"2023-11-10 16:27:32.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/36/900636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/900636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/1"}}},"commit":{"parents":[{"commit":"f1ec291df71cf155b7cbf700c30be75006e4b204","subject":"Merge \"Add libvirt-dev-alias spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f1ec291df71cf155b7cbf700c30be75006e4b204"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-11-10 16:26:23.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-11-10 16:26:23.000000000","tz":60},"subject":"Proposes mdev live-migration support in libvirt","message":"Proposes mdev live-migration support in libvirt\n\nRelates to blueprint libvirt-mdev-live-migrate\n\nChange-Id: Iee16cb71d31e1308226adc0a39caef5d9970e00c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dc17ac82cb48d43a9152a6bcdf9e25bb872e28b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dc17ac82cb48d43a9152a6bcdf9e25bb872e28b9"}]},"branch":"refs/heads/master"},"14c07d2e494677c70feb0ee73b5ab818264131ed":{"kind":"REWORK","_number":2,"created":"2023-11-16 15:12:12.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/36/900636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/900636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/2"}}},"commit":{"parents":[{"commit":"f1ec291df71cf155b7cbf700c30be75006e4b204","subject":"Merge \"Add libvirt-dev-alias spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f1ec291df71cf155b7cbf700c30be75006e4b204"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-11-10 16:26:23.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-11-16 15:12:05.000000000","tz":60},"subject":"Proposes mdev live-migration support in libvirt","message":"Proposes mdev live-migration support in libvirt\n\nRelates to blueprint libvirt-mdev-live-migrate\n\nChange-Id: Iee16cb71d31e1308226adc0a39caef5d9970e00c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/14c07d2e494677c70feb0ee73b5ab818264131ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/14c07d2e494677c70feb0ee73b5ab818264131ed"}]},"branch":"refs/heads/master"},"40f97f592024d1dda9255b1560b9d955e05a457a":{"kind":"REWORK","_number":3,"created":"2023-11-16 16:20:44.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/36/900636/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/900636/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/3"}}},"commit":{"parents":[{"commit":"f1ec291df71cf155b7cbf700c30be75006e4b204","subject":"Merge \"Add libvirt-dev-alias spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f1ec291df71cf155b7cbf700c30be75006e4b204"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-11-10 16:26:23.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-11-16 16:20:34.000000000","tz":60},"subject":"Proposes mdev live-migration support in libvirt","message":"Proposes mdev live-migration support in libvirt\n\nRelates to blueprint libvirt-mdev-live-migrate\n\nChange-Id: Iee16cb71d31e1308226adc0a39caef5d9970e00c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/40f97f592024d1dda9255b1560b9d955e05a457a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/40f97f592024d1dda9255b1560b9d955e05a457a"}]},"branch":"refs/heads/master"},"d563a6bb92a959c6ca7573b205720a51c85312c3":{"kind":"REWORK","_number":4,"created":"2023-12-08 12:31:22.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/36/900636/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/900636/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/4"}}},"commit":{"parents":[{"commit":"f1ec291df71cf155b7cbf700c30be75006e4b204","subject":"Merge \"Add libvirt-dev-alias spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f1ec291df71cf155b7cbf700c30be75006e4b204"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-11-10 16:26:23.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-12-08 12:31:15.000000000","tz":60},"subject":"Proposes mdev live-migration support in libvirt","message":"Proposes mdev live-migration support in libvirt\n\nRelates to blueprint libvirt-mdev-live-migrate\n\nChange-Id: Iee16cb71d31e1308226adc0a39caef5d9970e00c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d563a6bb92a959c6ca7573b205720a51c85312c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d563a6bb92a959c6ca7573b205720a51c85312c3"}]},"branch":"refs/heads/master"},"7aba16ed50111406519b9273f8965e87b85f3f98":{"kind":"REWORK","_number":5,"created":"2023-12-08 12:38:55.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/36/900636/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/900636/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/5"}}},"commit":{"parents":[{"commit":"f1ec291df71cf155b7cbf700c30be75006e4b204","subject":"Merge \"Add libvirt-dev-alias spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f1ec291df71cf155b7cbf700c30be75006e4b204"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-11-10 16:26:23.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-12-08 12:38:48.000000000","tz":60},"subject":"Proposes mdev live-migration support in libvirt","message":"Proposes mdev live-migration support in libvirt\n\nRelates to blueprint libvirt-mdev-live-migrate\n\nChange-Id: Iee16cb71d31e1308226adc0a39caef5d9970e00c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7aba16ed50111406519b9273f8965e87b85f3f98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7aba16ed50111406519b9273f8965e87b85f3f98"}]},"branch":"refs/heads/master"},"e65b69bdbe93576d51c54e59740237dbac9b0dc0":{"kind":"REWORK","_number":6,"created":"2023-12-15 16:34:33.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/36/900636/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/900636/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/6"}}},"commit":{"parents":[{"commit":"f1ec291df71cf155b7cbf700c30be75006e4b204","subject":"Merge \"Add libvirt-dev-alias spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f1ec291df71cf155b7cbf700c30be75006e4b204"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-11-10 16:26:23.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-12-15 16:34:19.000000000","tz":60},"subject":"Proposes mdev live-migration support in libvirt","message":"Proposes mdev live-migration support in libvirt\n\nRelates to blueprint libvirt-mdev-live-migrate\n\nChange-Id: Iee16cb71d31e1308226adc0a39caef5d9970e00c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e65b69bdbe93576d51c54e59740237dbac9b0dc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e65b69bdbe93576d51c54e59740237dbac9b0dc0"}]},"branch":"refs/heads/master"},"34e6e746b6f6b3a9d4d35f9e2e2cfb9154e19341":{"kind":"REWORK","_number":7,"created":"2023-12-18 08:37:33.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/36/900636/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/900636/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/7"}}},"commit":{"parents":[{"commit":"f1ec291df71cf155b7cbf700c30be75006e4b204","subject":"Merge \"Add libvirt-dev-alias spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f1ec291df71cf155b7cbf700c30be75006e4b204"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-11-10 16:26:23.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-12-18 08:36:53.000000000","tz":60},"subject":"Proposes mdev live-migration support in libvirt","message":"Proposes mdev live-migration support in libvirt\n\nRelates to blueprint libvirt-mdev-live-migrate\n\nChange-Id: Iee16cb71d31e1308226adc0a39caef5d9970e00c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/34e6e746b6f6b3a9d4d35f9e2e2cfb9154e19341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/34e6e746b6f6b3a9d4d35f9e2e2cfb9154e19341"}]},"branch":"refs/heads/master"},"24ac23166673e5169f3f396382c0bb7a472d978d":{"kind":"REWORK","_number":8,"created":"2023-12-19 17:34:23.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/36/900636/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/900636/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/900636/8"}}},"commit":{"parents":[{"commit":"f1ec291df71cf155b7cbf700c30be75006e4b204","subject":"Merge \"Add libvirt-dev-alias spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f1ec291df71cf155b7cbf700c30be75006e4b204"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-11-10 16:26:23.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-12-19 17:34:07.000000000","tz":60},"subject":"Proposes mdev live-migration support in libvirt","message":"Proposes mdev live-migration support in libvirt\n\nRelates to blueprint libvirt-mdev-live-migrate\n\nChange-Id: Iee16cb71d31e1308226adc0a39caef5d9970e00c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/24ac23166673e5169f3f396382c0bb7a472d978d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/24ac23166673e5169f3f396382c0bb7a472d978d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
