)]}'
{"id":"openstack%2Fnova-specs~908297","triplet_id":"openstack%2Fnova-specs~master~I3e3c2350e623a95e3b2831818012376df0508232","project":"openstack/nova-specs","branch":"master","topic":"bp/libvirt-stateless-firmware","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-05-14 12:52:16.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-14 13:01:47.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2024-05-14 13:01:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3e3c2350e623a95e3b2831818012376df0508232","subject":"libvirt: Stateless firmware support","status":"MERGED","created":"2024-02-07 12:08:16.000000000","updated":"2024-05-14 13:02:56.000000000","submitted":"2024-05-14 13:01:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"908297-bp/libvirt-stateless-firmware","meta_rev_id":"481efe53c7aedf1c2c80ca2e60e3b1a4cc1b6552","_number":908297,"virtual_id_number":908297,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-14 13:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2024-05-14 12:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2024-05-14 12:52:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-07 12:14:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-13 09:13:21.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-05-14 11:40:33.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"fdb9bc266b784581ff216938fa89c867f2091775","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-07 12:08:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80049b226a477e3de6feddb4e0e3c96a86cada12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-07 12:08:38.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c7771ff44018f9c609b4be6f7cdb23ad1ff75f40","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-07 12:11:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"528b10e6aa3599d2ffd28c8e53c66e56914204bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 12:14:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/71768df49d0940539b079c8a3aaf3df7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/524443edd38449c59e8efe2c5d153c4f : FAILURE in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d99763ed3d6b4427bd00783628fe8401 : SUCCESS in 3m 17s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/2024.1/approved/libvirt-stateless-firmware.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/2024.1/approved/libvirt-stateless-firmware.rst","accounts_in_message":[],"_revision_number":2},{"id":"ba325f42c22f13694a1cb85620b9dc416f6f77c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-07 14:01:47.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6ad621a625c1283f355b97245abcfbc02cd3bc4c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-07 14:03:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c4f4cc10a23dcf87d4c5da94ff4e63a937441ee1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-07 14:03:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"04748fce87ad514a5e44477b29a92902d6773d67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 14:17:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c83d06ac54e434a99dd71e0dee78a93\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d029bddf233b43f191b76cd4cbc9882e : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/644bb970eac34708bd6b75abfe3c408b : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":4},{"id":"0a68e8c45098b3592ff23e84107aac17fb1059e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-07 14:44:29.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"7e402df210b7588b07c70bf80badfdf75582ec46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 14:57:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ae77fe8cd6840339019c7c31deeaa76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbdf4e947dd84847894078859306bb16 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0bfa4d58b724ec8a2a16c90b3c2d02c : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":5},{"id":"8e0742f4e34afc39b627a3e6112a633ec78c4e26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-13 07:36:49.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"dc0839c48edf483e1766b1314760e2de854db84c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 07:49:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3f45a1f468e41a7952572e750c3de7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/195770028c624e79b8a27ddca4e27737 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c51f89b98b8c49f98ab68e2303df1125 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":6},{"id":"bcfcfbf7a70b0fef70008692c4da3b80f08a01a3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-02-13 09:13:21.000000000","message":"Patch Set 6: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ec0d9ae1649e6485329c9fb2515742db7c1cc372","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-19 13:23:27.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"7daec0f628de40155345f19d61670cb70dd02347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 13:30:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/606c838d1a6e4f45970fd47cf2b947a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dfc5e48396d4c4185a70251ff13bc51 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/665d8087e5af4a69bea716a17ce601d7 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":7},{"id":"c5b5d7881ec96a4f131cb699fa3d9cb01988b889","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-05-14 11:40:33.000000000","message":"Patch Set 7: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4d080e3c78e77e6b0f24628186fbad2a35e45000","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-05-14 11:49:06.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7b978715fce5f96bc540929b5d4f43411576807b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-14 12:13:04.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"f3c23b90dd5ff84b2c69558dd438e19cd549651f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-14 12:13:21.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ec493dbfe426de34b31bf4acd70ec5ed033c5ee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 12:24:54.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e275ed14a0cc45e18d8a5cd83efbd2db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6c1d832709c4162b080aece8fe01862 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d668c046fa14467a892c5afddfdb74c : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":8},{"id":"bc368311c649cafe2ed07798899029179175a21d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-05-14 12:52:16.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"84f8e6744df847cea669431d55593ba6b64772cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 12:52:35.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"4980bca7b1945e280f4128e9c0a7002e9a3725c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 13:01:47.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b8712f50df646948b065eaf6aeca185\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ad76c0a8a054388b76a60c769811891 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcfe713c18b04b1d85132a777fd416e2 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":8},{"id":"1dfddc6c610a84a16704aa328849b3ad6ce2cf45","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 13:01:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"481efe53c7aedf1c2c80ca2e60e3b1a4cc1b6552","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 13:02:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f9f861e75044e679bbc3f6046ec0c41\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/1586fb41631a4732995186e4406803af : SUCCESS in 51s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"5f07083318d1f44d3310ced285c4a0fefb757000","revisions":{"350de765f688b19e1222313b79563ceefd4009ef":{"kind":"REWORK","_number":1,"created":"2024-02-07 12:08:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/908297/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/97/908297/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/1"}}},"commit":{"parents":[{"commit":"f0ffcb6ddfbbb4baccfa7b86e0a9176aca4db698","subject":"add the blueprint link to the specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f0ffcb6ddfbbb4baccfa7b86e0a9176aca4db698"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-07 12:06:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-07 12:06:50.000000000","tz":540},"subject":"libvirt: AMD statless firmware support","message":"libvirt: AMD statless firmware support\n\nNOTE:\nThis will be submitted for reference and early discussions. This should\nbe resubmitted to 2024.2 once spec proposal is open.\n\nblueprint: libvirt-stateless-firmware\nChange-Id: I3e3c2350e623a95e3b2831818012376df0508232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/350de765f688b19e1222313b79563ceefd4009ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/350de765f688b19e1222313b79563ceefd4009ef"}]},"branch":"refs/heads/master"},"1ffb340577ceb816f56d7da2343d76c5ec4b59e7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-07 12:08:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/908297/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/97/908297/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/2"}}},"commit":{"parents":[{"commit":"f0ffcb6ddfbbb4baccfa7b86e0a9176aca4db698","subject":"add the blueprint link to the specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f0ffcb6ddfbbb4baccfa7b86e0a9176aca4db698"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-07 12:06:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-07 12:08:23.000000000","tz":540},"subject":"libvirt: Statless firmware support","message":"libvirt: Statless firmware support\n\nNOTE:\nThis will be submitted for reference and early discussions. This should\nbe resubmitted to 2024.2 once spec proposal is open.\n\nblueprint: libvirt-stateless-firmware\nChange-Id: I3e3c2350e623a95e3b2831818012376df0508232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1ffb340577ceb816f56d7da2343d76c5ec4b59e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1ffb340577ceb816f56d7da2343d76c5ec4b59e7"}]},"branch":"refs/heads/master"},"e8def8b17cba4d47a3de28b1be9ed9ddd2fb7bb9":{"kind":"REWORK","_number":3,"created":"2024-02-07 14:01:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/908297/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/97/908297/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/3"}}},"commit":{"parents":[{"commit":"f0ffcb6ddfbbb4baccfa7b86e0a9176aca4db698","subject":"add the blueprint link to the specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f0ffcb6ddfbbb4baccfa7b86e0a9176aca4db698"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-07 12:06:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-07 14:01:36.000000000","tz":540},"subject":"libvirt: Statless firmware support","message":"libvirt: Statless firmware support\n\nNOTE:\nThis will be submitted for reference and early discussions. This should\nbe resubmitted to 2024.2 once spec proposal is open.\n\nblueprint: libvirt-stateless-firmware\nChange-Id: I3e3c2350e623a95e3b2831818012376df0508232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e8def8b17cba4d47a3de28b1be9ed9ddd2fb7bb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e8def8b17cba4d47a3de28b1be9ed9ddd2fb7bb9"}]},"branch":"refs/heads/master"},"ddaa95c0d9a811797aa090f9fabd88b88a4e52f4":{"kind":"REWORK","_number":4,"created":"2024-02-07 14:03:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/908297/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/97/908297/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/4"}}},"commit":{"parents":[{"commit":"f0ffcb6ddfbbb4baccfa7b86e0a9176aca4db698","subject":"add the blueprint link to the specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f0ffcb6ddfbbb4baccfa7b86e0a9176aca4db698"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-07 12:06:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-07 14:03:35.000000000","tz":540},"subject":"libvirt: Statless firmware support","message":"libvirt: Statless firmware support\n\nNOTE:\nThis will be submitted for reference and early discussions. This should\nbe resubmitted to 2024.2 once spec proposal is open.\n\nblueprint: libvirt-stateless-firmware\nChange-Id: I3e3c2350e623a95e3b2831818012376df0508232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ddaa95c0d9a811797aa090f9fabd88b88a4e52f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ddaa95c0d9a811797aa090f9fabd88b88a4e52f4"}]},"branch":"refs/heads/master"},"9495a9a62725f35f4ad7b63d24dba1c7ae3c1155":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-02-07 14:44:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/908297/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/97/908297/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/5"}}},"commit":{"parents":[{"commit":"f0ffcb6ddfbbb4baccfa7b86e0a9176aca4db698","subject":"add the blueprint link to the specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f0ffcb6ddfbbb4baccfa7b86e0a9176aca4db698"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-07 12:06:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-07 14:44:18.000000000","tz":540},"subject":"libvirt: Stateless firmware support","message":"libvirt: Stateless firmware support\n\nNOTE:\nThis will be submitted for reference and early discussions. This should\nbe resubmitted to 2024.2 once spec proposal is open.\n\nblueprint: libvirt-stateless-firmware\nChange-Id: I3e3c2350e623a95e3b2831818012376df0508232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9495a9a62725f35f4ad7b63d24dba1c7ae3c1155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9495a9a62725f35f4ad7b63d24dba1c7ae3c1155"}]},"branch":"refs/heads/master"},"23a0ea36165e0bb10ac4409ae725b7be9289bb5e":{"kind":"REWORK","_number":6,"created":"2024-02-13 07:36:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/908297/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/97/908297/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/6"}}},"commit":{"parents":[{"commit":"f0ffcb6ddfbbb4baccfa7b86e0a9176aca4db698","subject":"add the blueprint link to the specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f0ffcb6ddfbbb4baccfa7b86e0a9176aca4db698"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-07 12:06:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-13 07:36:33.000000000","tz":540},"subject":"libvirt: Stateless firmware support","message":"libvirt: Stateless firmware support\n\nNOTE:\nThis will be submitted for reference and early discussions. This should\nbe resubmitted to 2024.2 once spec proposal is open.\n\nblueprint: libvirt-stateless-firmware\nChange-Id: I3e3c2350e623a95e3b2831818012376df0508232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/23a0ea36165e0bb10ac4409ae725b7be9289bb5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/23a0ea36165e0bb10ac4409ae725b7be9289bb5e"}]},"branch":"refs/heads/master"},"85ff87682f6ac95ad91cfc21d49d00cb1c1ed3c5":{"kind":"REWORK","_number":7,"created":"2024-02-19 13:23:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/908297/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/97/908297/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/7"}}},"commit":{"parents":[{"commit":"80295ed3c4ef1a1b45cf5971949576479bfbca0a","subject":"Create specs directory for 2024.2 Dalmatian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/80295ed3c4ef1a1b45cf5971949576479bfbca0a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-07 12:06:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-19 13:23:15.000000000","tz":540},"subject":"libvirt: Stateless firmware support","message":"libvirt: Stateless firmware support\n\nblueprint: libvirt-stateless-firmware\nChange-Id: I3e3c2350e623a95e3b2831818012376df0508232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/85ff87682f6ac95ad91cfc21d49d00cb1c1ed3c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/85ff87682f6ac95ad91cfc21d49d00cb1c1ed3c5"}]},"branch":"refs/heads/master"},"5f07083318d1f44d3310ced285c4a0fefb757000":{"kind":"REWORK","_number":8,"created":"2024-05-14 12:13:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/908297/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/97/908297/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/97/908297/8"}}},"commit":{"parents":[{"commit":"80295ed3c4ef1a1b45cf5971949576479bfbca0a","subject":"Create specs directory for 2024.2 Dalmatian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/80295ed3c4ef1a1b45cf5971949576479bfbca0a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-07 12:06:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-14 12:12:53.000000000","tz":540},"subject":"libvirt: Stateless firmware support","message":"libvirt: Stateless firmware support\n\nblueprint: libvirt-stateless-firmware\nChange-Id: I3e3c2350e623a95e3b2831818012376df0508232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f07083318d1f44d3310ced285c4a0fefb757000"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f07083318d1f44d3310ced285c4a0fefb757000"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
