)]}'
{"id":"openstack%2Fnova-specs~913997","triplet_id":"openstack%2Fnova-specs~master~Idd9cb72992fe9cda3539d3ee5c0b0d8ea6628bf1","project":"openstack/nova-specs","branch":"master","topic":"bp/manila_shares_attachments","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2024-05-08 12:10:03.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2024-05-14 09:43:08.000000000","reason":"\u003cGERRIT_ACCOUNT_7166\u003e replied on the change","reason_account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2024-05-14 09:57:52.000000000","reason":"Change was submitted"},"2271":{"account":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"last_update":"2024-05-14 09:57:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd9cb72992fe9cda3539d3ee5c0b0d8ea6628bf1","subject":"Update and Re-propose \"Allow Manila shares to be directly attached to an instance when using libvirt\" for Dalmatian","status":"MERGED","created":"2024-03-22 16:13:10.000000000","updated":"2024-05-14 10:00:17.000000000","submitted":"2024-05-14 09:57:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"913997-bp/manila_shares_attachments","meta_rev_id":"0b3b4d4abd3bfda748ca8c8dfe827c004fa5ce5d","_number":913997,"virtual_id_number":913997,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-14 09:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":2,"date":"2024-05-13 08:57:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-14 09:43:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-14 09:43:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-22 16:13:56.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2024-03-22 16:13:56.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2024-03-22 16:28:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-06 09:10:53.000000000","updated_by":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"reviewer":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"state":"CC"}],"messages":[{"id":"6f51ce7aaff2bbf06cda50d3941b7a30037df9aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-03-22 16:13:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7df27a3f368ca977b1b04066d740f516a941fea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:28:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31be246cf5a54c4ebe3e79df484a4bba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4e7d1a571994df6a9ca3605c09766ab : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02617115ab2246b2a66941a97ccecf5d : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a236b7f7f714a5158d4f90e235d71bacff4c4931","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-04-04 13:54:19.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"28ea97c5af50fcfc773dc3dee1ac84940e16ef34","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-04-05 14:08:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"584c72dbf8555377e4ff4cc6f3a1466e7ee7ba8a","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2024-04-06 09:10:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88b2a97a8b9123667a3548a63099ac617a73782f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-04-30 10:06:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0d70470c46253b862fd6fe697207bf676314f316","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-04-30 10:06:56.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"29bc39ba133e0bf859cc678277bd10b6d140aada","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 10:22:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c7aec130e8f4dbc83acb2e580509b93\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e209678e61364f5aa523cf24a34d76d9 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f345aa6639714985b38f7b9c90e38a1d : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"73ff444334ecf3171e9fa9b488d35a13256eb034","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-04-30 11:12:01.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7c39d8cc9a1d2bfadedf2d4df40c9ae76d95009b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-04-30 12:57:33.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7187f08aee2470b6c87d4263a72ec8b7235239fe","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-05-02 08:50:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1599a353c6a79be08e8206876cbee16efdaa1e62","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-05-02 09:53:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a8cacbc226124cf07f97e8dc38003cdf642d18e2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-05-02 10:03:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"58b53ff072674ef8bcaf0778c17df18b7ed258a9","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-05-07 16:23:08.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d7a9ff698991acc00c5d5b3498393a2e6794d1b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-05-07 16:25:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"704c5e70f050dd03c5df901e244d8b93efbca453","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 16:37:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6d08fb8b36548a9bce6e88a3f57b88a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa71fe2866b4daea81c81855f058c72 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49690221b7ac42c3aa87d72e2f79a4bc : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b69a1672bfdc9277dd4ccf4a14ff3bef7207c1b8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-05-08 12:10:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3554ba129611b2e120bd0745956f086180869933","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-05-10 12:39:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7b0fbce204d5ca51720d346d66b213b75be2b331","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-05-10 13:31:53.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"419a49980b51c3fd817a838d136442c8ff5abff5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 13:45:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e54149a12664569b489786c8f5aec47\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e5d2498321243eaa07f3e102384bcdf : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fc4ddb25c3a4f37a6f03b516a3107b3 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":4},{"id":"45b13831762023c02898206f3922fbf5b0f19478","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-05-13 08:56:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"33f6b523879bde422f4f9d05709d1f55426fa2dd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-05-13 08:57:02.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"38962f7f595c16b1dade44aa06c05e0172ed16b0","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-05-14 09:43:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b592012233664d1acd88065aa2e3962379a55b89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 09:43:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0b84315a82dd14ed31c422a5500eed237002a145","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 09:57:52.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73fec8910d484403ac1d4a8977b8fe42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/640f6081a6f84858bf731f529d02259c : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdf6bf7db8e64708ae74d7c228ee82ee : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":4},{"id":"73cfd507b23c52c5e4bb042efc4435b15df912f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 09:57:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"0b3b4d4abd3bfda748ca8c8dfe827c004fa5ce5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 10:00:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5d37560733d4acba9d70f21137002ba\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/a323e106efef4cd891f00764a48f1d6c : SUCCESS in 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b9fc21f22a49e3ec74bdbdb5009b1182eb651919","revisions":{"8d5d8f050480e0f97fba27a7d9662f3fdb799a4d":{"kind":"REWORK","_number":1,"created":"2024-03-22 16:13:10.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/97/913997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/97/913997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/1"}}},"commit":{"parents":[{"commit":"ae9a0631eadf7d8e8184f4b40fbd6bf47eec4ead","subject":"Move Caracal implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ae9a0631eadf7d8e8184f4b40fbd6bf47eec4ead"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-03-22 15:56:47.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-03-22 15:56:47.000000000","tz":60},"subject":"Update and Re-propose \"Allow Manila shares to be directly attached to an instance when using libvirt\" for Dalmatian","message":"Update and Re-propose \"Allow Manila shares to be directly attached to an instance\nwhen using libvirt\" for Dalmatian\n\nImplements: blueprint libvirt-virtiofs-attach-manila-shares\nChange-Id: Idd9cb72992fe9cda3539d3ee5c0b0d8ea6628bf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d5d8f050480e0f97fba27a7d9662f3fdb799a4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d5d8f050480e0f97fba27a7d9662f3fdb799a4d"}]},"branch":"refs/heads/master"},"d0689f5710eb9e4fcb58334258dbd484e8791055":{"kind":"REWORK","_number":2,"created":"2024-04-30 10:06:44.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/97/913997/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/97/913997/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/2"}}},"commit":{"parents":[{"commit":"ae9a0631eadf7d8e8184f4b40fbd6bf47eec4ead","subject":"Move Caracal implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ae9a0631eadf7d8e8184f4b40fbd6bf47eec4ead"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-03-22 15:56:47.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-04-30 10:05:57.000000000","tz":120},"subject":"Update and Re-propose \"Allow Manila shares to be directly attached to an instance when using libvirt\" for Dalmatian","message":"Update and Re-propose \"Allow Manila shares to be directly attached to an instance\nwhen using libvirt\" for Dalmatian\n\nImplements: blueprint libvirt-virtiofs-attach-manila-shares\nChange-Id: Idd9cb72992fe9cda3539d3ee5c0b0d8ea6628bf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d0689f5710eb9e4fcb58334258dbd484e8791055"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d0689f5710eb9e4fcb58334258dbd484e8791055"}]},"branch":"refs/heads/master"},"a26a47f59c5ecd619a6fe70630d6d8b0d25c55ef":{"kind":"REWORK","_number":3,"created":"2024-05-07 16:25:00.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/97/913997/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/97/913997/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/3"}}},"commit":{"parents":[{"commit":"ae9a0631eadf7d8e8184f4b40fbd6bf47eec4ead","subject":"Move Caracal implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ae9a0631eadf7d8e8184f4b40fbd6bf47eec4ead"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-03-22 15:56:47.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-05-07 16:23:30.000000000","tz":120},"subject":"Update and Re-propose \"Allow Manila shares to be directly attached to an instance when using libvirt\" for Dalmatian","message":"Update and Re-propose \"Allow Manila shares to be directly attached to an instance\nwhen using libvirt\" for Dalmatian\n\nImplements: blueprint libvirt-virtiofs-attach-manila-shares\nChange-Id: Idd9cb72992fe9cda3539d3ee5c0b0d8ea6628bf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a26a47f59c5ecd619a6fe70630d6d8b0d25c55ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a26a47f59c5ecd619a6fe70630d6d8b0d25c55ef"}]},"branch":"refs/heads/master"},"b9fc21f22a49e3ec74bdbdb5009b1182eb651919":{"kind":"REWORK","_number":4,"created":"2024-05-10 13:31:53.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/97/913997/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/97/913997/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/97/913997/4"}}},"commit":{"parents":[{"commit":"ae9a0631eadf7d8e8184f4b40fbd6bf47eec4ead","subject":"Move Caracal implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ae9a0631eadf7d8e8184f4b40fbd6bf47eec4ead"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-03-22 15:56:47.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-05-10 13:31:14.000000000","tz":120},"subject":"Update and Re-propose \"Allow Manila shares to be directly attached to an instance when using libvirt\" for Dalmatian","message":"Update and Re-propose \"Allow Manila shares to be directly attached to an instance\nwhen using libvirt\" for Dalmatian\n\nImplements: blueprint libvirt-virtiofs-attach-manila-shares\nChange-Id: Idd9cb72992fe9cda3539d3ee5c0b0d8ea6628bf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b9fc21f22a49e3ec74bdbdb5009b1182eb651919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b9fc21f22a49e3ec74bdbdb5009b1182eb651919"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
