)]}'
{"id":"openstack%2Fnova-specs~923807","triplet_id":"openstack%2Fnova-specs~master~I1b001364475e86e7f37f462dddacbdd1278dcef5","project":"openstack/nova-specs","branch":"master","topic":"bp/unified-limits-nova-unset-limits","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2024-07-16 15:10:46.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-07-16 15:10:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1b001364475e86e7f37f462dddacbdd1278dcef5","subject":"Propose config option to control behavior of unset unified limits","status":"MERGED","created":"2024-07-10 03:38:53.000000000","updated":"2024-07-16 20:10:25.000000000","submitted":"2024-07-16 15:10:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"923807-bp/unified-limits-nova-unset-limits","meta_rev_id":"ac6990e91b096b8bfc624da509b845f56b451407","_number":923807,"virtual_id_number":923807,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-16 15:10:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2024-07-16 12:46:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-16 15:00:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-16 15:00:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-10 03:45:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-16 12:46:39.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-07-16 15:00:09.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"64feebc1c26c2c15c9c64293588bbaed869dcc95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-07-10 03:38:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de46b36dcdbf636667ecdbfa9ecb665ffda36fae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 03:45:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50584332e8e14e46811d22e4add01834\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b5b7db74d754d0cbe5c426e6f35db12 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16818a40d32e4c9092ceff40224ef92c : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"fcebb7b3b5d1bb96d7f96934951693f8b7384ade","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-07-10 15:57:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"580bfc902019edfc6a3505e670cd680eaef8678d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-07-10 19:45:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5ad48812eee2f81261378baeb737938ff67809f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-07-13 02:02:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"91abac7f3f96b5f4b3eb701b7494eb97dac7696e","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-07-13 02:03:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84b29dec23c48d6fa7c508bf4ab3d0660ad6882f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-07-13 02:04:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9c2e1df3330d51fbafd5f016af5d22b40a0828f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 02:55:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/344fa3f453364736a6842fe9b2a73be8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a56c8506b3744d589da63d11e45c2fb : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/595d3f4925564353a2366481c7f27bb9 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":3},{"id":"b60f067bbb5b1853770e33e53b2da9aa61a7071f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-07-16 12:46:39.000000000","message":"Patch Set 3: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0e4009a6ff16e064c83f287b1e12892a5331cd46","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-07-16 15:00:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"072df98d5993b90ebdea10110c6c4e6b2c00e2d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 15:00:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d1c9e411aec738d764a061234cf827c587713c50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 15:10:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0766748b85694de4b462e814ac47aca7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d2341808b4941399aa5b5657f7e06e6 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cde49fbaee8041cda9c65df4e2c6c15b : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":3},{"id":"9995f6126c622b73d9ce3dd1df9050b4943a6815","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 15:10:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"00d71e56ed15498635763a8f4d8b8b202c4bb68d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 15:12:06.000000000","message":"Patch Set 3:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e44ab453f2941bda565e74ed09f17fa\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/6a6f5c3fff4840c29d8f9b23b3a5280a : FAILURE in 1m 01s","accounts_in_message":[],"_revision_number":3},{"id":"ac6990e91b096b8bfc624da509b845f56b451407","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-07-16 20:10:25.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"190e4cf9015a1c944f4283a9d99365f11a6214a3","revisions":{"ca4f10f592ad6ec54254c71ab2f0585b7217c7f4":{"kind":"REWORK","_number":1,"created":"2024-07-10 03:38:53.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/07/923807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/07/923807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/923807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/923807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/923807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/07/923807/1"}}},"commit":{"parents":[{"commit":"353ce30714c4d71ef65e210593cecc577ce5cfcb","subject":"Merge \"Follow-up for openapi spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/353ce30714c4d71ef65e210593cecc577ce5cfcb"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-07-10 03:08:59.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-07-10 03:31:21.000000000","tz":0},"subject":"Propose config option for unified limits of zero","message":"Propose config option for unified limits of zero\n\nThis proposes adding a config option to control whether a unified limit\nof zero should be considered as unlimited in Nova.\n\nRelated to blueprint unified-limits-nova-zero-as-unlimited\n\nChange-Id: I1b001364475e86e7f37f462dddacbdd1278dcef5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ca4f10f592ad6ec54254c71ab2f0585b7217c7f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ca4f10f592ad6ec54254c71ab2f0585b7217c7f4"}]},"branch":"refs/heads/master"},"b12bd38ddb0f35a514f9c47bbe1c36e8126fdfd2":{"kind":"REWORK","_number":2,"created":"2024-07-13 02:02:31.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/07/923807/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/07/923807/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/923807/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/923807/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/923807/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/07/923807/2"}}},"commit":{"parents":[{"commit":"353ce30714c4d71ef65e210593cecc577ce5cfcb","subject":"Merge \"Follow-up for openapi spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/353ce30714c4d71ef65e210593cecc577ce5cfcb"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-07-10 03:08:59.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-07-13 02:01:29.000000000","tz":0},"subject":"Propose config option to control behavior of unset unified limits","message":"Propose config option to control behavior of unset unified limits\n\nThis proposes adding a config option to control whether an unset\nunified limit should be considered as unlimited in Nova.\n\nRelated to blueprint unified-limits-nova-unset-limits\n\nChange-Id: I1b001364475e86e7f37f462dddacbdd1278dcef5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b12bd38ddb0f35a514f9c47bbe1c36e8126fdfd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b12bd38ddb0f35a514f9c47bbe1c36e8126fdfd2"}]},"branch":"refs/heads/master"},"190e4cf9015a1c944f4283a9d99365f11a6214a3":{"kind":"REWORK","_number":3,"created":"2024-07-13 02:04:14.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/07/923807/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/07/923807/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/923807/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/923807/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/923807/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/07/923807/3"}}},"commit":{"parents":[{"commit":"353ce30714c4d71ef65e210593cecc577ce5cfcb","subject":"Merge \"Follow-up for openapi spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/353ce30714c4d71ef65e210593cecc577ce5cfcb"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-07-10 03:08:59.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-07-13 02:04:04.000000000","tz":0},"subject":"Propose config option to control behavior of unset unified limits","message":"Propose config option to control behavior of unset unified limits\n\nThis proposes adding a config option to control whether an unset\nunified limit should be considered as unlimited in Nova.\n\nRelated to blueprint unified-limits-nova-unset-limits\n\nChange-Id: I1b001364475e86e7f37f462dddacbdd1278dcef5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/190e4cf9015a1c944f4283a9d99365f11a6214a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/190e4cf9015a1c944f4283a9d99365f11a6214a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
