)]}'
{"id":"openstack%2Fnova-specs~932653","triplet_id":"openstack%2Fnova-specs~master~Id7e3c0a8bc6509edc94d2e58be4611e5c6a416c2","project":"openstack/nova-specs","branch":"master","topic":"bp/add-burst-length-support-to-cinder-qos","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-01-15 14:47:16.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-01-14 21:42:02.000000000","reason":"\u003cGERRIT_ACCOUNT_37356\u003e replied on the change","reason_account":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"}},"37356":{"account":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"last_update":"2025-05-01 15:28:08.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-01-13 17:44:24.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-05-01 15:28:08.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"hashtags":[],"change_id":"Id7e3c0a8bc6509edc94d2e58be4611e5c6a416c2","subject":"Add Burst Length Support to Cinder QoS","status":"NEW","created":"2024-10-17 20:00:19.000000000","updated":"2025-05-15 19:31:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":41,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"eac4a3c030bbb7a3ee9ac197b589b95cb051edd2","_number":932653,"virtual_id_number":932653,"owner":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-19 19:38:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":-1,"date":"2025-05-15 19:31:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-17 20:14:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-24 03:28:48.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-10-25 14:17:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-01-08 15:13:45.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-01-08 16:06:26.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"}],"messages":[{"id":"d4a8e6ec84c4db195b9a0e2e84099cd212a32733","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"date":"2024-10-17 20:00:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8172280d30079ac3b00b43784528dfaba945dacb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 20:14:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0830c7c033094a348c602fdfb69dcd48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e53adbdb2a0945ca9cbfe448a62aad60 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73a4af7cedbe4a1ca1de2d61efdf098f : FAILURE in 2m 34s","accounts_in_message":[],"_revision_number":1},{"id":"50abaa2666973028b05276f7c890bebaae265646","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"date":"2024-10-17 20:25:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c05a23178caef4a6d3a3b8b0a9797fd0dc41fc3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 20:32:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b95232bc10634c93940bd67adf634679\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ed4db0fa2440908671547d25bf0071 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/141f2e3cd2f44fd7a3c3e33327181d20 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":2},{"id":"428fa3700f7c6eff94058c3ff3da579569d7e259","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"date":"2024-10-23 15:16:29.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"de1ecc130c648f7e65ee83f329f8a626fcb92361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 15:23:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4928b239b90748fdaa139d2a7202afa5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2044a88be374108a469713637f5eb79 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9f104b8548045c7a330a5dfc1b4bf93 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":3},{"id":"9c9982725338ecf42bc11ca58023ed69f5b2096e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-10-24 03:28:48.000000000","message":"Patch Set 3: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0b2d6245559bbd568eb7157537e4be3a262a3442","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"date":"2024-10-24 17:45:22.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a3caaf0aeb0eed3fa6553b3d0d2c45016a8074e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 17:52:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6752351d82f2432d9675fcb92ef06fdc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cecca15678e840f6bb5c2c39c5bcde03 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80177666f909452b95908e6e400db29e : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":4},{"id":"8a1577466140ff4d2431adfb545f9271987cd69c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-25 14:17:38.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"565710ed1feeb08bdb8c44da26f848830e3c6368","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-25 14:25:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"70f1d9a01dcee57e67d7a06f8de28ad9ec4e2173","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-29 16:49:56.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"59385b35767caeb0eb48c6c36a060f3874e656c5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-10-29 16:53:57.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6232b670aa9bbb7122a16446c23e24911f34cf45","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-10-29 16:56:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"87d6af0856d897cfb0bbae31442896f1c8395d30","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-29 17:06:34.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5828eddf85abfa04933a19d7253ecc2400c04f6a","author":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"date":"2024-12-17 19:33:57.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"469eab590d6d54eca04dd9c216b47ee62ae63293","author":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"date":"2024-12-18 18:56:02.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"483d055b84cc688a1e1be9fb806cea3f60a9e903","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-01-08 15:13:45.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ed55f63aa53559be376377e87969e2a61567bf8f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-01-08 16:06:26.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"771dc3d6d9c4662da30df9750205ef87c218a9dc","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-13 17:44:24.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4d8efbd81310561c2d1927b7d5d99916664c5dff","author":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"date":"2025-01-13 21:21:39.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"83c25887aaf1049a7f2e25e06405a5684ddf15b1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-14 16:51:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0b6f533c68e5a23639426321965f4340ae3d07d4","author":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"date":"2025-01-14 21:42:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"420d5bdbf8fa6feb970c7dad5df6c263dd463c2f","author":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"date":"2025-01-14 21:43:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d60a28ee2979728e2ff3b826aae07b29cf792683","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"date":"2025-01-15 14:47:16.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8f84fcb708b0f730a43f4994afd59664e882d7b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 14:54:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bebbf0faacde4bd3b28cdc630fc6a337\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a36dff0add94570ad4cdfc8e43894ba : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08f9b29759cc406cb18da864f196a2d7 : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":5},{"id":"637c81615bf6f81358043d68289933c343722084","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"date":"2025-01-16 20:26:30.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f36ca7d81155f29c6c4478e63eca317cb8346030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 20:37:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98928807a33d4b8aa30479e6ac7e5f5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f535900dc04eadab3bb9a7051c8c0c : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84d10a82d1144bca9d95a8fc9a40ea3c : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":6},{"id":"086e517a0fbea1b035d33683c9da29bc6a17129f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"date":"2025-01-16 20:43:18.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"4ba2b46229a74085f09b7a7e4f928cf62455d68d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 20:50:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08c08190e92640e18717aeda1705eedb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50a7fb9f5c9d40e39695fcbcc4789373 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7568b674dd86449d8f03f5c16ae90340 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":7},{"id":"4c51bcba9e0c59b81cf5830e99ecc36870b6f174","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-02-04 17:14:37.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"198742c0ffa0d1693f13e499042a863ce26c7438","author":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"date":"2025-02-04 21:06:54.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b93eba12b102756b69fd1a7581b86b82d2c961ef","author":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"date":"2025-02-19 14:34:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"56b63642ee6dbfc417bf3542202cce01dc61fdab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"date":"2025-02-19 19:30:02.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"9acbaaedab1f71978aff085674044616d17479d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 19:38:26.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7a39b621fb74a6294747836bd32ba47\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e28b23d2ae0453b9d8967f63509a8ee : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3291e7e747da4220bd474295da6e40be : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":8},{"id":"f65d254aa2d4f11a7ed41d3028054468ba684011","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-01 15:28:08.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"eac4a3c030bbb7a3ee9ac197b589b95cb051edd2","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-15 19:31:00.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"64e09e04f3a0ffd3b6df558fc8f106cc151516c4","revisions":{"a76dbb46085ac730e4e47d5029666badaa336e44":{"kind":"REWORK","_number":1,"created":"2024-10-17 20:00:19.000000000","uploader":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"ref":"refs/changes/53/932653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/53/932653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/1"}}},"commit":{"parents":[{"commit":"0cb32c15886f2beb61ed68e01b26b4deea7de4b7","subject":"Fix typos in release name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0cb32c15886f2beb61ed68e01b26b4deea7de4b7"}]}],"author":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2024-10-17 18:53:51.000000000","tz":-240},"committer":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2024-10-17 19:16:17.000000000","tz":-240},"subject":"Add Burst Length Support to Cinder QoS","message":"Add Burst Length Support to Cinder QoS\n\nQEMU 2.6 and Libvirt 2.4.0 introduce new iotune parameters that enhance\nthe flexibility of block storage quality of service (QoS). Specifically,\nthey enable the configuration of burst lengths for read, write, and\noverall IOPS and bandwidth. This blueprint proposes to incorporate these\nparameters into Cinder QoS specifications for connected volumes.\n\nChange-Id: Id7e3c0a8bc6509edc94d2e58be4611e5c6a416c2\nSigned-off-by: Mengyang Zhang \u003cmzhang741@bloomberg.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a76dbb46085ac730e4e47d5029666badaa336e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a76dbb46085ac730e4e47d5029666badaa336e44"}]},"branch":"refs/heads/master"},"9bd381746503403eed08021df8f26f2c1eaebaf2":{"kind":"REWORK","_number":2,"created":"2024-10-17 20:25:28.000000000","uploader":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"ref":"refs/changes/53/932653/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/53/932653/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/2"}}},"commit":{"parents":[{"commit":"0cb32c15886f2beb61ed68e01b26b4deea7de4b7","subject":"Fix typos in release name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0cb32c15886f2beb61ed68e01b26b4deea7de4b7"}]}],"author":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2024-10-17 18:53:51.000000000","tz":-240},"committer":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2024-10-17 20:25:04.000000000","tz":-240},"subject":"Add Burst Length Support to Cinder QoS","message":"Add Burst Length Support to Cinder QoS\n\nQEMU 2.6 and Libvirt 2.4.0 introduce new iotune parameters that enhance\nthe flexibility of block storage quality of service (QoS). Specifically,\nthey enable the configuration of burst lengths for read, write, and\noverall IOPS and bandwidth. This blueprint proposes to incorporate these\nparameters into Cinder QoS specifications for connected volumes.\n\nChange-Id: Id7e3c0a8bc6509edc94d2e58be4611e5c6a416c2\nSigned-off-by: Mengyang Zhang \u003cmzhang741@bloomberg.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9bd381746503403eed08021df8f26f2c1eaebaf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9bd381746503403eed08021df8f26f2c1eaebaf2"}]},"branch":"refs/heads/master"},"bf0ec0bc0487393519ca3eda00c4fb8bd7096045":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-10-23 15:16:29.000000000","uploader":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"ref":"refs/changes/53/932653/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/53/932653/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/3"}}},"commit":{"parents":[{"commit":"0cb32c15886f2beb61ed68e01b26b4deea7de4b7","subject":"Fix typos in release name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0cb32c15886f2beb61ed68e01b26b4deea7de4b7"}]}],"author":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2024-10-17 18:53:51.000000000","tz":-240},"committer":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2024-10-23 15:15:11.000000000","tz":-240},"subject":"Add Burst Length Support to Cinder QoS","message":"Add Burst Length Support to Cinder QoS\n\nblueprint: add-burst-length-support\n\nQEMU 2.6 and Libvirt 2.4.0 introduce new iotune parameters that enhance\nthe flexibility of block storage quality of service (QoS). Specifically,\nthey enable the configuration of burst lengths for read, write, and\noverall IOPS and bandwidth. This blueprint proposes to incorporate these\nparameters into Cinder QoS specifications for connected volumes.\n\nChange-Id: Id7e3c0a8bc6509edc94d2e58be4611e5c6a416c2\nSigned-off-by: Mengyang Zhang \u003cmzhang741@bloomberg.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bf0ec0bc0487393519ca3eda00c4fb8bd7096045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bf0ec0bc0487393519ca3eda00c4fb8bd7096045"}]},"branch":"refs/heads/master"},"e5b0d172b89e8218148ac31a3a4122bc2a4cc9ea":{"kind":"REWORK","_number":4,"created":"2024-10-24 17:45:22.000000000","uploader":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"ref":"refs/changes/53/932653/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/53/932653/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/4"}}},"commit":{"parents":[{"commit":"0cb32c15886f2beb61ed68e01b26b4deea7de4b7","subject":"Fix typos in release name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0cb32c15886f2beb61ed68e01b26b4deea7de4b7"}]}],"author":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2024-10-17 18:53:51.000000000","tz":-240},"committer":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2024-10-24 17:45:09.000000000","tz":-240},"subject":"Add Burst Length Support to Cinder QoS","message":"Add Burst Length Support to Cinder QoS\n\nblueprint: add-burst-length-support\n\nQEMU 2.6 and Libvirt 2.4.0 introduce new iotune parameters that enhance\nthe flexibility of block storage quality of service (QoS). Specifically,\nthey enable the configuration of burst lengths for read, write, and\noverall IOPS and bandwidth. This blueprint proposes to incorporate these\nparameters into Cinder QoS specifications for connected volumes.\n\nChange-Id: Id7e3c0a8bc6509edc94d2e58be4611e5c6a416c2\nSigned-off-by: Mengyang Zhang \u003cmzhang741@bloomberg.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e5b0d172b89e8218148ac31a3a4122bc2a4cc9ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e5b0d172b89e8218148ac31a3a4122bc2a4cc9ea"}]},"branch":"refs/heads/master"},"ccdfd0d388a6ca80f299b7f58d2b445aedfb8ae7":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-01-15 14:47:16.000000000","uploader":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"ref":"refs/changes/53/932653/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/53/932653/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/5"}}},"commit":{"parents":[{"commit":"0cb32c15886f2beb61ed68e01b26b4deea7de4b7","subject":"Fix typos in release name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0cb32c15886f2beb61ed68e01b26b4deea7de4b7"}]}],"author":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2024-10-17 18:53:51.000000000","tz":-240},"committer":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2025-01-15 14:47:16.000000000","tz":0},"subject":"Add Burst Length Support to Cinder QoS","message":"Add Burst Length Support to Cinder QoS\n\nblueprint: add-burst-length-support\n\nQEMU 2.6 and Libvirt 2.4.0 introduce new iotune parameters that enhance the flexibility of block storage quality of service (QoS). Specifically, they enable the configuration of burst lengths for read, write, and overall IOPS and bandwidth.\n\nCurrently, the Nova libvirt driver supports setting burst limits but does not provide the option to configure burst length. As a result, burst limits can only be applied for one second as QEMU defaults the burst length to this duration, reducing their effectiveness. This blueprint aims to incorporate burst length parameters into the existing QoS settings to allow for more flexible resource management.\n\nChange-Id: Id7e3c0a8bc6509edc94d2e58be4611e5c6a416c2\nSigned-off-by: Mengyang Zhang \u003cmzhang741@bloomberg.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ccdfd0d388a6ca80f299b7f58d2b445aedfb8ae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ccdfd0d388a6ca80f299b7f58d2b445aedfb8ae7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fc9da9eb07f089be6071a94f43b0efb515c141b0":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-01-16 20:26:30.000000000","uploader":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"ref":"refs/changes/53/932653/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/53/932653/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/6"}}},"commit":{"parents":[{"commit":"0cb32c15886f2beb61ed68e01b26b4deea7de4b7","subject":"Fix typos in release name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0cb32c15886f2beb61ed68e01b26b4deea7de4b7"}]}],"author":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2024-10-17 18:53:51.000000000","tz":-240},"committer":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2025-01-16 20:26:30.000000000","tz":0},"subject":"Add Burst Length Support to Cinder QoS","message":"Add Burst Length Support to Cinder QoS\n\nblueprint: add-burst-length-support\n\nQEMU 2.6 and Libvirt 2.4.0 introduce new iotune parameters that enhance the flexibility of block storage quality of service (QoS). Specifically, they enable the configuration of burst lengths for read, write, and overall IOPS and bandwidth.\n\nCurrently, the Nova libvirt driver supports setting burst limits but does not provide the option to configure burst length. As a result, burst limits can only be applied for one second as QEMU defaults the burst length to this duration, reducing their effectiveness.\n\nThis blueprint aims to incorporate burst length parameters into the existing QoS settings to allow for more flexible resource management.\n\nWith the changes, an admin can set front-end QoS specifications with burst limits and lengths for a specific volume type via the following openstack command:\n\n  openstack volume qos create \\\n    --consumer \"front-end\" \\\n    --property \"read_bytes_sec_max\u003d100000000\" \\\n    --property \"read_bytes_sec_max_length\u003d60\" \\\n    \u003cqos_name\u003e\n  openstack volume qos associate \u003cqos_name\u003e --volume-type \u003cvolume_type\u003e\n\nWhen a volume of this type attached to a VM, these values would be populated into the libvirt xml file under the iotune session.\n\n  \u003ciotune\u003e\n     \u003cread_bytes_sec\u003e1024000\u003c/read_bytes_sec\u003e\n     \u003cread_bytes_sec_max_length\u003e60\u003c/read_bytes_sec_max_length\u003e\n  \u003c/iotune\u003e\n\nThe complete list of supported burst length parameters is as follow.\nAvailable Additional QoS Specs:\n\nEach specification will add an extra line to the libvirt XML definition,\nspecifically within the \u003ciotune\u003e section for each device:\n\n* **read_bytes_sec_max_length**:\n\n  Maximum duration in seconds for the read_bytes_sec_max burst period (valid\n  only if read_bytes_sec_max is set).\n\n  ``\u003cread_bytes_sec_max_length\u003evalue\u003c/read_bytes_sec_max_length\u003e``\n\n\n* **write_bytes_sec_max_length**:\n\n  Maximum duration in seconds for the write_bytes_sec_max burst period (valid\n  only if write_bytes_sec_max is set).\n\n  ``\u003cwrite_bytes_sec_max_length\u003evalue\u003c/write_bytes_sec_max_length\u003e``\n\n\n* **total_bytes_sec_max_length**:\n\n  Maximum duration in seconds for the total_bytes_sec_max burst period (valid\n  only if total_bytes_sec_max is set).\n\n  ``\u003ctotal_bytes_sec_max_length\u003evalue\u003c/total_bytes_sec_max_length\u003e``\n\n* **read_iops_sec_max_length**:\n\n  Maximum duration in seconds for the read_iops_sec_max burst period (valid\n  only if read_iops_sec_max is set).\n\n  ``\u003cread_iops_sec_max_length\u003evalue\u003c/read_iops_sec_max_length\u003e``\n\n\n* **write_iops_sec_max_length**:\n\n  Maximum duration in seconds for the write_iops_sec_max burst period (valid\n  only if write_iops_sec_max is set).\n\n  ``\u003cwrite_iops_sec_max_length\u003evalue\u003c/write_iops_sec_max_length\u003e``\n\n\n* **total_iops_sec_max_length**:\n\n  Maximum duration in seconds for the total_iops_sec_max burst period (valid\n  only if total_iops_sec_max is set).\n\n  ``\u003ctotal_iops_sec_max_length\u003evalue\u003c/total_iops_sec_max_length\u003e``\n\nChange-Id: Id7e3c0a8bc6509edc94d2e58be4611e5c6a416c2\nSigned-off-by: Mengyang Zhang \u003cmzhang741@bloomberg.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fc9da9eb07f089be6071a94f43b0efb515c141b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fc9da9eb07f089be6071a94f43b0efb515c141b0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a5adc75200fe2c57b77197ab50239cdd0e98dff5":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-01-16 20:43:18.000000000","uploader":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"ref":"refs/changes/53/932653/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/53/932653/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/7"}}},"commit":{"parents":[{"commit":"0cb32c15886f2beb61ed68e01b26b4deea7de4b7","subject":"Fix typos in release name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0cb32c15886f2beb61ed68e01b26b4deea7de4b7"}]}],"author":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2024-10-17 18:53:51.000000000","tz":-240},"committer":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2025-01-16 20:43:18.000000000","tz":0},"subject":"Add Burst Length Support to Cinder QoS","message":"Add Burst Length Support to Cinder QoS\n\nblueprint: add-burst-length-support\n\nQEMU 2.6 and Libvirt 2.4.0 introduce new iotune parameters that enhance the flexibility of block storage quality of service (QoS). Specifically, they enable the configuration of burst lengths for read, write, and overall IOPS and bandwidth.\n\nCurrently, the Nova libvirt driver supports setting burst limits but does not provide the option to configure burst length. As a result, burst limits can only be applied for one second as QEMU defaults the burst length to this duration, reducing their effectiveness.\n\nThis blueprint aims to incorporate burst length parameters into the existing QoS settings to allow for more flexible resource management.\n\nWith these changes, users can set front-end QoS specifications with burst limits and lengths for a specific volume type using the following openstack commands:\n\n  openstack volume qos create \\\n    --consumer \"front-end\" \\\n    --property \"read_bytes_sec_max\u003d100000000\" \\\n    --property \"read_bytes_sec_max_length\u003d60\" \\\n    \u003cqos_name\u003e\n\n  openstack volume qos associate \u003cqos_name\u003e --volume-type \\\n    \u003cvolume_type\u003e\n\nWhen a volume of this type is attached to a VM, these values will be populated into the libvirt XML file under the iotune session:\n\n  \u003ciotune\u003e\n     \u003cread_bytes_sec\u003e1024000\u003c/read_bytes_sec\u003e\n     \u003cread_bytes_sec_max_length\u003e60\u003c/read_bytes_sec_max_length\u003e\n  \u003c/iotune\u003e\n\nThe complete list of supported burst length specifications is as follow. Each specification will add an extra line to the libvirt XML definition,\nspecifically within the \u003ciotune\u003e section for each device:\n\n- read_bytes_sec_max_length:\n  Maximum duration in seconds for the read_bytes_sec_max burst period\n  (valid only if read_bytes_sec_max is set).\n\n  \u003cread_bytes_sec_max_length\u003evalue\u003c/read_bytes_sec_max_length\u003e\n\n- write_bytes_sec_max_length:\n  Maximum duration in seconds for the write_bytes_sec_max burst period\n  (valid only if write_bytes_sec_max is set).\n\n  \u003cwrite_bytes_sec_max_length\u003evalue\u003c/write_bytes_sec_max_length\u003e\n\n- total_bytes_sec_max_length:\n  Maximum duration in seconds for the total_bytes_sec_max burst period\n  (valid only if total_bytes_sec_max is set).\n\n  \u003ctotal_bytes_sec_max_length\u003evalue\u003c/total_bytes_sec_max_length\u003e\n\n- read_iops_sec_max_length:\n  Maximum duration in seconds for the read_iops_sec_max burst period\n  (valid only if read_iops_sec_max is set).\n\n  \u003cread_iops_sec_max_length\u003evalue\u003c/read_iops_sec_max_length\u003e\n\n\n- write_iops_sec_max_length:\n  Maximum duration in seconds for the write_iops_sec_max burst period\n  (valid only if write_iops_sec_max is set).\n\n  \u003cwrite_iops_sec_max_length\u003evalue\u003c/write_iops_sec_max_length\u003e\n\n\n- total_iops_sec_max_length:\n  Maximum duration in seconds for the total_iops_sec_max burst period\n  (valid only if total_iops_sec_max is set).\n\n  \u003ctotal_iops_sec_max_length\u003evalue\u003c/total_iops_sec_max_length\u003e\n\nChange-Id: Id7e3c0a8bc6509edc94d2e58be4611e5c6a416c2\nSigned-off-by: Mengyang Zhang \u003cmzhang741@bloomberg.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a5adc75200fe2c57b77197ab50239cdd0e98dff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a5adc75200fe2c57b77197ab50239cdd0e98dff5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"64e09e04f3a0ffd3b6df558fc8f106cc151516c4":{"kind":"REWORK","_number":8,"created":"2025-02-19 19:30:02.000000000","uploader":{"_account_id":37356,"name":"Mengyang Zhang","display_name":"Mengyang","email":"mzhang741@bloomberg.net","username":"mzhang741"},"ref":"refs/changes/53/932653/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/53/932653/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/53/932653/8"}}},"commit":{"parents":[{"commit":"ee3b8dd2b1509dece7351488398dff879495403e","subject":"Fix a line that was not removed before merging the specification","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee3b8dd2b1509dece7351488398dff879495403e"}]}],"author":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2024-10-17 18:53:51.000000000","tz":-240},"committer":{"name":"Mengyang Zhang","email":"mzhang741@bloomberg.net","date":"2025-02-19 19:28:22.000000000","tz":-300},"subject":"Add Burst Length Support to Cinder QoS","message":"Add Burst Length Support to Cinder QoS\n\nblueprint: add-burst-length-support\n\nQEMU 2.6 and Libvirt 2.4.0 introduce new iotune parameters that enhance\nthe flexibility of block storage quality of service (QoS). Specifically,\nthey enable the configuration of burst lengths for read, write, and\noverall IOPS and bandwidth. This blueprint proposes to incorporate these\nparameters into Cinder QoS specifications for connected volumes.\n\nChange-Id: Id7e3c0a8bc6509edc94d2e58be4611e5c6a416c2\nSigned-off-by: Mengyang Zhang \u003cmzhang741@bloomberg.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/64e09e04f3a0ffd3b6df558fc8f106cc151516c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/64e09e04f3a0ffd3b6df558fc8f106cc151516c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
