)]}'
{"id":"openstack%2Fnova-specs~934391","triplet_id":"openstack%2Fnova-specs~master~I27c1d527f465f9547ee4159ab01a471c2198b9b0","project":"openstack/nova-specs","branch":"master","topic":"bp/unified-limits-nova-unset-limits","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2024-11-20 10:24:07.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-11-20 10:14:40.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-11-19 14:50:37.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}}},"hashtags":[],"change_id":"I27c1d527f465f9547ee4159ab01a471c2198b9b0","subject":"Re-propose config option for behavior of unset unified limits","status":"MERGED","created":"2024-11-08 01:46:34.000000000","updated":"2024-11-20 10:25:47.000000000","submitted":"2024-11-20 10:24:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"934391-bp/unified-limits-nova-unset-limits","meta_rev_id":"c49847ba15970e7de7bf92c276588101ed5e9f1c","_number":934391,"virtual_id_number":934391,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-20 10:24:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2024-11-20 10:14:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2024-11-19 14:50:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2024-11-20 10:14:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-08 01:54:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-08 21:15:19.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2024-11-08 21:26:06.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-11-08 21:26:06.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"4f396d249da494f3ac35935d0a035357a9fd487f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-11-08 01:46:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e76273701eecca60bcf44150b0853fc61339f939","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 01:54:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/400fe71909a5480fa39aba233107c654\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/997108203f7f4521978ff49bf8104a08 : FAILURE in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/621aff6abe57447ab399e933cdba5570 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"27e993a0b63c2bb435cbc265f10643ad49c0706d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-08 21:15:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6512c2169152a431653dd00217e88ee7bfc98e1d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-08 21:26:06.000000000","message":"Patch Set 1: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0bf8403cb4d1c33bceea8d2a84e006eb4c75ad3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"real_author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-08 21:31:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_4690\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}],"_revision_number":2},{"id":"e3f459d9e1781eff88af4228627fefef850bf8d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 21:37:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb8b7c1a1b614221914087c548785f66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7ed4348e7d6492dabec927d94963416 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0962aa02beae4d8db33dc940b03baf17 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9a5bfdb50d62c8c69331e2ea1314a8dcb6f4a896","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-11-10 01:40:03.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6ef2183e2a7c13e594ef07d70c7a428464c418f9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-11-14 20:39:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d8ba35b470ad3e36cc27f4b25512d858f01ef72e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-14 20:51:11.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"79b54d592dd1199dc0d4ecae67c3adee96e9e4c5","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-11-14 21:55:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4ab2ff0a6c216cc9bdb299bf4a4654ed1ee9769","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-11-14 21:57:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8ba6b497836a999ea95518cdf91fc50ad8989b27","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-11-14 22:18:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"013d8df4dc72d15e9217f2bde091dcec1f0105f6","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-11-15 04:06:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d0fae3dd102f010f392faa49e15887095ee0eb5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-15 10:25:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8689288d7e82a88b19e9177b7318fb01b2ce5861","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-11-15 14:38:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c9efa9f87aca8f5e8a4eed8e632cf2cf4d42aa1","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-11-15 17:34:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c61076867ee643da944842231c456526dd6b3dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-11-19 04:29:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1167c508cdd0738fd76b5899a4bee907a1fe8cca","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-11-19 04:30:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e001f707c3c7ec61f72a3f4ff005ae88d31a6cf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 04:39:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2301ffa4b70b4c359f11c28b3d4cdf5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc11ebcfbc2048c2b17d1f671a958ec5 : FAILURE in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77404b2fcd13432d8e8b6d6b8b6d9b1d : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":3},{"id":"75a86c3e60e1f84db4cd5aca4bd8a852ab70de32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-11-19 05:14:59.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ad042ec095b105437cb5b380cdda56c0445c3181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 05:21:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b30164751f77403999f38fa606c17066\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9b13797c1c04c1d9d1598598457e88e : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90690814874b4418926a1596f8c198b8 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":4},{"id":"e022fb2194580d4c3aa7c5fd8c18fada265dc844","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-11-19 14:50:37.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5b54e40bf51fe4b8cc941e2862a5f823846a0a1b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-20 10:14:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e2e03122a8686ea5038324217e10c6db1420b06b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 10:14:59.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"56f6d09c0e24fac920dca3c03c2af1b1518cc02c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 10:24:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00117ceb1d5440dface53271efe78ed6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f56656a705bf4c44a9422f73147d05ca : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2879925b68af40d9bf74dc37052ef58d : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":4},{"id":"1d60cb47d90d41b61af28efafd73e3637165b805","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 10:24:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c49847ba15970e7de7bf92c276588101ed5e9f1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 10:25:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf4ee0b3a0a24b9f9384a913598a0745\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/dd6e83f2432e486bb94eaba7ef1bbe65 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","revisions":{"e6c975c428c0cdd8bc6ef1fefd24721005ff399e":{"kind":"REWORK","_number":1,"created":"2024-11-08 01:46:34.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/91/934391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/934391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/1"}}},"commit":{"parents":[{"commit":"0cb32c15886f2beb61ed68e01b26b4deea7de4b7","subject":"Fix typos in release name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0cb32c15886f2beb61ed68e01b26b4deea7de4b7"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-11-08 01:12:20.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-11-08 01:15:35.000000000","tz":0},"subject":"Re-propose config option for behavior of unset unified limits","message":"Re-propose config option for behavior of unset unified limits\n\nThis re-proposal contains significant changes from the original 2024.2\n(Dalmatian) proposal.\n\nRelated to blueprint unified-limits-nova-unset-limits\n\nChange-Id: I27c1d527f465f9547ee4159ab01a471c2198b9b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e6c975c428c0cdd8bc6ef1fefd24721005ff399e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e6c975c428c0cdd8bc6ef1fefd24721005ff399e"}]},"branch":"refs/heads/master"},"b4b4aa4fe1b40c3695bf39b6c5a54ffa41f7b5c8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-08 21:31:19.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"real_uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/91/934391/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/934391/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/2"}}},"commit":{"parents":[{"commit":"1bf26c951d2b9f3ba1a315260179a14ea70d3179","subject":"fix footnote refernces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1bf26c951d2b9f3ba1a315260179a14ea70d3179"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-11-08 01:12:20.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-11-08 21:31:19.000000000","tz":0},"subject":"Re-propose config option for behavior of unset unified limits","message":"Re-propose config option for behavior of unset unified limits\n\nThis re-proposal contains significant changes from the original 2024.2\n(Dalmatian) proposal.\n\nRelated to blueprint unified-limits-nova-unset-limits\n\nChange-Id: I27c1d527f465f9547ee4159ab01a471c2198b9b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b4b4aa4fe1b40c3695bf39b6c5a54ffa41f7b5c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b4b4aa4fe1b40c3695bf39b6c5a54ffa41f7b5c8"}]},"branch":"refs/heads/master","description":"Rebase"},"5a558658d71db67d9a754ce59ecd9e79e65e9d4a":{"kind":"REWORK","_number":3,"created":"2024-11-19 04:29:58.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/91/934391/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/934391/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/3"}}},"commit":{"parents":[{"commit":"0cb32c15886f2beb61ed68e01b26b4deea7de4b7","subject":"Fix typos in release name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0cb32c15886f2beb61ed68e01b26b4deea7de4b7"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-11-08 01:12:20.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-11-19 03:03:05.000000000","tz":0},"subject":"Re-propose config option for behavior of unset unified limits","message":"Re-propose config option for behavior of unset unified limits\n\nThis re-proposal contains significant changes from the original 2024.2\n(Dalmatian) proposal.\n\nRelated to blueprint unified-limits-nova-unset-limits\n\nChange-Id: I27c1d527f465f9547ee4159ab01a471c2198b9b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5a558658d71db67d9a754ce59ecd9e79e65e9d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5a558658d71db67d9a754ce59ecd9e79e65e9d4a"}]},"branch":"refs/heads/master"},"6cae04b088a72509b6ea01d2db8d400e2bccd6cc":{"kind":"REWORK","_number":4,"created":"2024-11-19 05:14:59.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/91/934391/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/91/934391/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/91/934391/4"}}},"commit":{"parents":[{"commit":"a422b51b992716de61ad3d8db91feff5dc23d82a","subject":"Merge \"Repropose spice-direct console support.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a422b51b992716de61ad3d8db91feff5dc23d82a"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-11-08 01:12:20.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-11-19 05:05:45.000000000","tz":0},"subject":"Re-propose config option for behavior of unset unified limits","message":"Re-propose config option for behavior of unset unified limits\n\nThis re-proposal contains significant changes from the original 2024.2\n(Dalmatian) proposal.\n\nRelated to blueprint unified-limits-nova-unset-limits\n\nChange-Id: I27c1d527f465f9547ee4159ab01a471c2198b9b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
