)]}'
{"id":"openstack%2Fnova-specs~936140","triplet_id":"openstack%2Fnova-specs~master~I0c771c647e11ce7bdb409a7c0c47c04e7bb94a66","project":"openstack/nova-specs","branch":"master","topic":"bp/show-scheduler-hints-in-server-details","attention_set":{"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2025-02-26 19:57:36.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-12-11 15:02:08.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2024-12-11 15:02:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0c771c647e11ce7bdb409a7c0c47c04e7bb94a66","subject":"Add spec for show scheduler hints in server details","status":"MERGED","created":"2024-11-25 11:27:39.000000000","updated":"2025-02-26 19:57:36.000000000","submitted":"2024-12-11 15:02:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":62,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"936140-bp/show-scheduler-hints-in-server-details","meta_rev_id":"dc1190ad30b0024578a8aaa50c7fe0b5d3c37e5e","_number":936140,"virtual_id_number":936140,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-11 15:02:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2024-12-11 14:40:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2024-12-11 14:48:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2024-12-11 14:48:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-25 11:57:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-26 04:44:24.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-11-26 08:30:39.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"CC"},{"updated":"2024-12-10 14:04:53.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2024-12-10 14:04:59.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2024-12-11 14:48:35.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2024-12-16 11:52:39.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2025-02-26 19:57:36.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"}],"messages":[{"id":"441b7f7d4585c1c4283bf197797b83a7135307ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2024-11-25 11:27:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06196b6f5d675b168a3f2e7eb7520431fecf989d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 11:57:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a0629eb66424215aa7ad7f743085743\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/148e5730f29e4a6f84c3b5524f8854f1 : SUCCESS in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30facce102e34527a9fdb049f1eceebb : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b9e8ef29efb447a6232523c68ec00c4e14b89201","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-26 04:44:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"10455a27b699ae729d24fac44b04fedc4781eba7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2024-11-26 19:01:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3b5cb1de3e6b449dbb35d72afb0046d3a1d8b5a4","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2024-11-26 19:07:00.000000000","message":"Patch Set 2:\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f0295c3b86bc5ed22a22b28ee236595335c3e0b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 19:08:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa9f3a8b2e8c497abeae8a9d43caee39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fa7e594037d436b9d3ae267a62e5c27 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd8ee9ffc1ce4c18b9afb1397b4a46b5 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":2},{"id":"035e471711b2abb7b81b54dd5846e942976d6edd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2024-12-09 18:03:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9b5a8226a9a5899239413f898d415b539bff0a26","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2024-12-09 18:07:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"71e26dfa5451077bf44fd8751639be43bb1393ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 18:14:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a83291cd1dbf4037885f01c2ed3abb14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/113240958fdd40cba6113072c3a139a0 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4cb85b7ef1a41e4a4438fbcaf5b6294 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":3},{"id":"d216612f58fff650b98bbb4f677eb913f8599175","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-12-09 18:27:30.000000000","message":"Patch Set 3: Code-Review+2\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c48da86bd0fce32f74c13dde6bbdf63a353d6bec","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-10 14:04:53.000000000","message":"Patch Set 3:\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ef8e83e8226e312359cedb7e28d33d52e0f7d82e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-10 14:04:59.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"961c42a2de372b0d5153990b860d173e94c9b327","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2024-12-10 19:01:20.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"46bd636fda60c03467c3983d0700fb758c9f890e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 19:09:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a0cae61c40e4b5da8bfc8d47dd61a29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba715d8af76c408188efa80b8b56e6cf : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30d238fb5d2c4e7284ba14f7e214408c : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":4},{"id":"60283e45471b833099e84c1237242f42e44f4af4","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2024-12-10 19:52:29.000000000","message":"Patch Set 4:\n\n(10 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e789751f773f4e9805670a46dc0ef45271deea1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2024-12-11 13:27:22.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"bf26d71d37a88f1f809ed22dbc7f48dac4ad389e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2024-12-11 13:30:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1614111d80dfa4ccaad10696f85d777639d7f506","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 13:35:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e04b829bc4c1476280340279a9ce1464\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b552edcb22df46e4a895e1e77c4077e9 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26dd5979b4ff418c9e6e16a4c6734a05 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":5},{"id":"f3c1079740a0885fca1151456d27e8936e764b39","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-12-11 14:40:26.000000000","message":"Patch Set 5: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4d9e75d6bc823bf0d9c53ed7bb254d8510fb5c86","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-12-11 14:48:35.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"83c2a7fea25525a2c9e81ed0acd8408700261a81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 14:48:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"948b147d13a9f011f76db044f540f6e5d64d6658","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 15:02:08.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/018a5ba1de9f4c88ab880de001596e6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f4c862c3be4f8a851bf9c433b7f187 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58db4e3083e542c09b72dfb5d596bbdb : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":5},{"id":"9db61a429a3f454da6b27eb2590b22aa9ec8a737","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 15:02:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"9a25b1978d9a81799a4a8b964d591e7cf5f62500","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 15:09:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b52fab5c9bc84225978ff4ab8e8ea87d\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/92e5bf55610d4cb4a2a33d6dfd7c5a9b : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":5},{"id":"ca2738479859956903f463f8ad76d8f8aa568ca8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-13 10:15:11.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"43456e6627186208cdf963c9d48185f248e1712e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-12-14 02:20:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e0e91f505063ae871c47b655a7aaff3fa1cf1c60","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-12-16 11:52:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"aa469b3799e1c22eba2ff7544918a72e9e6c3332","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-06 10:38:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8242b482418b84d6d96cbab51da48b3427d3eae8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-07 11:49:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f918426b90fd2e1967cb61404b91a5df7d310bbc","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-01-14 19:37:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dc1190ad30b0024578a8aaa50c7fe0b5d3c37e5e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-26 19:57:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f55f6012eb34ae7750666a5c4b551c5818f51eff","revisions":{"1f0918ac861f7e49ae2c57137dd34ed60e8f7994":{"kind":"REWORK","_number":1,"created":"2024-11-25 11:27:39.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/40/936140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/40/936140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/1"}}},"commit":{"parents":[{"commit":"a422b51b992716de61ad3d8db91feff5dc23d82a","subject":"Merge \"Repropose spice-direct console support.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a422b51b992716de61ad3d8db91feff5dc23d82a"}]}],"author":{"name":"Douglas Viroel","email":"dviroel@redhat.com","date":"2024-11-25 11:18:52.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"dviroel@redhat.com","date":"2024-11-25 11:26:48.000000000","tz":-180},"subject":"Add spec for show scheduler hints in server details","message":"Add spec for show scheduler hints in server details\n\nNova currently lacks a straightforward way to expose scheduler hints\nassociated with a server. This proposal suggests extending existing Nova\u0027s\nAPI to allow users to retrieve this information when it is available.\nNote that there was a spec proposal in the past, which was abandoned\n[1].\n\nAPIImpact\n\nblueprint show-scheduler-hints-in-server-details\n\n[1] https://review.opendev.org/c/openstack/nova-specs/+/440580\n\nChange-Id: I0c771c647e11ce7bdb409a7c0c47c04e7bb94a66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1f0918ac861f7e49ae2c57137dd34ed60e8f7994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1f0918ac861f7e49ae2c57137dd34ed60e8f7994"}]},"branch":"refs/heads/master"},"9f05d70bdb82918cd9af84b5303f1950745e5c23":{"kind":"REWORK","_number":2,"created":"2024-11-26 19:01:04.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/40/936140/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/40/936140/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/2"}}},"commit":{"parents":[{"commit":"a422b51b992716de61ad3d8db91feff5dc23d82a","subject":"Merge \"Repropose spice-direct console support.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a422b51b992716de61ad3d8db91feff5dc23d82a"}]}],"author":{"name":"Douglas Viroel","email":"dviroel@redhat.com","date":"2024-11-25 11:18:52.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"dviroel@redhat.com","date":"2024-11-26 19:00:03.000000000","tz":-180},"subject":"Add spec for show scheduler hints in server details","message":"Add spec for show scheduler hints in server details\n\nNova currently lacks a straightforward way to expose scheduler hints\nassociated with a server. This proposal suggests extending existing Nova\u0027s\nAPI to allow users to retrieve this information when it is available.\nNote that there was a spec proposal in the past, which was abandoned\n[1].\n\nAPIImpact\n\nblueprint show-scheduler-hints-in-server-details\n\n[1] https://review.opendev.org/c/openstack/nova-specs/+/440580\n\nChange-Id: I0c771c647e11ce7bdb409a7c0c47c04e7bb94a66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9f05d70bdb82918cd9af84b5303f1950745e5c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9f05d70bdb82918cd9af84b5303f1950745e5c23"}]},"branch":"refs/heads/master"},"403819d4c1b0d0ae9ecb3ca8e69579240feffa99":{"kind":"REWORK","_number":3,"created":"2024-12-09 18:03:20.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/40/936140/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/40/936140/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/3"}}},"commit":{"parents":[{"commit":"a422b51b992716de61ad3d8db91feff5dc23d82a","subject":"Merge \"Repropose spice-direct console support.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a422b51b992716de61ad3d8db91feff5dc23d82a"}]}],"author":{"name":"Douglas Viroel","email":"dviroel@redhat.com","date":"2024-11-25 11:18:52.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"dviroel@redhat.com","date":"2024-12-09 18:02:49.000000000","tz":-180},"subject":"Add spec for show scheduler hints in server details","message":"Add spec for show scheduler hints in server details\n\nNova currently lacks a straightforward way to expose scheduler hints\nassociated with a server. This proposal suggests extending existing Nova\u0027s\nAPI to allow users to retrieve this information when it is available.\nNote that there was a spec proposal in the past, which was abandoned\n[1].\n\nAPIImpact\n\nblueprint show-scheduler-hints-in-server-details\n\n[1] https://review.opendev.org/c/openstack/nova-specs/+/440580\n\nChange-Id: I0c771c647e11ce7bdb409a7c0c47c04e7bb94a66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/403819d4c1b0d0ae9ecb3ca8e69579240feffa99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/403819d4c1b0d0ae9ecb3ca8e69579240feffa99"}]},"branch":"refs/heads/master"},"c3b764bea6e4060ae22f4d979ca2d3f7b73b08e3":{"kind":"REWORK","_number":4,"created":"2024-12-10 19:01:20.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/40/936140/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/40/936140/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/4"}}},"commit":{"parents":[{"commit":"a422b51b992716de61ad3d8db91feff5dc23d82a","subject":"Merge \"Repropose spice-direct console support.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a422b51b992716de61ad3d8db91feff5dc23d82a"}]}],"author":{"name":"Douglas Viroel","email":"dviroel@redhat.com","date":"2024-11-25 11:18:52.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2024-12-10 19:00:33.000000000","tz":-180},"subject":"Add spec for show scheduler hints in server details","message":"Add spec for show scheduler hints in server details\n\nNova currently lacks a straightforward way to expose scheduler hints\nassociated with a server. This proposal suggests extending existing Nova\u0027s\nAPI to allow users to retrieve this information when it is available.\nNote that there was a spec proposal in the past, which was abandoned\n[1].\n\nAPIImpact\n\nblueprint show-scheduler-hints-in-server-details\n\n[1] https://review.opendev.org/c/openstack/nova-specs/+/440580\n\nChange-Id: I0c771c647e11ce7bdb409a7c0c47c04e7bb94a66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c3b764bea6e4060ae22f4d979ca2d3f7b73b08e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c3b764bea6e4060ae22f4d979ca2d3f7b73b08e3"}]},"branch":"refs/heads/master"},"f55f6012eb34ae7750666a5c4b551c5818f51eff":{"kind":"REWORK","_number":5,"created":"2024-12-11 13:27:22.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/40/936140/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/40/936140/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/40/936140/5"}}},"commit":{"parents":[{"commit":"a422b51b992716de61ad3d8db91feff5dc23d82a","subject":"Merge \"Repropose spice-direct console support.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a422b51b992716de61ad3d8db91feff5dc23d82a"}]}],"author":{"name":"Douglas Viroel","email":"dviroel@redhat.com","date":"2024-11-25 11:18:52.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2024-12-11 13:27:11.000000000","tz":-180},"subject":"Add spec for show scheduler hints in server details","message":"Add spec for show scheduler hints in server details\n\nNova currently lacks a straightforward way to expose scheduler hints\nassociated with a server. This proposal suggests extending existing Nova\u0027s\nAPI to allow users to retrieve this information when it is available.\nNote that there was a spec proposal in the past, which was abandoned\n[1].\n\nAPIImpact\n\nblueprint show-scheduler-hints-in-server-details\n\n[1] https://review.opendev.org/c/openstack/nova-specs/+/440580\n\nChange-Id: I0c771c647e11ce7bdb409a7c0c47c04e7bb94a66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f55f6012eb34ae7750666a5c4b551c5818f51eff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f55f6012eb34ae7750666a5c4b551c5818f51eff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
