)]}'
{"id":"openstack%2Fnova-specs~936389","triplet_id":"openstack%2Fnova-specs~master~I36f42e08d3213976ea68823d9fa8349037f36269","project":"openstack/nova-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"37442":{"account":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"last_update":"2025-01-10 00:30:22.000000000","reason":"Change was abandoned"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-01-06 21:29:06.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"hashtags":[],"change_id":"I36f42e08d3213976ea68823d9fa8349037f36269","subject":"Proposes `Add Distributed Locking for Host Discovery`.","status":"ABANDONED","created":"2024-11-27 16:30:29.000000000","updated":"2025-01-10 00:30:22.000000000","total_comment_count":18,"unresolved_comment_count":9,"has_review_started":true,"meta_rev_id":"eb02f742bbc73b5f98dc1c8c769cc2e0e7dee1dd","_number":936389,"virtual_id_number":936389,"owner":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-05 15:12:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-01-08 15:23:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-27 16:44:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-27 19:45:38.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-01-08 15:23:29.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"c686724e3fe9579ff120f50dd762d468ac851ebf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2024-11-27 16:30:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bff8a05fc44c3a5c9174dce8fa434b1934cacd15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 16:44:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3362bde42ede463aaa8bed08f091ff79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b10238193cd641b183ff0e44db9b83cb : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70a7febe0aa747b7999f8f7e4bcd5d74 : FAILURE in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"5c00f700a6603b5745c3f5c9bed840b4fe14bb35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2024-11-27 16:53:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"785a238a233f8e1ffa5ba2cfed1cafcdff6724a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 17:08:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c426a4cb281f45f99c20da7ecad18d14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e05988130944b1821cdba1e7d51e2b : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7df3092873e4492ab37b8b16408054ac : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"277c4cc530de5af1d92d85822b41a2e9b1b44c75","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-27 19:45:38.000000000","message":"Patch Set 2: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":2},{"id":"98ac04a1b05ac7a590658bc7918e7fe275bc0e86","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2024-11-27 19:56:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a9d195d7633557872faa99861742249ffb97bbd3","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2024-12-05 13:55:04.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3b7a557767e032d2c88c2994f5d2f585373c2d83","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2024-12-05 14:00:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c644c0bcdb4b62b74eaa5ea908bfcf1bc5c289b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2024-12-05 14:59:31.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5d211bd233386b7a29e2fb09328644b687a89428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 15:12:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3e58a837f9743ca93de496e521691e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9394576219f64d5085a125fb61de9d4b : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/787aa33a8a5c4d2ba73e3987d8d58e6e : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"54ee79f5b49f02544f0db72e64a2d9d23759c109","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-06 21:29:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"436ad07d2cd0858c3c4d05ddbad5540b3c3376e2","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-01-08 15:23:29.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eb02f742bbc73b5f98dc1c8c769cc2e0e7dee1dd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-10 00:30:22.000000000","message":"Abandoned\n\nill abandon this for now and we can restore it if https://blueprints.launchpad.net/nova/+spec/distributed-host-discovery turns out to not be enough.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f9d0d54f2dea5ca19b2e5a1f4e433ff123a04636","revisions":{"4fc173eab8d5952c96f43f03436e47e837862398":{"kind":"REWORK","_number":1,"created":"2024-11-27 16:30:29.000000000","uploader":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"ref":"refs/changes/89/936389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/89/936389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/936389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/936389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/936389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/89/936389/1"}}},"commit":{"parents":[{"commit":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","subject":"Re-propose config option for behavior of unset unified limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]}],"author":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2024-11-27 16:28:35.000000000","tz":60},"committer":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2024-11-27 16:28:35.000000000","tz":60},"subject":"[SPEC] `Add Distributed Locking for Host Discovery`.","message":"[SPEC] `Add Distributed Locking for Host Discovery`.\n\nChange-Id: I36f42e08d3213976ea68823d9fa8349037f36269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4fc173eab8d5952c96f43f03436e47e837862398"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4fc173eab8d5952c96f43f03436e47e837862398"}]},"branch":"refs/heads/master"},"e9ed10aaa3c493a0a5b39d51eeeba44a323e6ab2":{"kind":"REWORK","_number":2,"created":"2024-11-27 16:53:14.000000000","uploader":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"ref":"refs/changes/89/936389/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/89/936389/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/936389/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/936389/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/936389/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/89/936389/2"}}},"commit":{"parents":[{"commit":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","subject":"Re-propose config option for behavior of unset unified limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]}],"author":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2024-11-27 16:28:35.000000000","tz":60},"committer":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2024-11-27 16:53:07.000000000","tz":60},"subject":"[SPEC] `Add Distributed Locking for Host Discovery`.","message":"[SPEC] `Add Distributed Locking for Host Discovery`.\n\nChange-Id: I36f42e08d3213976ea68823d9fa8349037f36269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e9ed10aaa3c493a0a5b39d51eeeba44a323e6ab2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e9ed10aaa3c493a0a5b39d51eeeba44a323e6ab2"}]},"branch":"refs/heads/master"},"f9d0d54f2dea5ca19b2e5a1f4e433ff123a04636":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-12-05 14:59:31.000000000","uploader":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"ref":"refs/changes/89/936389/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/89/936389/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/936389/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/936389/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/936389/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/89/936389/3"}}},"commit":{"parents":[{"commit":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","subject":"Re-propose config option for behavior of unset unified limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]}],"author":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2024-11-27 16:28:35.000000000","tz":60},"committer":{"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","date":"2024-12-05 14:59:31.000000000","tz":0},"subject":"Proposes `Add Distributed Locking for Host Discovery`.","message":"Proposes `Add Distributed Locking for Host Discovery`.\n\nChange-Id: I36f42e08d3213976ea68823d9fa8349037f36269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f9d0d54f2dea5ca19b2e5a1f4e433ff123a04636"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f9d0d54f2dea5ca19b2e5a1f4e433ff123a04636"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
