)]}'
{"id":"openstack%2Fnova-specs~936407","triplet_id":"openstack%2Fnova-specs~master~Ibb16cc3e438e4766efbbe7c443cc439df8f3b9ff","project":"openstack/nova-specs","branch":"master","topic":"bp/enable-vfio-devices-with-kernel-variant-drivers","attention_set":{"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2025-01-14 13:17:47.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-01-09 14:59:32.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-01-09 15:09:50.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2025-01-09 15:09:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibb16cc3e438e4766efbbe7c443cc439df8f3b9ff","subject":"Enable VFIO devices with kernel variant drivers","status":"MERGED","created":"2024-11-27 19:22:19.000000000","updated":"2025-01-16 14:09:05.000000000","submitted":"2025-01-09 15:09:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":232,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"936407-bp/enable-vfio-devices-with-kernel-variant-drivers","meta_rev_id":"3ee1ec45c0c8840cb7071c495b464a9f01573653","_number":936407,"virtual_id_number":936407,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-09 15:09:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2025-01-09 14:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-09 12:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2025-01-08 17:56:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2025-01-09 14:59:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-27 19:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-28 10:00:08.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2024-12-02 10:42:32.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2024-12-02 14:11:35.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"97d5f3ccb0a3144bc7ff6a446ba7726d3f56dd3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-11-27 19:22:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0dce4effe62f05d29ca28bb52e66a49abcc3b3e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 19:39:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32ced0ba1abc4e3c9a93cd3b88311f47\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab432850fbf94daeb7437733ad129339 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68a4b54afdc64983a380ff75cb9c7b72 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"9c50f5c44a11d99ed09de617c841d58039d15eec","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-11-28 10:00:08.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2248767d89c97da0d15340e3f783bb7075f1a24c","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-11-28 14:32:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d71c68fa0f49bf98eabc5aee7c3814a65435ad81","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-02 10:42:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(27 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f8102588bacd2dc25ec12b8d7a3bf4af1da31fc3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-12-02 14:11:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(21 comments)","accounts_in_message":[],"_revision_number":1},{"id":"de0a8d92ff711a5e3fb34997fdf3c1c27df1131e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-12-03 01:05:38.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0cad971593e2f53ce79d5844693a9fa3e9e43d92","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-03 12:14:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"685288e8e27a4da679bce9e469a02ba24e4050d1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-12-04 10:49:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e848b53816d6887f5cb8053c391c96bf4afb1026","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-12-11 17:33:19.000000000","message":"Patch Set 1:\n\n(29 comments)","accounts_in_message":[],"_revision_number":1},{"id":"85f77133bb32017e5a520564a1f70833e45a68a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-12-11 17:37:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"03453d0cfbe8ab2e58a605e1c7118be41ab6a091","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-12-11 17:38:32.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"45eaa19f0b95ab5b347512a19ffef58dfbacb4aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 17:49:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/614fed00a1734140aca0373ab717475a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/759fa8300d4c42458ab79ed3a81f5fb2 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e21f7c62d8af4ea983871a907b9facfa : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":3},{"id":"515e80675e17769ecde8109deae7e1ee19ad2405","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-12-12 20:35:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d9e8478343e6ae791e8c2c2f5c393deb7258f766","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 20:45:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa986faa02f1470f967b6f57c851e594\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5c1634b4144473abfb025c87470e306 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e39724719781489d863f685946956b2f : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":4},{"id":"7792ff7f566309a831a11fe5c9c23fe184cdc843","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-13 11:05:38.000000000","message":"Patch Set 4: Code-Review-1\n\n(18 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f5812dec7dd6007a2d5361f8a4db403f8b386b9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-12-18 15:43:55.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0bde15c97ff92f90fa3c07e02831be7f351e14ea","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-12-18 15:44:17.000000000","message":"Patch Set 5:\n\n(17 comments)","accounts_in_message":[],"_revision_number":5},{"id":"603ccc7e6d7f5a6e7da46c3137f0f447e9bfa617","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 16:03:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/319cf07320704b87b10e2e1f4fac242d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f16275aef72f4b29866328fa1d6ac53d : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7515fbe050c94d0faec455782369b19e : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":5},{"id":"23edd9a0292b21c03b1bfcbe4138e84df262caa0","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-19 10:37:30.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fa46d320c3c4f72c6b91477f0e7acd62c8373104","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-19 10:37:35.000000000","message":"Patch Set 5: Code-Review-1","accounts_in_message":[],"_revision_number":5},{"id":"83b0729518adeb0a3f29a6cb7a001fdc42f45ec1","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-01-03 16:09:34.000000000","message":"Patch Set 5:\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"64f8e3d1a168667b96266adaf5f880731e30ace1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-07 12:06:02.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"56e6044a2fe7d514ba89b4b62852027796b81415","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-07 14:56:46.000000000","message":"Patch Set 5: Code-Review-1\n\n(18 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c3fd564bfc001cff574b345601f1f8b842ebb94f","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-01-07 16:55:46.000000000","message":"Patch Set 5: Code-Review-1\n\n(23 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f6916d7fc9a3e7948fef294eaa3e2f033cdc3180","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-08 11:09:36.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"dfde2085cd5f50475a3ecefd1e25cb9b5672eacf","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-01-08 15:28:28.000000000","message":"Patch Set 5:\n\n(34 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8e4fb77a02d99f0d8d6afe0f2f35ed0fb514bc3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-01-08 15:29:01.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"fa90ce028e449b55679bceb220f501cac54dd6fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 15:39:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b47dc2d4d91341d28d52c44cbf098f1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d45e5cc314f4061b80c2787bc9d32fd : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcd512d282e14dddb8b982e21b16a545 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":6},{"id":"eddee4c7ee6e0e7ddbb2463f86c20bd54898a0d9","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-01-08 17:56:33.000000000","message":"Patch Set 6: Code-Review+2\n\n(8 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1961057d165c806a79e9fa4e134a6333d9d0b6ee","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-09 12:09:51.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"70295324394cf2f7b52762436f9ce0ded63113e9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-09 14:59:32.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"362fbfbe0a8e028cd89a99d175638d81e9764dcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 14:59:51.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7f6b65a37a55153fb2fb8bc8b9e01fc73163dc92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 15:09:50.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8031bf18e54f4d40af430342186c8b73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10b5661dffc440bc8799dd1970a2d234 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b61723365c418e8992860eb1c48a14 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":6},{"id":"4282aa6bb102c47b27e06b17bb5bd825cca58420","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 15:09:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"cb00a403ef55e00643a35d499904e12a26e726c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 15:11:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f8da872a7df4b4fb4a643811e42f2b8\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/cc95ea1af5b6454eaba354fb3489a0c3 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":6},{"id":"dd7a06402a56ec01ee6a7a2d6fa0f5fc471e9f5a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-14 13:17:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7e41d120b49c909934f7fb3bd2b765bcbbecacd6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-14 13:57:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3ee1ec45c0c8840cb7071c495b464a9f01573653","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-16 14:09:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f972c6d8df5f4fbed368109b9035cb0d6a9fd280","revisions":{"df1dd6b102f5f3c45b29a6b96f7aa731e4613da1":{"kind":"REWORK","_number":1,"created":"2024-11-27 19:22:19.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/07/936407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/07/936407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/1"}}},"commit":{"parents":[{"commit":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","subject":"Re-propose config option for behavior of unset unified limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-11-27 19:20:41.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-11-27 19:20:41.000000000","tz":60},"subject":"WIP: Allow vfio gpu support with Vendor-Specific VFIO framework","message":"WIP: Allow vfio gpu support with Vendor-Specific VFIO framework\n\nChange-Id: Ibb16cc3e438e4766efbbe7c443cc439df8f3b9ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/df1dd6b102f5f3c45b29a6b96f7aa731e4613da1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/df1dd6b102f5f3c45b29a6b96f7aa731e4613da1"}]},"branch":"refs/heads/master"},"422384c36db0f71a4695c2dc2edc48123e6b05ca":{"kind":"REWORK","_number":2,"created":"2024-12-11 17:37:14.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/07/936407/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/07/936407/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/2"}}},"commit":{"parents":[{"commit":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","subject":"Re-propose config option for behavior of unset unified limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-11-27 19:20:41.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-12-11 17:36:23.000000000","tz":60},"subject":"Allow vfio gpu support with Vendor-Specific VFIO framework","message":"Allow vfio gpu support with Vendor-Specific VFIO framework\n\nThis spec outlines the necessary steps to enable support for SR-IOV\ndevices using the new kernel variant driver interface.\n\nImplements: blueprint enable-vfio-devices-with-kernel-variant-drivers\nChange-Id: Ibb16cc3e438e4766efbbe7c443cc439df8f3b9ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/422384c36db0f71a4695c2dc2edc48123e6b05ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/422384c36db0f71a4695c2dc2edc48123e6b05ca"}]},"branch":"refs/heads/master"},"31e965eff54c90b8b47b38248760c055b51871e3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-12-11 17:38:32.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/07/936407/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/07/936407/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/3"}}},"commit":{"parents":[{"commit":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","subject":"Re-propose config option for behavior of unset unified limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-11-27 19:20:41.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-12-11 17:38:14.000000000","tz":60},"subject":"Enable VFIO devices with kernel variant drivers","message":"Enable VFIO devices with kernel variant drivers\n\nThis spec outlines the necessary steps to enable support for SR-IOV\ndevices using the new kernel variant driver interface.\n\nImplements: blueprint enable-vfio-devices-with-kernel-variant-drivers\nChange-Id: Ibb16cc3e438e4766efbbe7c443cc439df8f3b9ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/31e965eff54c90b8b47b38248760c055b51871e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/31e965eff54c90b8b47b38248760c055b51871e3"}]},"branch":"refs/heads/master"},"8d66fcf0b05b6cdf2e116c93f7c4e3aba9ff0a88":{"kind":"REWORK","_number":4,"created":"2024-12-12 20:35:52.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/07/936407/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/07/936407/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/4"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-11-27 19:20:41.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-12-12 20:35:51.000000000","tz":60},"subject":"Enable VFIO devices with kernel variant drivers","message":"Enable VFIO devices with kernel variant drivers\n\nThis spec outlines the necessary steps to enable support for SR-IOV\ndevices using the new kernel variant driver interface.\n\nImplements: blueprint enable-vfio-devices-with-kernel-variant-drivers\nChange-Id: Ibb16cc3e438e4766efbbe7c443cc439df8f3b9ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d66fcf0b05b6cdf2e116c93f7c4e3aba9ff0a88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d66fcf0b05b6cdf2e116c93f7c4e3aba9ff0a88"}]},"branch":"refs/heads/master"},"f440e848c2123d6c4ea43db80c2afa85f183b6b5":{"kind":"REWORK","_number":5,"created":"2024-12-18 15:43:55.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/07/936407/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/07/936407/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/5"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-11-27 19:20:41.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-12-18 15:43:04.000000000","tz":60},"subject":"Enable VFIO devices with kernel variant drivers","message":"Enable VFIO devices with kernel variant drivers\n\nThis spec outlines the necessary steps to enable support for SR-IOV\ndevices using the new kernel variant driver interface.\n\nImplements: blueprint enable-vfio-devices-with-kernel-variant-drivers\nChange-Id: Ibb16cc3e438e4766efbbe7c443cc439df8f3b9ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f440e848c2123d6c4ea43db80c2afa85f183b6b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f440e848c2123d6c4ea43db80c2afa85f183b6b5"}]},"branch":"refs/heads/master"},"f972c6d8df5f4fbed368109b9035cb0d6a9fd280":{"kind":"REWORK","_number":6,"created":"2025-01-08 15:29:01.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/07/936407/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/07/936407/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/07/936407/6"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-11-27 19:20:41.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-01-08 15:28:48.000000000","tz":60},"subject":"Enable VFIO devices with kernel variant drivers","message":"Enable VFIO devices with kernel variant drivers\n\nThis spec outlines the necessary steps to enable support for SR-IOV\ndevices using the new kernel variant driver interface.\n\nImplements: blueprint enable-vfio-devices-with-kernel-variant-drivers\nChange-Id: Ibb16cc3e438e4766efbbe7c443cc439df8f3b9ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f972c6d8df5f4fbed368109b9035cb0d6a9fd280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f972c6d8df5f4fbed368109b9035cb0d6a9fd280"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
