)]}'
{"id":"openstack%2Fnova-specs~937185","triplet_id":"openstack%2Fnova-specs~master~I069aa57cad3bd836bcd974ad57decb28771e2278","project":"openstack/nova-specs","branch":"master","topic":"graceful-compute-shutdown","attention_set":{"37442":{"account":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"last_update":"2025-05-14 09:53:24.000000000","reason":"\u003cGERRIT_ACCOUNT_16207\u003e replied on the change","reason_account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-05-14 10:41:11.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"21813":{"account":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"last_update":"2025-01-08 15:48:12.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e mentioned you in a comment","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-01-08 15:48:12.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e mentioned you in a comment","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-05-14 10:41:11.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"8878":{"account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"last_update":"2025-01-09 17:29:27.000000000","reason":"\u003cGERRIT_ACCOUNT_37442\u003e replied on the change","reason_account":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"}}},"removed_from_attention_set":{"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2025-05-14 13:06:41.000000000","reason":"\u003cGERRIT_ACCOUNT_16207\u003e replied on the change","reason_account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}}},"hashtags":[],"change_id":"I069aa57cad3bd836bcd974ad57decb28771e2278","subject":"Proposes \u0027Add Graceful Nova Compute Shutdown\u0027","status":"NEW","created":"2024-12-05 15:01:25.000000000","updated":"2025-10-20 18:31:05.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":24,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"14ecd7638019737338725d73f8b99c34f2445659","_number":937185,"virtual_id_number":937185,"owner":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-15 10:23:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":-1,"date":"2025-10-20 18:31:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-05 15:08:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-05 16:18:27.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2024-12-09 23:42:24.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-01-08 15:48:12.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"state":"CC"},{"updated":"2025-01-08 15:48:12.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2025-01-09 14:08:14.000000000","updated_by":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"CC"},{"updated":"2025-01-09 17:24:18.000000000","updated_by":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"CC"},{"updated":"2025-04-15 13:59:46.000000000","updated_by":{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},"reviewer":{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},"state":"CC"},{"updated":"2025-05-14 09:53:24.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2025-10-20 18:31:05.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"dfe76327c6b87c39159107dd6fa5142db8ee021f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2024-12-05 15:01:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58a1f4b72ccf74d9069b5940984beefa962387db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 15:08:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b8af9bcbd70455ab7200e62f579b018\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/641f4a39825a4c75a3bc9c9b7aaac787 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08aef433a54b408f9c0350069dcd1c98 : FAILURE in 2m 14s","accounts_in_message":[],"_revision_number":1},{"id":"bbc9f9cab5a6cf4c7bd4418b0567bf9aa360ce77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2024-12-05 15:27:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"da29f013bc683c4ad545b6166fb6306ad67c4308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 15:34:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e9960092d16477890c86a5bc8dd2491\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6f7efdc347148c9b052b7d19292a522 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b51eac69b3b40b193325bef8661fdd5 : FAILURE in 3m 01s","accounts_in_message":[],"_revision_number":2},{"id":"9a675ce59234159cb7f005a955b04cf9fe9655d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2024-12-05 15:39:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8230f4280290ad74121b0e8f3fa6548e78912545","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 15:46:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d243cbf0abc404bbc5058b1c5b69e47\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44dba6084fb4e16bc7bfaefc95e7008 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4465fa99bc0643059f4cf174c311496e : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":3},{"id":"67ca0b2ceb5493cf5ffaf00452d9d284555c9afb","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-05 16:18:27.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"798ae8bdc6f2e9724beb27fb0a028f02b4c5d978","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2024-12-05 20:20:32.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bd46719c6f7598331b876da768cf8ddb962be8fd","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-12-09 23:42:24.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d2c94f7042b8449632d18bf30b2f28a31824ea83","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-08 15:48:12.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"61c341db80043227542f88e9c186264848fdab96","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2025-01-09 14:08:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"39feb1c5eb53601f06a2e9c0314a0b41d77fced5","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2025-01-09 14:10:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f4ec28922e1d95165fb4557f89869bf610ef0417","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-01-09 17:24:18.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4b3d4d069ec9b55333c88a8ba45d11913c9594ae","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2025-01-09 17:29:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a2ba8e90605336ffa14ba493a4af9c7020216f1d","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-05-14 09:53:24.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"609d4cddbd8383275270abc4b72cd10aa3f8f412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2025-05-14 10:41:11.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"09286eb41e0bf3758aabaee04830ff385e8528fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 10:51:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a9c53eb04b44f2da9f3485a6b203311\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e3e9254054482bb2144458b9bcb421 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/648a0ece2fd24af7abacaf64b0859013 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":4},{"id":"ec75900f080fef486a4c99a80f3183eca1300ee0","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-05-14 13:06:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ac68e08928692988940a78c118db55b99c2ee579","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2025-05-15 10:10:02.000000000","message":"Topic set to graceful-compute-shutdown","accounts_in_message":[],"_revision_number":4},{"id":"1c9edbaae52d7f39df09c96037b368a0059d3a12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2025-05-15 10:11:43.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"9027166671d0757aa04c451b31d3724b5998b029","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2025-05-15 10:14:40.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"555024a384cdc205c3f9a24f5764b181038171de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 10:23:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb880cfd9a834e4890e601f7a56bdee5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd29c26cee974e808bffd34d4a330154 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b15226bf19954f7a8239a554d87e6bae : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":6},{"id":"14ecd7638019737338725d73f8b99c34f2445659","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-10-20 18:31:05.000000000","message":"Patch Set 6: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"8d8a38328bfe300a726eae7308f5389fa5915973","revisions":{"e74f8eb3c50278e5b56a946285d053b3ce301243":{"kind":"REWORK","_number":1,"created":"2024-12-05 15:01:25.000000000","uploader":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"ref":"refs/changes/85/937185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/937185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/1"}}},"commit":{"parents":[{"commit":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","subject":"Re-propose config option for behavior of unset unified limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]}],"author":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2024-12-05 14:58:13.000000000","tz":60},"committer":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2024-12-05 15:00:16.000000000","tz":60},"subject":"Proposes `Add Graceful Nova Compute Shutdown`","message":"Proposes `Add Graceful Nova Compute Shutdown`\n\nChange-Id: I069aa57cad3bd836bcd974ad57decb28771e2278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e74f8eb3c50278e5b56a946285d053b3ce301243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e74f8eb3c50278e5b56a946285d053b3ce301243"}]},"branch":"refs/heads/master"},"972965ff31486b18f9c7419af243fea603764e64":{"kind":"REWORK","_number":2,"created":"2024-12-05 15:27:07.000000000","uploader":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"ref":"refs/changes/85/937185/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/937185/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/2"}}},"commit":{"parents":[{"commit":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","subject":"Re-propose config option for behavior of unset unified limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]}],"author":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2024-12-05 14:58:13.000000000","tz":60},"committer":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2024-12-05 15:26:58.000000000","tz":60},"subject":"Proposes `Add Graceful Nova Compute Shutdown`","message":"Proposes `Add Graceful Nova Compute Shutdown`\n\nChange-Id: I069aa57cad3bd836bcd974ad57decb28771e2278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/972965ff31486b18f9c7419af243fea603764e64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/972965ff31486b18f9c7419af243fea603764e64"}]},"branch":"refs/heads/master"},"4ad1041c87b582096f48506318764f9318600cda":{"kind":"REWORK","_number":3,"created":"2024-12-05 15:39:38.000000000","uploader":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"ref":"refs/changes/85/937185/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/937185/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/3"}}},"commit":{"parents":[{"commit":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","subject":"Re-propose config option for behavior of unset unified limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]}],"author":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2024-12-05 14:58:13.000000000","tz":60},"committer":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2024-12-05 15:39:31.000000000","tz":60},"subject":"Proposes `Add Graceful Nova Compute Shutdown`","message":"Proposes `Add Graceful Nova Compute Shutdown`\n\nChange-Id: I069aa57cad3bd836bcd974ad57decb28771e2278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4ad1041c87b582096f48506318764f9318600cda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4ad1041c87b582096f48506318764f9318600cda"}]},"branch":"refs/heads/master"},"14ba1c46f6ed2bcb2060c99748b9b86d78993929":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-05-14 10:41:11.000000000","uploader":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"ref":"refs/changes/85/937185/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/937185/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/4"}}},"commit":{"parents":[{"commit":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","subject":"Re-propose config option for behavior of unset unified limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]}],"author":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2024-12-05 14:58:13.000000000","tz":60},"committer":{"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","date":"2025-05-14 10:41:11.000000000","tz":0},"subject":"Proposes `Add Graceful Nova Compute Shutdown`","message":"Proposes `Add Graceful Nova Compute Shutdown`\n\nBP: https://blueprints.launchpad.net/nova/+spec/graceful-compute-shutdown\n\nChange-Id: I069aa57cad3bd836bcd974ad57decb28771e2278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/14ba1c46f6ed2bcb2060c99748b9b86d78993929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/14ba1c46f6ed2bcb2060c99748b9b86d78993929"}]},"branch":"refs/heads/master","description":"Edit commit message"},"066de37ea65c5c4d7d9060f0840ba1d32a7cae8f":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-05-15 10:11:43.000000000","uploader":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"ref":"refs/changes/85/937185/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/937185/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/5"}}},"commit":{"parents":[{"commit":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","subject":"Re-propose config option for behavior of unset unified limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]}],"author":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2024-12-05 14:58:13.000000000","tz":60},"committer":{"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","date":"2025-05-15 10:11:43.000000000","tz":0},"subject":"Proposes \u0027Add Graceful Nova Compute Shutdown\u0027","message":"Proposes \u0027Add Graceful Nova Compute Shutdown\u0027\n\nRelated to blueprint graceful-compute-shutdown\n\nref: https://blueprints.launchpad.net/nova/+spec/graceful-compute-shutdown\n\nChange-Id: I069aa57cad3bd836bcd974ad57decb28771e2278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/066de37ea65c5c4d7d9060f0840ba1d32a7cae8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/066de37ea65c5c4d7d9060f0840ba1d32a7cae8f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8d8a38328bfe300a726eae7308f5389fa5915973":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-05-15 10:14:40.000000000","uploader":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"ref":"refs/changes/85/937185/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/937185/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/937185/6"}}},"commit":{"parents":[{"commit":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","subject":"Re-propose config option for behavior of unset unified limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]}],"author":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2024-12-05 14:58:13.000000000","tz":60},"committer":{"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","date":"2025-05-15 10:14:40.000000000","tz":0},"subject":"Proposes \u0027Add Graceful Nova Compute Shutdown\u0027","message":"Proposes \u0027Add Graceful Nova Compute Shutdown\u0027\n\nRelated to blueprint graceful-compute-shutdown\n\nChange-Id: I069aa57cad3bd836bcd974ad57decb28771e2278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d8a38328bfe300a726eae7308f5389fa5915973"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d8a38328bfe300a726eae7308f5389fa5915973"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
