)]}'
{"id":"openstack%2Fnova-specs~937615","triplet_id":"openstack%2Fnova-specs~master~I9d44e9e16d17a86093aa2bf2be56f3d893b358ca","project":"openstack/nova-specs","branch":"master","topic":"bp/migrate-vfio-devices-using-kernel-variant-drivers","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-01-10 08:15:31.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-01-10 08:08:30.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2025-01-10 08:15:31.000000000","reason":"Change was submitted"},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2025-01-10 08:15:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9d44e9e16d17a86093aa2bf2be56f3d893b358ca","subject":"Migrate VFIO devices using kernel variant drivers","status":"MERGED","created":"2024-12-12 11:59:50.000000000","updated":"2025-01-15 18:10:41.000000000","submitted":"2025-01-10 08:15:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":161,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"937615-bp/migrate-vfio-devices-using-kernel-variant-drivers","meta_rev_id":"b676d342ee0c557b2b5a0d0ec6089bceed6a1c94","_number":937615,"virtual_id_number":937615,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 08:15:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2025-01-09 18:52:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-10 08:08:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-10 08:09:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-12 12:14:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-12 15:08:38.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2024-12-13 13:32:51.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-01-07 13:42:40.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2025-01-07 19:41:24.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"0ab08ee8369451840efe059fdcad551e1bb7297c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-12-12 11:59:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e157f6c3b7b761f807227363782b4787e12f8e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 12:14:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/941e41a1338a4cf384816eea55f13d02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f396d585853468ab4b864f2f7acdc13 : FAILURE in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acc06671998c4b3bb9810f018adccaaa : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"4d9df698ec15e61cbfbc935ce96835875a735fcd","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-12-12 15:08:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(20 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cfd5bccd40609cefb0b78003497f2d60bcec9f3e","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-12-12 20:26:49.000000000","message":"Patch Set 1:\n\n(11 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9f716195c6209f005885df40b2cc7cefa2fe9460","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2024-12-12 20:34:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"700e492c6c0455dfc3659bad6686d2bd18644599","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 20:41:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3133c26eece43b4b9b2809ef685945f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96823746d6c94796bd7ee95fb9094268 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84e406f13faf47b1947935c531edeb82 : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":2},{"id":"384cb9dfc30d368a62eddf10105937c9f7ef05b5","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-13 13:32:51.000000000","message":"Patch Set 2: Code-Review-1\n\n(19 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ace9c901d5b792531bb7a56e07220d68435e5ee3","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-01-03 17:11:30.000000000","message":"Patch Set 2:\n\n(18 comments)","accounts_in_message":[],"_revision_number":2},{"id":"488af5ead808ef3c86fc0e833b8bbf88d6f0ee80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-01-03 17:18:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"92abae4fc8ee0042b05a34aa882872c825365a20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 17:26:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9940ffd42814c1bb3873a2b91fb534a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76619de42ce54ea6b1f72d5175814ae4 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd525074423944c5817c736d73b913fa : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":3},{"id":"a3f4f50b989ad3c7ac2f66bc7c6b19457dc2aa88","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-07 13:23:47.000000000","message":"Patch Set 3: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7e3ed641562dc287be22da5a26417e6373d40b7f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-07 13:42:40.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"84df61f025b4a10b286a0edee919924d317a42e5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-07 19:41:24.000000000","message":"Patch Set 3: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":3},{"id":"353a7d272cd14c5f4335990403d10bac667237a5","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-01-08 09:16:35.000000000","message":"Patch Set 3: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bdbdd6f93aa54e975a7d093edfc9b09cab9808f2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-08 11:25:26.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4fe6410518b4138186919f5e07b84676f9cdcd21","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-01-08 17:04:41.000000000","message":"Patch Set 3:\n\n(19 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fdc4a8a0245f95e4ca0308750002ba7a85bbbe86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-01-08 17:04:58.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a03125c17bff0900b0b9cfc95a80ec1961019bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 17:17:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c46915fc4bcd4f76b7c2f364545418de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/277a254a9c04403dbbd687039527ea39 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef2607f7483d4c63bc1a57890a5d96a0 : FAILURE in 3m 01s","accounts_in_message":[],"_revision_number":4},{"id":"7b210771ca2db7c619746a507e0770938b6b1749","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-08 17:57:56.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7404a5476be0078fa4128b7bb0ed9461f579cd26","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-01-09 11:37:15.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"79171d0126997e1af993fb5e8af033241216ab5c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-09 12:46:40.000000000","message":"Patch Set 4: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5798123ee3ab584376502554e947222175994f0e","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-01-09 13:29:57.000000000","message":"Patch Set 4: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9a795671863d7fa7d55cd93d1ab5c097bc7d5ad9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-01-09 16:58:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"72f81d9001ab7153f531d9372ab644af908a11a5","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-01-09 17:01:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fa91f60898265f37bd36e00b1205d2214cee80ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 17:04:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51e84d778c394cd8a4d53f0f7ac2a1e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbad5e5b61144f1f922a6f2bbec550e4 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/933dc4243cf94d88a9b2ae3e3a3098e3 : FAILURE in 2m 28s","accounts_in_message":[],"_revision_number":5},{"id":"a081662882b4bfa81fe516465abe28873a241235","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-01-09 17:21:59.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"b829ce5dd97827dc1d1834c789ada176e280b70a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 17:29:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b277bb12e1914252837e560ba80c2475\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9be2a85b59764a1abdfa936e5f8284f3 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91ba649f140e44aaba9502b8d5a34be1 : FAILURE in 3m 00s","accounts_in_message":[],"_revision_number":6},{"id":"68281c83233b680aaaa193efe66418cf4345f00d","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-01-09 17:44:09.000000000","message":"Patch Set 6: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"db0b528600e3ace6c9ac88c8b4174ec065742dd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-01-09 17:55:15.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"bff7f819954920c637789e44e744042dff995802","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-09 17:59:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"96af04838fbc8033e5647fceb2dcbbcde551a5f5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-09 18:01:23.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bb2b5d7582002d9f479e7e93ca2be87d9fad5e97","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-01-09 18:13:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"29c02c1f87d89089a627036e24b695cffc177a96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 18:16:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab68e3c928e0441fbc9f9fcb7f3b76a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ded71f1799a4608abec0dd1eb9caf47 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d71770a762d401b98860b216a0f7409 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":7},{"id":"79586bfef17bf8d135a43ed340862f6a0e0be01c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-09 18:52:32.000000000","message":"Patch Set 7: Code-Review+2\n\n(7 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4997d1c74095ae62beacf7bb33379c59cc92c7db","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-09 20:23:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a6a07cc87082af6825f738fb74f603c7356a4db4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-10 08:08:30.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cd8157d54463680783680e1c8e27289db9527d0b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-10 08:09:23.000000000","message":"Patch Set 7: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a82b050fd93ef4eb84d3dfa4eafc630e72d973ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 08:09:41.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"95ece512f15b8af0b81d687c65e4487a4d12e5ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 08:15:31.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dad55dbcb4a418c9a853994028ff3f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0653b97cfe42e5bb6744476f1b52d0 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7e6be43c1684d57961bad59b4060507 : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":7},{"id":"e3bafed90319fdbe34f9c98dd9758740fb17a9fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 08:15:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"03e0cb958adcc79bb0fb5dc6d53d037f47d5f2d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 08:17:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/769d5346bc8e4fa188a597d6a5a1e49f\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/2bb1c0754d254aabbf1fdce44916fb7c : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":7},{"id":"b676d342ee0c557b2b5a0d0ec6089bceed6a1c94","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-01-15 18:10:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9b433f0025b2fd4f81e87f2e595a03808accba0b","revisions":{"82744730b28336487fed793b50f53c79bbc63f78":{"kind":"REWORK","_number":1,"created":"2024-12-12 11:59:50.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/15/937615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/15/937615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/1"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-12-12 11:51:23.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-12-12 11:58:33.000000000","tz":60},"subject":"WIP: Migrate VFIO devices using kernel variant drivers","message":"WIP: Migrate VFIO devices using kernel variant drivers\n\nThis spec outlines the necessary steps to migrate SR-IOV devices\nusing the new kernel variant driver interface.\n\nImplements: blueprint migrate-vfio-devices-using-kernel-variant-drivers\nChange-Id: I9d44e9e16d17a86093aa2bf2be56f3d893b358ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/82744730b28336487fed793b50f53c79bbc63f78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/82744730b28336487fed793b50f53c79bbc63f78"}]},"branch":"refs/heads/master"},"d21e0686c0b114bd78460c07f27e3c60d73562ef":{"kind":"REWORK","_number":2,"created":"2024-12-12 20:34:56.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/15/937615/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/15/937615/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/2"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-12-12 11:51:23.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-12-12 20:34:20.000000000","tz":60},"subject":"Migrate VFIO devices using kernel variant drivers","message":"Migrate VFIO devices using kernel variant drivers\n\nThis spec outlines the necessary steps to migrate SR-IOV devices\nusing the new kernel variant driver interface.\n\nImplements: blueprint migrate-vfio-devices-using-kernel-variant-drivers\nChange-Id: I9d44e9e16d17a86093aa2bf2be56f3d893b358ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d21e0686c0b114bd78460c07f27e3c60d73562ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d21e0686c0b114bd78460c07f27e3c60d73562ef"}]},"branch":"refs/heads/master"},"6b76defb2c44c71728f336f29f11ffc269cdf5fd":{"kind":"REWORK","_number":3,"created":"2025-01-03 17:18:59.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/15/937615/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/15/937615/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/3"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-12-12 11:51:23.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-01-03 17:18:37.000000000","tz":60},"subject":"Migrate VFIO devices using kernel variant drivers","message":"Migrate VFIO devices using kernel variant drivers\n\nThis spec outlines the necessary steps to migrate SR-IOV devices\nusing the new kernel variant driver interface.\n\nImplements: blueprint migrate-vfio-devices-using-kernel-variant-drivers\nChange-Id: I9d44e9e16d17a86093aa2bf2be56f3d893b358ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6b76defb2c44c71728f336f29f11ffc269cdf5fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6b76defb2c44c71728f336f29f11ffc269cdf5fd"}]},"branch":"refs/heads/master"},"48c3bb127aa3a8729ef28479b22fd058002485a7":{"kind":"REWORK","_number":4,"created":"2025-01-08 17:04:58.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/15/937615/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/15/937615/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/4"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-12-12 11:51:23.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-01-08 16:52:42.000000000","tz":60},"subject":"Migrate VFIO devices using kernel variant drivers","message":"Migrate VFIO devices using kernel variant drivers\n\nThis spec outlines the necessary steps to migrate SR-IOV devices\nusing the new kernel variant driver interface.\n\nImplements: blueprint migrate-vfio-devices-using-kernel-variant-drivers\nChange-Id: I9d44e9e16d17a86093aa2bf2be56f3d893b358ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/48c3bb127aa3a8729ef28479b22fd058002485a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/48c3bb127aa3a8729ef28479b22fd058002485a7"}]},"branch":"refs/heads/master"},"9c21ae4d38a5d94bf038b77505acaa1b27c3674f":{"kind":"REWORK","_number":5,"created":"2025-01-09 16:58:49.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/15/937615/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/15/937615/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/5"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-12-12 11:51:23.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-01-09 16:58:32.000000000","tz":60},"subject":"Migrate VFIO devices using kernel variant drivers","message":"Migrate VFIO devices using kernel variant drivers\n\nThis spec outlines the necessary steps to migrate SR-IOV devices\nusing the new kernel variant driver interface.\n\nImplements: blueprint migrate-vfio-devices-using-kernel-variant-drivers\nChange-Id: I9d44e9e16d17a86093aa2bf2be56f3d893b358ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9c21ae4d38a5d94bf038b77505acaa1b27c3674f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9c21ae4d38a5d94bf038b77505acaa1b27c3674f"}]},"branch":"refs/heads/master"},"ad4ab77cc43c367c5ef1f72940c64b10780f81ef":{"kind":"REWORK","_number":6,"created":"2025-01-09 17:21:59.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/15/937615/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/15/937615/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/6"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-12-12 11:51:23.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-01-09 17:21:43.000000000","tz":60},"subject":"Migrate VFIO devices using kernel variant drivers","message":"Migrate VFIO devices using kernel variant drivers\n\nThis spec outlines the necessary steps to migrate SR-IOV devices\nusing the new kernel variant driver interface.\n\nImplements: blueprint migrate-vfio-devices-using-kernel-variant-drivers\nChange-Id: I9d44e9e16d17a86093aa2bf2be56f3d893b358ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ad4ab77cc43c367c5ef1f72940c64b10780f81ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ad4ab77cc43c367c5ef1f72940c64b10780f81ef"}]},"branch":"refs/heads/master"},"9b433f0025b2fd4f81e87f2e595a03808accba0b":{"kind":"REWORK","_number":7,"created":"2025-01-09 17:55:15.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/15/937615/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/15/937615/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/15/937615/7"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2024-12-12 11:51:23.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-01-09 17:50:50.000000000","tz":60},"subject":"Migrate VFIO devices using kernel variant drivers","message":"Migrate VFIO devices using kernel variant drivers\n\nThis spec outlines the necessary steps to migrate SR-IOV devices\nusing the new kernel variant driver interface.\n\nImplements: blueprint migrate-vfio-devices-using-kernel-variant-drivers\nChange-Id: I9d44e9e16d17a86093aa2bf2be56f3d893b358ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9b433f0025b2fd4f81e87f2e595a03808accba0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9b433f0025b2fd4f81e87f2e595a03808accba0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
