)]}'
{"id":"openstack%2Fnova-specs~937650","triplet_id":"openstack%2Fnova-specs~master~Ie32d6e65bf7df3aac13c0c3dff6a467484cba745","project":"openstack/nova-specs","branch":"master","topic":"bp/policy-manager-role-default","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2025-06-06 22:32:54.000000000","reason":"\u003cGERRIT_ACCOUNT_4690\u003e replied on the change","reason_account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-06-06 22:38:10.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-06-06 22:38:10.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-06-06 22:38:10.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-06-06 22:38:10.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2025-06-06 22:38:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie32d6e65bf7df3aac13c0c3dff6a467484cba745","subject":"Propose API policy manager role spec","status":"MERGED","created":"2024-12-12 20:51:35.000000000","updated":"2025-06-06 22:39:30.000000000","submitted":"2025-06-06 22:38:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":82,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"937650-bp/policy-manager-role-default","meta_rev_id":"8260cfd43bcc57735cb8b2a8555ef3688cde90a2","_number":937650,"virtual_id_number":937650,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-06 22:38:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2025-06-06 22:32:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2025-06-06 00:38:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2025-06-05 13:05:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2025-06-06 22:32:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-12 21:03:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-13 02:25:52.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2024-12-13 08:37:11.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2024-12-16 08:09:45.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2025-01-08 14:07:53.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-01-09 11:49:10.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-03-12 19:27:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-06-05 13:05:09.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"beda29357225193a7af2411f6784affe4e14e1ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-12-12 20:51:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a846142670f6be94507679f0c62b96c0b9d691d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-12-12 20:54:06.000000000","message":"Topic set to bp/policy-service-and-manager-role-default","accounts_in_message":[],"_revision_number":1},{"id":"9e3345ddd27dd8cd7834304f74243a97b2deeb1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 21:03:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e390aa6ac69142b2a9dfcf787def2f9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/185592a494d7486b8231564b5a303973 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c48703f0bbfd4c3daac79e9a000dea9f : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b86feff536a01deb63514e539e9957ca460b6311","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-12-13 02:25:52.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b13fe9dc288336893b80be085a944cdc02085251","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-12-13 08:37:11.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6d3d2aecbc962117541f9f443f9454a224dda1f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-12-13 21:20:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec205328d6e1ac076f38a0a1addf382df88389fa","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-12-13 21:23:06.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6ca2a7d7c9960f3ea80abeb9e81425e4b25ea3b7","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-12-16 08:09:45.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0b80a74ee48f250d220bf13e1728a684b6d2b6e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-12-17 04:32:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7b16f15f84bac9f9d3abfd5faec2504734a71e68","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-12-17 04:32:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"487115e7ecba3f330c810a6e54bd9cf0a3bc45e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 04:37:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/492bf3e122704652aa76436a856d8c27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df9d2af5d3cf4ecdac2c8aa4e3530abf : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aae2cccc90414a00a4e4108330d43fdb : FAILURE in 2m 12s","accounts_in_message":[],"_revision_number":2},{"id":"58165538e1607db32f65b75d7222538086b4b52a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-12-17 05:07:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a7938851617b9376ab42562a65517f6d3c3e676b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 05:22:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77fc6ebda20e404da63e7cd686961ce2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccd674c3266c41b9a60a210ae9bde306 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb3bccf27ee54000b72bee5ae57de377 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":3},{"id":"5e4e4d4685843d1df812821b7926ca919bfce25c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-12-17 05:23:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"55f25636e92cd5290ffb7def64f32a4b6fb5e6ba","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-08 14:07:53.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e90472133b233dfefcde9df137bae244bf429b79","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-01-08 15:20:01.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2d5fdfae891f0d7958b5155abff6cac6173c4275","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-09 01:40:57.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"78e4c22ba91991dac1bd321387f1d0bbb0ec7693","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-09 02:54:15.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a922f1cffc9b6ecb2c6f6647b2744e7c118020aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 03:07:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35b90d8394224e8bad87ba4e6114070d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edb48668a4254e718335598df333b37e : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eef1827217af469da9b66834af58d500 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":4},{"id":"97ee3f80d138816136f4d9daf675402a82aaed56","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-09 11:49:10.000000000","message":"Patch Set 4: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6310fb37add327b91b647c172ebf2898cf689c02","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-09 12:01:30.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"35e189118f30370a47218ad7aae0d7985b0ccc62","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-09 16:28:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"42e4c7db7eb3cff72efcf837c72b409f74c03c05","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-10 21:22:19.000000000","message":"Patch Set 4:\n\n(10 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9ef60444a71efe2004c6ee6803b72709f5132666","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-10 22:26:00.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"a2745fe57935910be1087f46569f59b467fb89ae","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-10 22:26:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f29a83ff35f4ef3bb0e1a0ec2eea256972e4a90c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 22:32:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82cefd5d133244f9b7894f33548fafad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bdd751bc7e84c6596d913d2b2117911 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32f9a7c212024a6da61239aca7a6de14 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":5},{"id":"7be78d8e44e4359cbc2e5989fe1aa48ab6da62fb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-13 13:42:08.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6f6d4d2d3883cb81e84e0aa15b972896f2432ede","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-15 18:22:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0c11ede39084d0704fc7e90f59a9f906c8e05165","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-15 18:29:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ede0ea0863860cc5efa5f8d330aa827aa754b535","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-15 18:50:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e55e6b7f8bf8c55cbd7a871b82d2b55970d0a576","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-20 16:30:58.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8582180fefff9b9fcfdf05ff38d9ec80639083ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-29 00:15:52.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3e569caf256e48973f98f944f0b193a2b4f909f8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-29 00:16:33.000000000","message":"Topic bp/policy-service-and-manager-role-default removed","accounts_in_message":[],"_revision_number":6},{"id":"433fa52c0c30927289fd940ead5153ac98e0d5a1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-29 00:16:40.000000000","message":"Topic set to bp/policy-manager-role-default","accounts_in_message":[],"_revision_number":6},{"id":"e3b3a5b3bc7f4f4d94beb06fcc287c7295e82fbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 00:23:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cb1375eb2744166ada70167708d65f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f6acbec45474062a1fb5638ebd85717 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1913c0739fc4e20b5d0868ce5b44813 : FAILURE in 3m 03s","accounts_in_message":[],"_revision_number":6},{"id":"a520b28ddfbaf33157db567f841d0098525795e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-29 00:27:04.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"6700db13ffd6f37a07a37225693672b11f031e78","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-29 00:28:04.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2b0fee429dcfd9c6bbe0a5510f8e4989b21f5821","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 00:34:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c6115f0163843aeab22636af5f0e772\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74c857779838489fade4388b49ef282d : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5626377716da45dc8e41a41b3be586f7 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":7},{"id":"db42a8e6bc14a4ce653e71e0cd44f70a7419b59b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-29 15:44:27.000000000","message":"Patch Set 7: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":7},{"id":"14b2899b0a8d2829161ecc34c462887f92c0d587","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-29 18:01:57.000000000","message":"Patch Set 7:\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"970845f5bab4381d1d49bdedd06b078c41fa2f91","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-29 18:08:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6a6d6ed260926716d44faa38129ecf3c1e48a97f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-29 18:24:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"325cee89fa84be01640d6b9898dd0128c7e09675","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-05-30 07:02:56.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"426aeab1a00da95d768ce3a74765527731070f28","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-30 15:10:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8ae512d806ce92ad1883b76074653db535f240fc","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-30 15:32:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4ba9e66eefe2f32b46b8f0a3ad825a6954357caa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-30 15:45:40.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"3b7a3db429e3e53ea702403cb8747d5b5f0f0f69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 16:25:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43b646be5fe14258bc03ad54d7c01369\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a51f031f9aa74c439a9d479e3d5900af : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c72c8bf0c3eb40aa90481095174eac3d : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":8},{"id":"511c2a31eaa0af02486c6188bb25594d4455e06c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-30 18:09:43.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"41fac423b73e4137171a939dbcd2a851af599280","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-30 18:10:37.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"79c07b43eea9c93959546b356c0253de939f63c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 18:18:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f474aa90d374540bca0f75b6368890d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ca14645dee241f08280bcdabb603436 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87caccdf45294d3c8e3afef2dff30995 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":9},{"id":"8b7c618998d70f16831dac4be8c641f54674699d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-05 13:05:09.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e8ce87c192ab58dfd581aa1abd04ddf1f92c9e2f","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-06-06 00:00:34.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3b769c7438ae478f01b9f8dd040c3429cac31e64","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-06 00:38:03.000000000","message":"Patch Set 9: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a83a11c21b4e5d61f748096e13396195f7eb2fe1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-06 00:40:55.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6b698a932533a1e6148e2924836d95a57e8ea9b1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-06 18:47:16.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8902b9a0018342b27dfa691c742917ff9fadf367","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-06-06 22:32:54.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"06128ccdc0ca291c8626a20bd5683e76f129effe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 22:33:02.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"3563816d5b72d6c4bea7a99d301cab1241b2da97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 22:38:10.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3a174cfdca5499bbbb3a5ed2f87f9ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/629fea953c49411b886ed1a24ec0f250 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02c0935b2b274da2a79853a0fbfe75b5 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":9},{"id":"0610099d3e0556264a193f837d12b13b9afa68e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 22:38:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"8260cfd43bcc57735cb8b2a8555ef3688cde90a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 22:39:30.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93ed7e8ea850417f882f27dce18e4c72\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/dd4c8684ec024985b2df65fa1fe2e727 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"da3d034db214c26254568af4ba19fb36af0081fb","revisions":{"a0dfe8a8361542067d0f4402267e0e3aba0db019":{"kind":"REWORK","_number":1,"created":"2024-12-12 20:51:35.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/937650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/937650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/1"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-12-12 20:49:34.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-12-12 20:51:34.000000000","tz":-480},"subject":"Propose API policy service and manager role spec","message":"Propose API policy service and manager role spec\n\nThis spec re-proposes the service role to nova service-to-service\nAPIs policies and manager role for project level APIs.\n\nPartial implement blueprint policy-service-and-manager-role-default\n\nAPIImpact\n\nChange-Id: Ie32d6e65bf7df3aac13c0c3dff6a467484cba745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a0dfe8a8361542067d0f4402267e0e3aba0db019"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a0dfe8a8361542067d0f4402267e0e3aba0db019"}]},"branch":"refs/heads/master"},"4003f79e8655c82d98e31abb3182e276b103798c":{"kind":"REWORK","_number":2,"created":"2024-12-17 04:32:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/937650/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/937650/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/2"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-12-12 20:49:34.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-12-17 04:31:25.000000000","tz":-480},"subject":"Propose API policy service and manager role spec","message":"Propose API policy service and manager role spec\n\nThis spec proposes the service role to nova service-to-service\nAPIs policies and manager role for project level APIs.\n\nNOTE: service role proposal is re-proposed but manager role proposal\nis new in this cycle.\n\nPartial implement blueprint policy-service-and-manager-role-default\n\nAPIImpact\n\nChange-Id: Ie32d6e65bf7df3aac13c0c3dff6a467484cba745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4003f79e8655c82d98e31abb3182e276b103798c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4003f79e8655c82d98e31abb3182e276b103798c"}]},"branch":"refs/heads/master"},"90bfb079e670b981929494db8cab678ea7679a80":{"kind":"REWORK","_number":3,"created":"2024-12-17 05:07:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/937650/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/937650/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/3"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-12-12 20:49:34.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-12-17 05:07:11.000000000","tz":-480},"subject":"Propose API policy service and manager role spec","message":"Propose API policy service and manager role spec\n\nThis spec proposes the service role to nova service-to-service\nAPIs policies and manager role for project level APIs.\n\nNOTE: service role proposal is re-proposed but manager role proposal\nis new in this cycle.\n\nPartial implement blueprint policy-service-and-manager-role-default\n\nAPIImpact\n\nChange-Id: Ie32d6e65bf7df3aac13c0c3dff6a467484cba745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/90bfb079e670b981929494db8cab678ea7679a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/90bfb079e670b981929494db8cab678ea7679a80"}]},"branch":"refs/heads/master"},"1989c67d5e377c77a0d2885d080fdf3d35b5f1e1":{"kind":"REWORK","_number":4,"created":"2025-01-09 02:54:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/937650/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/937650/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/4"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-12-12 20:49:34.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-01-09 02:54:07.000000000","tz":-480},"subject":"Propose API policy service and manager role spec","message":"Propose API policy service and manager role spec\n\nThis spec proposes the service role to nova service-to-service\nAPIs policies and manager role for project level APIs.\n\nNOTE: service role proposal is re-proposed but manager role proposal\nis new in this cycle.\n\nPartial implement blueprint policy-service-and-manager-role-default\n\nAPIImpact\n\nChange-Id: Ie32d6e65bf7df3aac13c0c3dff6a467484cba745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1989c67d5e377c77a0d2885d080fdf3d35b5f1e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1989c67d5e377c77a0d2885d080fdf3d35b5f1e1"}]},"branch":"refs/heads/master"},"53bd67a092a972f592a8e70eb3b2614649dbed7d":{"kind":"REWORK","_number":5,"created":"2025-01-10 22:26:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/937650/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/937650/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/5"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-12-12 20:49:34.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-01-10 22:25:51.000000000","tz":-480},"subject":"Propose API policy service and manager role spec","message":"Propose API policy service and manager role spec\n\nThis spec proposes the service role to nova service-to-service\nAPIs policies and manager role for project level APIs.\n\nNOTE: service role proposal is re-proposed but manager role proposal\nis new in this cycle.\n\nPartial implement blueprint policy-service-and-manager-role-default\n\nAPIImpact\n\nChange-Id: Ie32d6e65bf7df3aac13c0c3dff6a467484cba745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/53bd67a092a972f592a8e70eb3b2614649dbed7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/53bd67a092a972f592a8e70eb3b2614649dbed7d"}]},"branch":"refs/heads/master"},"4ad9a83e16e9e5ddf72e73353c9988b577c2b28b":{"kind":"REWORK","_number":6,"created":"2025-05-29 00:15:52.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/937650/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/937650/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/6"}}},"commit":{"parents":[{"commit":"9d9d0ccf53b45c55664f6cb6ff033fe0dd5c96a0","subject":"Add reminder about Tempest API response schemas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d9d0ccf53b45c55664f6cb6ff033fe0dd5c96a0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-12-12 20:49:34.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmaan@ghanshyammann.com","date":"2025-05-29 00:15:15.000000000","tz":-420},"subject":"Propose API policy manager role spec","message":"Propose API policy manager role spec\n\nPartial implement blueprint policy-manager-role-default\n\nAPIImpact\n\nChange-Id: Ie32d6e65bf7df3aac13c0c3dff6a467484cba745\nSigned-off-by: Ghanshyam \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4ad9a83e16e9e5ddf72e73353c9988b577c2b28b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4ad9a83e16e9e5ddf72e73353c9988b577c2b28b"}]},"branch":"refs/heads/master"},"7dfa09d5e51f6b5dfd14c8a1ff2e92c11158ed5a":{"kind":"REWORK","_number":7,"created":"2025-05-29 00:27:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/937650/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/937650/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/7"}}},"commit":{"parents":[{"commit":"9d9d0ccf53b45c55664f6cb6ff033fe0dd5c96a0","subject":"Add reminder about Tempest API response schemas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d9d0ccf53b45c55664f6cb6ff033fe0dd5c96a0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-12-12 20:49:34.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmaan@ghanshyammann.com","date":"2025-05-29 00:26:21.000000000","tz":-420},"subject":"Propose API policy manager role spec","message":"Propose API policy manager role spec\n\nPartial implement blueprint policy-manager-role-default\n\nAPIImpact\n\nChange-Id: Ie32d6e65bf7df3aac13c0c3dff6a467484cba745\nSigned-off-by: Ghanshyam \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7dfa09d5e51f6b5dfd14c8a1ff2e92c11158ed5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7dfa09d5e51f6b5dfd14c8a1ff2e92c11158ed5a"}]},"branch":"refs/heads/master"},"8add454e407e351c8d7436c2d9432b216a9bf8c9":{"kind":"REWORK","_number":8,"created":"2025-05-30 15:45:40.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/937650/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/937650/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/8"}}},"commit":{"parents":[{"commit":"4b8f26f46e7c5e8c714516719cff66df519d2d19","subject":"Merge \"Add remove-os-volumes_boot-api spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4b8f26f46e7c5e8c714516719cff66df519d2d19"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-12-12 20:49:34.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmaan@ghanshyammann.com","date":"2025-05-30 15:45:33.000000000","tz":-420},"subject":"Propose API policy manager role spec","message":"Propose API policy manager role spec\n\nPartial implement blueprint policy-manager-role-default\n\nAPIImpact\n\nChange-Id: Ie32d6e65bf7df3aac13c0c3dff6a467484cba745\nSigned-off-by: Ghanshyam \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8add454e407e351c8d7436c2d9432b216a9bf8c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8add454e407e351c8d7436c2d9432b216a9bf8c9"}]},"branch":"refs/heads/master"},"da3d034db214c26254568af4ba19fb36af0081fb":{"kind":"REWORK","_number":9,"created":"2025-05-30 18:10:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/937650/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/937650/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/937650/9"}}},"commit":{"parents":[{"commit":"4b8f26f46e7c5e8c714516719cff66df519d2d19","subject":"Merge \"Add remove-os-volumes_boot-api spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4b8f26f46e7c5e8c714516719cff66df519d2d19"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-12-12 20:49:34.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmaan@ghanshyammann.com","date":"2025-05-30 18:10:30.000000000","tz":-420},"subject":"Propose API policy manager role spec","message":"Propose API policy manager role spec\n\nPartial implement blueprint policy-manager-role-default\n\nAPIImpact\n\nChange-Id: Ie32d6e65bf7df3aac13c0c3dff6a467484cba745\nSigned-off-by: Ghanshyam \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/da3d034db214c26254568af4ba19fb36af0081fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/da3d034db214c26254568af4ba19fb36af0081fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
