)]}'
{"id":"openstack%2Fnova-specs~938276","triplet_id":"openstack%2Fnova-specs~master~I12279e15a9e8fe2df6f0fd5f87ddab41eb140be6","project":"openstack/nova-specs","branch":"master","topic":"bp/arm-cca-libvirt-support","attention_set":{},"removed_from_attention_set":{"35307":{"account":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"last_update":"2025-02-07 11:14:34.000000000","reason":"\u003cGERRIT_ACCOUNT_35307\u003e replied on the change","reason_account":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-07-04 11:35:22.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I12279e15a9e8fe2df6f0fd5f87ddab41eb140be6","subject":"Add a spec for libvirt driver launching Arm CCA-encrypted instances","status":"NEW","created":"2024-12-27 03:15:00.000000000","updated":"2025-07-04 11:35:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":3,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"9cce0c8a40872177af4cd39a40b12c6d89693a83","_number":938276,"virtual_id_number":938276,"owner":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-07 11:25:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"all":[{"value":-1,"date":"2025-02-07 11:14:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-27 03:24:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-08 14:20:47.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2025-01-08 14:20:51.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"feccf08f46c85c1bd1932ca3ae12c963eb663182","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"date":"2024-12-27 03:15:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bf27191dc33d02b718c644c3b10684a972e9a12","author":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"date":"2024-12-27 03:24:27.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b4b881c5e2588b19002e6d916956103a5ca3b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 03:24:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f917a1c239f48d69444cea09519ce90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5855251070a4943a5e77fdeeb0f1184 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc810a8a3654e95973bcd659f1a2233 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f15c192a720f7786489f92d194abb485f729f1ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"date":"2024-12-27 06:48:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"03429647346fa361a5b65db09ccf56fc2471caa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 07:25:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/535aa2a97bf04398961b3d2cd2081bfc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c171cd9c51864ec2b391993b09519b06 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bd88f30788544b0a4a8a484735dd693 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"ae1bacc06b82e550d48d23c07cf9834aad23ea64","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-08 14:20:47.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"abcbfe2e5d67e283c0194923ee43fe8283e42305","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-08 14:20:51.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"e92187f471beed262aef09cacea1c4d47519cbb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"date":"2025-02-07 11:10:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"46dfe929a9ab1cc11a64b5be0e990af716f44ab8","author":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"date":"2025-02-07 11:14:34.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bd2d547589b895f05c3dceab60c41cd5807cb0e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 11:25:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61725b44b9b849a78d84ac78a86fe2f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1214bf36726f4f2695654c19cc889df3 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fc947255f924c5cb57063075a6f580d : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":3},{"id":"9cce0c8a40872177af4cd39a40b12c6d89693a83","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"date":"2025-07-04 11:35:22.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8158ca9138d339b9c5422a7784687ba26bbf9660","revisions":{"9c11c43a47472f547f1eb09590c2c057a86c70f4":{"kind":"REWORK","_number":1,"created":"2024-12-27 03:15:00.000000000","uploader":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"ref":"refs/changes/76/938276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/76/938276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/76/938276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/76/938276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/76/938276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/76/938276/1"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","date":"2024-12-27 03:11:44.000000000","tz":540},"committer":{"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","date":"2024-12-27 03:11:44.000000000","tz":540},"subject":"Add a spec for libvirt driver launching Arm CCA-encrypted instances","message":"Add a spec for libvirt driver launching Arm CCA-encrypted instances\n\nThis spec proposes work required for nova\u0027s libvirt driver to\nsupport launching instances encrypted using\nArm CCA (Confidential Computing Architecture) technology:\n\nhttps://www.arm.com/architecture/security-features/arm-confidential-compute-architecture\n\nImplements: blueprint arm-cca-libvirt-support\nChange-Id: I12279e15a9e8fe2df6f0fd5f87ddab41eb140be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9c11c43a47472f547f1eb09590c2c057a86c70f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9c11c43a47472f547f1eb09590c2c057a86c70f4"}]},"branch":"refs/heads/master"},"ee75a0677b97824dbb6cf3ea7fab4d8e3dec0186":{"kind":"REWORK","_number":2,"created":"2024-12-27 06:48:40.000000000","uploader":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"ref":"refs/changes/76/938276/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/76/938276/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/76/938276/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/76/938276/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/76/938276/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/76/938276/2"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","date":"2024-12-27 03:11:44.000000000","tz":540},"committer":{"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","date":"2024-12-27 06:18:04.000000000","tz":540},"subject":"Add a spec for libvirt driver launching Arm CCA-encrypted instances","message":"Add a spec for libvirt driver launching Arm CCA-encrypted instances\n\nThis spec proposes work required for nova\u0027s libvirt driver to\nsupport launching instances encrypted using\nArm CCA (Confidential Computing Architecture) technology:\n\nhttps://www.arm.com/architecture/security-features/arm-confidential-compute-architecture\n\nImplements: blueprint arm-cca-libvirt-support\nChange-Id: I12279e15a9e8fe2df6f0fd5f87ddab41eb140be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee75a0677b97824dbb6cf3ea7fab4d8e3dec0186"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee75a0677b97824dbb6cf3ea7fab4d8e3dec0186"}]},"branch":"refs/heads/master"},"8158ca9138d339b9c5422a7784687ba26bbf9660":{"kind":"REWORK","_number":3,"created":"2025-02-07 11:10:35.000000000","uploader":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"ref":"refs/changes/76/938276/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/76/938276/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/76/938276/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/76/938276/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/76/938276/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/76/938276/3"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","date":"2024-12-27 03:11:44.000000000","tz":540},"committer":{"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","date":"2025-02-07 11:07:25.000000000","tz":540},"subject":"Add a spec for libvirt driver launching Arm CCA-encrypted instances","message":"Add a spec for libvirt driver launching Arm CCA-encrypted instances\n\nThis spec proposes work required for nova\u0027s libvirt driver to\nsupport launching instances encrypted using\nArm CCA (Confidential Computing Architecture) technology:\n\nhttps://www.arm.com/architecture/security-features/arm-confidential-compute-architecture\n\nImplements: blueprint arm-cca-libvirt-support\nChange-Id: I12279e15a9e8fe2df6f0fd5f87ddab41eb140be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8158ca9138d339b9c5422a7784687ba26bbf9660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8158ca9138d339b9c5422a7784687ba26bbf9660"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
