)]}'
{"id":"openstack%2Fnova-specs~938823","triplet_id":"openstack%2Fnova-specs~master~I688e8525244c9b6509b0404de680148c9b175381","project":"openstack/nova-specs","branch":"master","topic":"bp/network-id-query","attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-07-03 18:27:40.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-03 18:27:40.000000000","reason":"Vote got outdated and was removed: Code-Review+2"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-07-03 18:27:40.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-07-03 18:27:40.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"8878":{"account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"last_update":"2025-07-03 18:27:28.000000000","reason":"\u003cGERRIT_ACCOUNT_8878\u003e replied on the change","reason_account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}}},"hashtags":[],"change_id":"I688e8525244c9b6509b0404de680148c9b175381","subject":"Add spec for the network id query in server list and server details","status":"NEW","created":"2025-01-09 16:11:05.000000000","updated":"2025-07-03 18:31:57.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":27,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"78d77e3812c953a4b2211b0afc96eb6f2f902a9a","_number":938823,"virtual_id_number":938823,"owner":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-03 18:31:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-09 16:22:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-09 17:26:44.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-06-05 21:18:44.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-06-20 10:35:17.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-06-23 19:58:56.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"68d40dbd8e1813616426c82555ec43304c282eb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-01-09 16:11:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b37a3af11749045a50aab176d9ac7d9beb202d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 16:22:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbf8c61877604d00806e10830d8a5006\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1ac6c5e6256444daa3b6b092e7281ea : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6192b1de34af4e99a090ffd2d9fcaa37 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8c403172951485c860628915906f484e71385b31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-01-09 16:32:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e6c2eaef52b4517ebee002ca22c149741fb715f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 16:49:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54bd807bb3464310a937b6588714c75b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74f0ec3645dd4044a42a2a83073bde00 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f327f2d21f03495ebcdd6c889403c391 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"a3533127319dd031f00cb1ab5525639dc35f17bd","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-09 17:26:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"72c71fc47fdbbf6f1efc3ef37fc423390937d827","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-02-22 02:13:09.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6e95a95b987c1dfc70042f462c5628414a9ac30d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-06-03 08:21:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"349fabd9d8c1a6a7f5ed6237ac576c6524567798","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-06-03 08:24:30.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"443b8954065aaba5d6470c79d1e5406780673372","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-06-03 08:30:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5d5d86c96f4d15c109dc48c666fae3b3b17992a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 09:07:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8e205a5825e4ecdb7df7226f4ddc7c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f6b0cf556ab499abbca82e7b752b945 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b5426fa9474d369aa6e1cb1de57539 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":3},{"id":"7c3a6b8df297d65a7c78b2b2478244fa92e27583","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-03 11:29:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f72caff86c595292ed8418d2d1696ab1ec85c73d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-05 21:18:44.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"845a97653335df9877eb3ffe46b1b5bb0d594a77","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-06-20 10:35:17.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5733b617f6a6c69f83c6f6b8de3725fc9f3e43a6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-20 11:58:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3603aac428e83a08278b5e8a27e3b930ac9f6196","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-23 19:58:56.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3a30027fe78997ed2989b052e02dc62fa208b572","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-30 19:51:40.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3fdb5936a678a09b8958834104591a86e63da541","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-07-03 18:27:28.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4868ac911994f31becb4b2065e6c18da69fe1f20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-07-03 18:27:40.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"78d77e3812c953a4b2211b0afc96eb6f2f902a9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 18:31:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be227de1ce804c559fb3ee053ca17202\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a991ac787447d79f254d3205bf2457 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4185d7b2325b481facb41caa1f0e2f65 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cb5fae3e90b166a6641569a1de912a82babe8cb3","revisions":{"9fd2984f3d0a24b75b8c141e718e8a7cfb97a7ac":{"kind":"REWORK","_number":1,"created":"2025-01-09 16:11:05.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/23/938823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/23/938823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/1"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-01-09 16:06:13.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-01-09 16:06:13.000000000","tz":540},"subject":"Add spec for the network id query in server list and server details","message":"Add spec for the network id query in server list and server details\n\nCloud users want to list servers attached to a spcific project\nnetwork. However, they need to call lots of Show server API\nsince there is no way to fetch server list basing on network\ninformation.\n\nThis spec enables cloud user to get server information by adding\nnetwork id in the list server API and list server detailed API.\n\nAPIImpact\n\nblueprint: network-id-query\nChange-Id: I688e8525244c9b6509b0404de680148c9b175381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9fd2984f3d0a24b75b8c141e718e8a7cfb97a7ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9fd2984f3d0a24b75b8c141e718e8a7cfb97a7ac"}]},"branch":"refs/heads/master"},"853889409450b89bfee9a932b740d70c5b1de0c2":{"kind":"REWORK","_number":2,"created":"2025-01-09 16:32:21.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/23/938823/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/23/938823/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/2"}}},"commit":{"parents":[{"commit":"7b296b40e4883eb14fd55a8701602ee421f98d6d","subject":"Merge \"Add spec for show scheduler hints in server details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7b296b40e4883eb14fd55a8701602ee421f98d6d"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-01-09 16:06:13.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-01-09 16:32:05.000000000","tz":540},"subject":"Add spec for the network id query in server list and server details","message":"Add spec for the network id query in server list and server details\n\nCloud users want to list servers attached to a spcific project\nnetwork. However, they need to call lots of Show server API\nsince there is no way to fetch server list basing on network\ninformation.\n\nThis spec enables cloud user to get server information by adding\nnetwork id in the list server API and list server detailed API.\n\nAPIImpact\n\nblueprint: network-id-query\nChange-Id: I688e8525244c9b6509b0404de680148c9b175381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/853889409450b89bfee9a932b740d70c5b1de0c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/853889409450b89bfee9a932b740d70c5b1de0c2"}]},"branch":"refs/heads/master"},"3fab7d808be8ceed92c6e6be9d04d1b31d56b85f":{"kind":"REWORK","_number":3,"created":"2025-06-03 08:21:43.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/23/938823/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/23/938823/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/3"}}},"commit":{"parents":[{"commit":"4b8f26f46e7c5e8c714516719cff66df519d2d19","subject":"Merge \"Add remove-os-volumes_boot-api spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4b8f26f46e7c5e8c714516719cff66df519d2d19"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-01-09 16:06:13.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-06-03 08:17:50.000000000","tz":540},"subject":"Add spec for the network id query in server list and server details","message":"Add spec for the network id query in server list and server details\n\nCloud users want to list servers attached to a spcific project\nnetwork. However, they need to call lots of Show server API\nsince there is no way to fetch server list basing on network\ninformation.\n\nThis spec enables cloud user to get server information by adding\nnetwork id in the list server API and list server detailed API.\n\nAPIImpact\n\nblueprint: network-id-query\nChange-Id: I688e8525244c9b6509b0404de680148c9b175381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3fab7d808be8ceed92c6e6be9d04d1b31d56b85f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3fab7d808be8ceed92c6e6be9d04d1b31d56b85f"}]},"branch":"refs/heads/master"},"cb5fae3e90b166a6641569a1de912a82babe8cb3":{"kind":"REWORK","_number":4,"created":"2025-07-03 18:27:40.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/23/938823/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/23/938823/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/23/938823/4"}}},"commit":{"parents":[{"commit":"4b8f26f46e7c5e8c714516719cff66df519d2d19","subject":"Merge \"Add remove-os-volumes_boot-api spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4b8f26f46e7c5e8c714516719cff66df519d2d19"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-01-09 16:06:13.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-07-03 18:25:16.000000000","tz":540},"subject":"Add spec for the network id query in server list and server details","message":"Add spec for the network id query in server list and server details\n\nCloud users want to list servers attached to a spcific project\nnetwork. However, they need to call lots of Show server API\nsince there is no way to fetch server list basing on network\ninformation.\n\nThis spec enables cloud user to get server information by adding\nnetwork id in the list server API and list server detailed API.\n\nAPIImpact\n\nblueprint: network-id-query\nChange-Id: I688e8525244c9b6509b0404de680148c9b175381\nSigned-off-by: Masahito Muroi \u003cmasahito.muroi@linecorp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cb5fae3e90b166a6641569a1de912a82babe8cb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cb5fae3e90b166a6641569a1de912a82babe8cb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
