)]}'
{"id":"openstack%2Fnova-specs~938910","triplet_id":"openstack%2Fnova-specs~master~Ia4d13ff6a2aff464b9c97b1c17266b03b64eccd5","project":"openstack/nova-specs","branch":"master","topic":"bp/image-props-in-server-show","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2025-01-13 12:57:41.000000000","reason":"Change was submitted"},"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2025-01-13 12:57:41.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-01-13 12:57:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia4d13ff6a2aff464b9c97b1c17266b03b64eccd5","subject":"Image properties in server show","status":"MERGED","created":"2025-01-10 14:56:26.000000000","updated":"2025-01-13 12:59:04.000000000","submitted":"2025-01-13 12:57:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"938910-bp/image-props-in-server-show","meta_rev_id":"ca05a3f43ae28aedd80a2beca18710deabaf80c6","_number":938910,"virtual_id_number":938910,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-13 12:57:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2025-01-10 21:52:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2025-01-10 22:04:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2025-01-10 21:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-13 12:44:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-13 12:44:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-10 15:12:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 16:28:19.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2025-01-10 16:46:09.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2025-01-10 17:00:16.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-01-10 21:43:43.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2025-01-10 22:04:22.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-01-13 12:44:45.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"9a5b22064e4d4fa7f79e82e4c88ad454746a13f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2025-01-10 14:56:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82cca3278578d36869aafbf72c658910104ad4f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 15:12:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d320265de4d4cd4900ebca0ed52790e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e19fa8a5222c47f69f13c7c403f1001c : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7002a0d7dd8d4cb69f207d8b0685a33a : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f1219fb5d0354b016f2bbcd6522a3ce19706e3c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-10 16:21:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"203b4e159e3494851d471d369b765757f12b4ae9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-10 16:28:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"69ff39bb9d39ee778ff4dba6a03bd852969a0f57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 16:32:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c7b147775fb417e885666ce8688446a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/867672d6b311499dbfe7283d6dc865a1 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78a9b9ac43b840b28060831f17b24ca9 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"6452c793ff5f7189c5cdf40b5e0aac504e181908","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-10 16:46:09.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2769d7277614a877d2f1685b641bdc074feb9d23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-10 16:57:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b8ee29fb6ef7d33eeefb767620c28dbb77a38995","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-10 17:00:16.000000000","message":"Patch Set 3: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f80f79953a1a892619d673d92ee4158593dc7074","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 17:10:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df72ad3512674f158564f269eb9cf069\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9cb670b0bdd4a1ea10c15cb2c7dbd4e : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f48c50a3fa142b3a58bdde497b5bbcd : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":3},{"id":"32218ac7c5c3ee7290916c77d496bbf850664fb6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-10 19:40:41.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"30cd5464d54f5094ad1a9f7ba3ff11b26c2f8a4b","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-01-10 21:43:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7d1cbbf6d21cf20cb48ca6b470362091154730e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 21:44:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"249e5fa7ef6f8e5146d8415f15c5021fc93b1892","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"real_author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-01-10 21:51:15.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_9708\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1\n","accounts_in_message":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"_revision_number":4},{"id":"4327737aa147c1b0143b90c31409cd8a86ad1e8d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-01-10 21:52:31.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f60342b66417786bd5703951a7921691f056a580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 22:00:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53b3d420c5fd4d53b7f4b27fd46f423e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a8c34dd23224823b6512a1e0e1b3539 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4438997f04df44d18bed942ae098b852 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":4},{"id":"e92a687bb828f2fc79709f2a5965d7dc6912d6a3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-10 22:04:22.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"dcf44dc8d9b879a96325137d2e67acdf69aa6c80","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-13 08:33:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a93ff2e8416110b605b57a1a23b6274f8e7c9138","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-01-13 12:44:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"55b2430de4d3479951868faeea89f7c39ea915ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 12:45:05.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"16d91a52cc0790f270cfcdb21b7ddb0eaca85656","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 12:57:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0534afdf971846358010e8e86c7f1ba2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0114b5a0b00149b7a8c137b388ce18d7 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d9e68bf862f4d2cb2d27ea58803122f : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":4},{"id":"9d5f4600d1c96240671a01f0ce5b3f98198dc4b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 12:57:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ca05a3f43ae28aedd80a2beca18710deabaf80c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 12:59:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd5da0a424e745b5bc20d5f3d441c9e4\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/b5e71142cafc49578b7c7816a7416254 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7bbf54e6bf6298c97cc4cf6e524e86ccb2ba7050","revisions":{"43fc7ce78fcc1deb7767742f93e8f67bc0655bbc":{"kind":"REWORK","_number":1,"created":"2025-01-10 14:56:26.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/10/938910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/10/938910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/1"}}},"commit":{"parents":[{"commit":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","subject":"Re-propose config option for behavior of unset unified limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2025-01-10 14:55:40.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2025-01-10 14:56:22.000000000","tz":-300},"subject":"WIP: Image properties in server show","message":"WIP: Image properties in server show\n\nThis spec proposes to show an instance\u0027s embedded image properties in the\nserver show API. This has lots of uses, but is particularly required for `vTPM\nlive migration \u003chttps://review.opendev.org/c/openstack/nova-specs/+/936775\u003e`_\nto show users the vTPM secret security level that is set on their instances.\n\nImplements bp/image-props-in-server-show\nChange-Id: Ia4d13ff6a2aff464b9c97b1c17266b03b64eccd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/43fc7ce78fcc1deb7767742f93e8f67bc0655bbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/43fc7ce78fcc1deb7767742f93e8f67bc0655bbc"}]},"branch":"refs/heads/master"},"7e3ef9b6a3e12ef3732835043c9cdb788b1a8bfa":{"kind":"REWORK","_number":2,"created":"2025-01-10 16:21:50.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/10/938910/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/10/938910/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/2"}}},"commit":{"parents":[{"commit":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","subject":"Re-propose config option for behavior of unset unified limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2025-01-10 14:55:40.000000000","tz":-300},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-01-10 16:21:32.000000000","tz":60},"subject":"Image properties in server show","message":"Image properties in server show\n\nThis spec proposes to show an instance\u0027s embedded image properties in the\nserver show API. This has lots of uses, but is particularly required for `vTPM\nlive migration \u003chttps://review.opendev.org/c/openstack/nova-specs/+/936775\u003e`_\nto show users the vTPM secret security level that is set on their instances.\n\nImplements: bp/image-props-in-server-show\nChange-Id: Ia4d13ff6a2aff464b9c97b1c17266b03b64eccd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7e3ef9b6a3e12ef3732835043c9cdb788b1a8bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7e3ef9b6a3e12ef3732835043c9cdb788b1a8bfa"}]},"branch":"refs/heads/master"},"99be7f9b68fce70af3bf978cbd886d479029f3f5":{"kind":"REWORK","_number":3,"created":"2025-01-10 16:57:41.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/10/938910/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/10/938910/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/3"}}},"commit":{"parents":[{"commit":"6cae04b088a72509b6ea01d2db8d400e2bccd6cc","subject":"Re-propose config option for behavior of unset unified limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cae04b088a72509b6ea01d2db8d400e2bccd6cc"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2025-01-10 14:55:40.000000000","tz":-300},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-01-10 16:57:35.000000000","tz":60},"subject":"Image properties in server show","message":"Image properties in server show\n\nThis spec proposes to show an instance\u0027s embedded image properties in the\nserver show API. This has lots of uses, but is particularly required for `vTPM\nlive migration \u003chttps://review.opendev.org/c/openstack/nova-specs/+/936775\u003e`_\nto show users the vTPM secret security level that is set on their instances.\n\nImplements: bp/image-props-in-server-show\nChange-Id: Ia4d13ff6a2aff464b9c97b1c17266b03b64eccd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/99be7f9b68fce70af3bf978cbd886d479029f3f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/99be7f9b68fce70af3bf978cbd886d479029f3f5"}]},"branch":"refs/heads/master"},"7bbf54e6bf6298c97cc4cf6e524e86ccb2ba7050":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-01-10 21:51:15.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"real_uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/10/938910/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/10/938910/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/10/938910/4"}}},"commit":{"parents":[{"commit":"b7fe7d8d25cf9bbbd58b33815e337164d50e148b","subject":"Merge \"Migrate VFIO devices using kernel variant drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b7fe7d8d25cf9bbbd58b33815e337164d50e148b"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2025-01-10 14:55:40.000000000","tz":-300},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-01-10 21:51:15.000000000","tz":0},"subject":"Image properties in server show","message":"Image properties in server show\n\nThis spec proposes to show an instance\u0027s embedded image properties in the\nserver show API. This has lots of uses, but is particularly required for `vTPM\nlive migration \u003chttps://review.opendev.org/c/openstack/nova-specs/+/936775\u003e`_\nto show users the vTPM secret security level that is set on their instances.\n\nImplements: bp/image-props-in-server-show\nChange-Id: Ia4d13ff6a2aff464b9c97b1c17266b03b64eccd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7bbf54e6bf6298c97cc4cf6e524e86ccb2ba7050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7bbf54e6bf6298c97cc4cf6e524e86ccb2ba7050"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
