)]}'
{"id":"openstack%2Fnova-specs~939190","triplet_id":"openstack%2Fnova-specs~master~If10eab618275f266e6207597dc880cb9bb1dd2b0","project":"openstack/nova-specs","branch":"master","topic":"bp/instance-metadata-protection","attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-11-17 10:46:14.000000000","reason":"\u003cGERRIT_ACCOUNT_26250\u003e replied on the change","reason_account":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"}},"36776":{"account":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"last_update":"2025-01-14 02:15:24.000000000","reason":"A robot voted negatively on a label"},"2271":{"account":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"last_update":"2025-04-07 10:46:05.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If10eab618275f266e6207597dc880cb9bb1dd2b0","subject":"Repropose instance metadata protection","status":"NEW","created":"2025-01-14 01:55:56.000000000","updated":"2026-03-10 10:52:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":8,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"8808172e8dfd18d2c2f4e0b6b61ddeaa66428411","_number":939190,"virtual_id_number":939190,"owner":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-14 02:15:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"all":[{"value":-1,"date":"2025-01-14 12:04:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-04-07 09:23:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-14 02:15:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-14 12:04:36.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-04-07 09:23:37.000000000","updated_by":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"reviewer":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"state":"REVIEWER"},{"updated":"2025-11-17 10:46:14.000000000","updated_by":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"reviewer":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"state":"CC"},{"updated":"2026-03-10 10:52:44.000000000","updated_by":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"CC"}],"messages":[{"id":"99cad8af18a54f078822ba5a7820ed6de5e799d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"date":"2025-01-14 01:55:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3f9f9448715df8bee702266dff81f78ff0f9176","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"date":"2025-01-14 01:59:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ca94573bf23d055153a1466451bb965324aada29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 02:15:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aadf4f4c83314b71935d045527cfd2de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8291c06a5f5a48909e2ea77ad33c160e : FAILURE in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aeed9ba589b453b8c913f372edc360a : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"06080c4e9c4fca347c8a1f86234d571bbe4c27b7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-14 12:04:36.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c374c11e4aa62af61cfb410794ae35c51bf6b92d","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-04-07 09:23:37.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e323e6a73b640b2264eb06084bca84e20d5c6d66","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-07 10:46:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aec74564a73e9a71f1bd624e6243ea45cfc5e48d","author":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"date":"2025-11-17 10:46:14.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8808172e8dfd18d2c2f4e0b6b61ddeaa66428411","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-03-10 10:52:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"56aee56d3e60204d17a6721ad6b10e31bafe3a61","revisions":{"bc55315a06ffeb141b06cf71c5438bb6d0470f9d":{"kind":"REWORK","_number":1,"created":"2025-01-14 01:55:56.000000000","uploader":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"ref":"refs/changes/90/939190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/90/939190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/90/939190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/90/939190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/90/939190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/90/939190/1"}}},"commit":{"parents":[{"commit":"a422b51b992716de61ad3d8db91feff5dc23d82a","subject":"Merge \"Repropose spice-direct console support.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a422b51b992716de61ad3d8db91feff5dc23d82a"}]}],"author":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2025-01-14 01:51:46.000000000","tz":540},"committer":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2025-01-14 01:55:53.000000000","tz":540},"subject":"Repropose instance metadata protection","message":"Repropose instance metadata protection\n\nRelated: blueprint instance-metadata-protection\nChange-Id: If10eab618275f266e6207597dc880cb9bb1dd2b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bc55315a06ffeb141b06cf71c5438bb6d0470f9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bc55315a06ffeb141b06cf71c5438bb6d0470f9d"}]},"branch":"refs/heads/master"},"56aee56d3e60204d17a6721ad6b10e31bafe3a61":{"kind":"REWORK","_number":2,"created":"2025-01-14 01:59:46.000000000","uploader":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"ref":"refs/changes/90/939190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/90/939190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/90/939190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/90/939190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/90/939190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/90/939190/2"}}},"commit":{"parents":[{"commit":"a422b51b992716de61ad3d8db91feff5dc23d82a","subject":"Merge \"Repropose spice-direct console support.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a422b51b992716de61ad3d8db91feff5dc23d82a"}]}],"author":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2025-01-14 01:51:46.000000000","tz":540},"committer":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2025-01-14 01:59:30.000000000","tz":540},"subject":"Repropose instance metadata protection","message":"Repropose instance metadata protection\n\nRelated: blueprint instance-metadata-protection\nChange-Id: If10eab618275f266e6207597dc880cb9bb1dd2b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/56aee56d3e60204d17a6721ad6b10e31bafe3a61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/56aee56d3e60204d17a6721ad6b10e31bafe3a61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
