)]}'
{"id":"openstack%2Fnova-specs~940440","triplet_id":"openstack%2Fnova-specs~master~I1716921c16e56f8fae00bccb825220ab6d410c51","project":"openstack/nova-specs","branch":"master","topic":"bp/flavor-search-by-name","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-06-20 16:01:16.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-02 18:50:54.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-07-02 18:50:54.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-06-20 18:07:10.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2025-07-02 18:50:54.000000000","reason":"Change was submitted"},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2025-07-02 18:50:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1716921c16e56f8fae00bccb825220ab6d410c51","subject":"Add flavor-search-by-name spec","status":"MERGED","created":"2025-01-30 12:07:57.000000000","updated":"2025-07-02 18:52:04.000000000","submitted":"2025-07-02 18:50:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":64,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940440-bp/flavor-search-by-name","meta_rev_id":"67132965dd4678f4bfd52ba3c4e6114a97f4ba39","_number":940440,"virtual_id_number":940440,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-02 18:50:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2025-06-20 16:01:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2025-06-20 18:07:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2025-07-02 18:39:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-30 12:20:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-31 12:18:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-01-31 12:18:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-05-14 09:39:48.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2025-05-29 20:07:11.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2025-06-17 16:03:56.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"886611f96cf66447e417d5768d77c787f9d467ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-30 12:07:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a573d8f6e0847784afd5515e97744acc2b86f8b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 12:20:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/785700198a6b4e99b5a962b9ab29d826\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36eca40ab99450ebe4b175d2485659e : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2620a2bfc384250a3ca7f774325c0dd : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2394a74651b6e57ba53f7d8cd979901e3767ef2f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-31 12:18:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6426738ea82c05cac9da644c83797a55f57af15","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-14 10:54:26.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e883274994149f4d63efd99d889e251a8a3a4d4b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-15 10:32:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"622ee19970ca4721a1871ea1647f76135eaa4200","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-15 11:00:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a925f449b2747dc90635ae4b1ec3eaa0331a80ec","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-15 11:04:54.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1988c777ea4efba87e9a0caad0d29aec2e11ce9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 11:05:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36d18746b2f745b0b0cd04bc262f6a91\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92804dc2aeab403f9e7587589e106947 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d2c47907c124ad79e0298820ef6d0b2 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":2},{"id":"ebc1faf6a79695e6f644624bb94e16e13b4b2afb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-15 11:10:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1f3581b3bfbbded93c8aabc6d99a782c8a0ffec9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-15 11:11:44.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"15ea1f122687f2de45280cc121ee7512850ddb02","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-15 11:13:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ac54a5388c9be2ce8d9ea458058093df6a8ef7e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 11:16:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d43195e6003740d1b859b6d373608e32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e30d37f664aa4cecbeab46dcb613608a : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e48d1f66ea384eada363847d7b15aa37 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":3},{"id":"70a91e19ce45ce6adc9548528ff74e3a2bbc736c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-15 11:50:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e9539fc6870d691e3d727310ec302c23d48d4cb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-15 11:50:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ef6f13eb4500428494aeb00b89134b9eb8772db1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-15 11:53:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8eb4b1747f07c55165b85c6213e722478aba33c7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-15 11:59:33.000000000","message":"Patch Set 4: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b45c4d1d226eefcd88606b0545d4fce223bf162e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 12:00:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92eaa7e8e8d541dfb2b5c61bc1d80069\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a73eb642a3804384b739e13194b04c6b : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f1f8cc3b712461eb3302d95346e781f : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":5},{"id":"92ce4ce2ba74a3de4ca5afdec0476944a8f9f5ed","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-16 09:54:30.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"480ce8afa808c7fd94332bc9e53e8f48fed4a2c6","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-05-14 09:39:48.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"056329f4b5d8728a8b1bb6c770a70a18a4611946","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-19 13:00:24.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"ce532606de7debf3f478bde5986201333d9f35d9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-19 13:00:30.000000000","message":"Topic flavor-search-by-name removed","accounts_in_message":[],"_revision_number":6},{"id":"be56a080de3b3a27d9423d4a4f7dece04bc0650b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-19 13:00:33.000000000","message":"Topic set to bp/flavor-search-by-name","accounts_in_message":[],"_revision_number":6},{"id":"89f1d0e2a0f44c7eeea52960b72abd58542f950a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 13:19:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3d869c5a5384dd18e807f5a6b371c92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa18aa23b30d4e75b87f8845e9072911 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e85a74b049494c25ae227dc1a27626bf : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":6},{"id":"5fac094e1c31be3cb5eda97c74022813fc2bfa67","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-29 16:05:14.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"89fb46230b99405157f0948e3546539d49579598","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-29 20:07:11.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7eb9180d36762e5521960e39ceca843ab7e1ac04","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-05 13:16:33.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d2d74cb61179dc99793b412466d29e9ca55c0563","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-05 17:32:06.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"36b8228a08129d7f15c31783770458be4d4e6e4c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-05 17:45:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b9d323b555c48e23e0f12c2320af0657a33089aa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-05 21:16:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"41157639c7646fa028ebe91fbb2ce2f12c422fb1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-09 16:57:53.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c592f8454b8cc8cfacb64bfbb1ea98873ff60754","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-10 16:23:53.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"d3cb238e59c5028b4741b7fa2b55f6e37e62cf3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 16:33:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/309d13f07b674870b4e58f47926978b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49c37baa103549a7ab74a4b946125777 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/700a9369ae044bc8bdf1ebcc9135625b : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":7},{"id":"f7bd409096bd8952ed0154818b8e05e66cb0e5d6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-17 14:37:07.000000000","message":"Patch Set 7: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"dceedeec9d146fc759520f5567171bcccbbeea95","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-06-17 16:03:56.000000000","message":"Patch Set 7: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":7},{"id":"172987574f82629efbdc31e2eed2373bf6d57a9b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-17 18:03:59.000000000","message":"Patch Set 7: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c0d785044f542969822a592ed75e8379117a3c73","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-17 18:40:20.000000000","message":"Patch Set 7: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"91d65385426b3f83f56c62b5ed4fb26f66d11877","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-06-20 10:19:39.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c8d925d4df65e48a3b49548c5ac3fd50cb199633","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-06-20 11:04:47.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1c5f981a030d3d438505b202135f05646fcce080","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-20 11:33:47.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ed6d5d317e83d0ac6db1817b7dd2a763fa447b95","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-20 13:57:16.000000000","message":"Patch Set 7:\n\n(8 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b630db304333094b28eb2677d0e2fb4ec4de2b7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-20 13:57:44.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"3c5c9312e28fc858af83082f556f276a51a48153","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 14:04:31.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8121b9a3fe6c430198d25f406893bf9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/281e934e55884e64b8edef71f4aa035d : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb648661c88412daba1a5c5f7e09efa : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":8},{"id":"c477fcc9e0cb81f73827974d83e30d1665a500a9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-20 16:01:16.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"fbbba16d19d5bf025bcc50e1f87f63070bb12857","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-20 18:07:10.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"07a5cd46d8d42ca8fd7eb41c95df2056cde78d0f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-02 12:16:14.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"57aa417c32ce583fe0a4c70c2f8f5506d349a1f2","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-02 18:39:57.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"209392c54b4f471cda50fa4272cdf3c7d2551fe4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 18:40:07.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"33ecebd866b346a34267cef975d92569dca12b7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 18:50:54.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be7178be142049acb1fa32dc5c98e030\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d08489858fcb4f979b37d717961712b6 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dee411f5d4dc4308adc1907b83d6dea0 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":8},{"id":"259d168e7efce6762e67b98bbb8224a878809905","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 18:50:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"67132965dd4678f4bfd52ba3c4e6114a97f4ba39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 18:52:04.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14a9d0e22e924dfea7f8347fe50e0821\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/1ee079744c324cd7910d019ca3fbcd34 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"33cbcb66184235f778b266a398a5b3d4e09191e6","revisions":{"825f1b68a31da7cc9df8dfe65bf6884e462bf50b":{"kind":"REWORK","_number":1,"created":"2025-01-30 12:07:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/940440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/40/940440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/1"}}},"commit":{"parents":[{"commit":"ee3b8dd2b1509dece7351488398dff879495403e","subject":"Fix a line that was not removed before merging the specification","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee3b8dd2b1509dece7351488398dff879495403e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-01-30 12:07:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-01-30 12:07:51.000000000","tz":0},"subject":"Add flavor-search-by-name spec","message":"Add flavor-search-by-name spec\n\nAllow filtering of flavors by name.\n\nChange-Id: I1716921c16e56f8fae00bccb825220ab6d410c51\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/825f1b68a31da7cc9df8dfe65bf6884e462bf50b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/825f1b68a31da7cc9df8dfe65bf6884e462bf50b"}]},"branch":"refs/heads/master"},"864cb27966ee37e9a21d275e9b7f254840aa0e04":{"kind":"REWORK","_number":2,"created":"2025-04-15 11:00:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/940440/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/40/940440/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/2"}}},"commit":{"parents":[{"commit":"2b660022a0eba45416eed9030191fbc62b617b90","subject":"Merge \"tox: Drop envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b660022a0eba45416eed9030191fbc62b617b90"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-01-30 12:07:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-04-15 10:59:20.000000000","tz":60},"subject":"Add flavor-search-by-name spec","message":"Add flavor-search-by-name spec\n\nAllow filtering of flavors by name. Take the opportunity to address\nother issues with the API while we\u0027re here.\n\nChange-Id: I1716921c16e56f8fae00bccb825220ab6d410c51\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/864cb27966ee37e9a21d275e9b7f254840aa0e04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/864cb27966ee37e9a21d275e9b7f254840aa0e04"}]},"branch":"refs/heads/master"},"74f933eaff90c522a68e80ce347226a4f9feab49":{"kind":"REWORK","_number":3,"created":"2025-04-15 11:10:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/940440/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/40/940440/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/3"}}},"commit":{"parents":[{"commit":"2b660022a0eba45416eed9030191fbc62b617b90","subject":"Merge \"tox: Drop envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b660022a0eba45416eed9030191fbc62b617b90"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-01-30 12:07:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-04-15 11:10:16.000000000","tz":60},"subject":"Add flavor-search-by-name spec","message":"Add flavor-search-by-name spec\n\nAllow filtering of flavors by name. Take the opportunity to address\nother issues with the API while we\u0027re here.\n\nChange-Id: I1716921c16e56f8fae00bccb825220ab6d410c51\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/74f933eaff90c522a68e80ce347226a4f9feab49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/74f933eaff90c522a68e80ce347226a4f9feab49"}]},"branch":"refs/heads/master"},"0ffd37d85ef3d5924cf03d3d93f39cdc13f4da32":{"kind":"REWORK","_number":4,"created":"2025-04-15 11:50:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/940440/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/40/940440/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/4"}}},"commit":{"parents":[{"commit":"2b660022a0eba45416eed9030191fbc62b617b90","subject":"Merge \"tox: Drop envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b660022a0eba45416eed9030191fbc62b617b90"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-01-30 12:07:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-04-15 11:50:40.000000000","tz":60},"subject":"Add flavor-search-by-name spec","message":"Add flavor-search-by-name spec\n\nAllow filtering of flavors by name. Take the opportunity to address\nother issues with the API while we\u0027re here.\n\nChange-Id: I1716921c16e56f8fae00bccb825220ab6d410c51\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0ffd37d85ef3d5924cf03d3d93f39cdc13f4da32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0ffd37d85ef3d5924cf03d3d93f39cdc13f4da32"}]},"branch":"refs/heads/master"},"79b7e76ca17f5bdd94b8706522f55c1fb998098f":{"kind":"REWORK","_number":5,"created":"2025-04-15 11:53:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/940440/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/40/940440/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/5"}}},"commit":{"parents":[{"commit":"2b660022a0eba45416eed9030191fbc62b617b90","subject":"Merge \"tox: Drop envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b660022a0eba45416eed9030191fbc62b617b90"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-01-30 12:07:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-04-15 11:53:44.000000000","tz":60},"subject":"Add flavor-search-by-name spec","message":"Add flavor-search-by-name spec\n\nAllow filtering of flavors by name. Take the opportunity to address\nother issues with the API while we\u0027re here.\n\nChange-Id: I1716921c16e56f8fae00bccb825220ab6d410c51\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/79b7e76ca17f5bdd94b8706522f55c1fb998098f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/79b7e76ca17f5bdd94b8706522f55c1fb998098f"}]},"branch":"refs/heads/master"},"012cc198c64e5a60b71cd3da55748701d2d24747":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-05-19 13:00:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/940440/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/40/940440/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/6"}}},"commit":{"parents":[{"commit":"2b660022a0eba45416eed9030191fbc62b617b90","subject":"Merge \"tox: Drop envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b660022a0eba45416eed9030191fbc62b617b90"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-01-30 12:07:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-19 13:00:24.000000000","tz":0},"subject":"Add flavor-search-by-name spec","message":"Add flavor-search-by-name spec\n\nAllow filtering of flavors by name. Take the opportunity to address\nother issues with the API while we\u0027re here.\n\nChange-Id: I1716921c16e56f8fae00bccb825220ab6d410c51\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nBlueprint: flavor-search-by-name\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/012cc198c64e5a60b71cd3da55748701d2d24747"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/012cc198c64e5a60b71cd3da55748701d2d24747"}]},"branch":"refs/heads/master","description":"Edit commit message"},"32306441c90ae158f2d147dab2cbb30267477266":{"kind":"REWORK","_number":7,"created":"2025-06-10 16:23:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/940440/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/40/940440/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/7"}}},"commit":{"parents":[{"commit":"2b660022a0eba45416eed9030191fbc62b617b90","subject":"Merge \"tox: Drop envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b660022a0eba45416eed9030191fbc62b617b90"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-01-30 12:07:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-10 16:23:45.000000000","tz":60},"subject":"Add flavor-search-by-name spec","message":"Add flavor-search-by-name spec\n\nAllow filtering of flavors by name. Take the opportunity to address\nother issues with the API while we\u0027re here.\n\nChange-Id: I1716921c16e56f8fae00bccb825220ab6d410c51\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nBlueprint: flavor-search-by-name\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/32306441c90ae158f2d147dab2cbb30267477266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/32306441c90ae158f2d147dab2cbb30267477266"}]},"branch":"refs/heads/master"},"33cbcb66184235f778b266a398a5b3d4e09191e6":{"kind":"REWORK","_number":8,"created":"2025-06-20 13:57:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/940440/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/40/940440/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/40/940440/8"}}},"commit":{"parents":[{"commit":"2aaeef5f9ffdabdcdd5b78f79961289ef38b2e72","subject":"Merge \"Re-propose \"libvirt: AMD SEV-ES support\" for 2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2aaeef5f9ffdabdcdd5b78f79961289ef38b2e72"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-01-30 12:07:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-20 13:57:36.000000000","tz":60},"subject":"Add flavor-search-by-name spec","message":"Add flavor-search-by-name spec\n\nAllow filtering of flavors by name. Take the opportunity to address\nother issues with the API while we\u0027re here.\n\nChange-Id: I1716921c16e56f8fae00bccb825220ab6d410c51\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nBlueprint: flavor-search-by-name\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/33cbcb66184235f778b266a398a5b3d4e09191e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/33cbcb66184235f778b266a398a5b3d4e09191e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
