)]}'
{"id":"openstack%2Fnova-specs~947542","triplet_id":"openstack%2Fnova-specs~master~Iaa6ddbbc583195d51df417d5e276e7e7832c3fc9","project":"openstack/nova-specs","branch":"master","topic":"bp/vtpm-live-migration","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2025-05-29 16:11:05.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-05-22 15:55:45.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2025-05-22 15:58:20.000000000","reason":"\u003cGERRIT_ACCOUNT_16207\u003e replied on the change","reason_account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}}},"hashtags":[],"change_id":"Iaa6ddbbc583195d51df417d5e276e7e7832c3fc9","subject":"Re-propose vTPM live migration","status":"MERGED","created":"2025-04-16 22:45:21.000000000","updated":"2025-05-29 16:12:45.000000000","submitted":"2025-05-29 16:11:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"947542-bp/vtpm-live-migration","meta_rev_id":"e2440aa9f8afe8adb89c3db96bb54184b38022d9","_number":947542,"virtual_id_number":947542,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-29 16:11:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2025-05-29 16:03:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-22 15:55:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2025-05-22 15:58:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2025-05-29 16:03:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2025-05-22 15:58:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-16 22:57:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-14 09:16:12.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2025-05-22 11:24:47.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-05-29 16:03:03.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"d1b578e296717d094a0883242e4a2f8f1aa42447","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-04-16 22:45:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c913a866c95d56bde4dcfe39f2b85c1bcc947bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 22:57:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e47f568cde934f1180965ec1b6ba58d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cfc176322f74c14a350ac23ff144bc8 : FAILURE in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e595131f9da4dbc9b4414ca570c3ad7 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4313b07e2dbc7f95c1cd0ef8a6027b3a82784ee1","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-04-16 23:06:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"405e74ab099582e9980aeaec19878f44152310be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-05-01 00:39:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5b94a389c913a9f18003e3bf3ac814632f817e0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 00:48:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bc8aee7182746b2af53a9c63313fd41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a972614cd1b7459fb5cc19919bd2007b : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fff6f14e17134e3fafc5665a4fc37e8f : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"5b681641a57eb6a0adccd3569ae159290297b9d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-05-01 22:09:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ed4c543b8781ebd66cd2c92bbe462f853db20d76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 22:16:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1f940b87def4a1c9129b2170ef6018d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7e3afcb8d834e968a72cc4882395f67 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d661568343c4b5f85aa9269dc7b41ee : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":3},{"id":"720af212241b21e26e486eedf9f7eb9337216721","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-05-14 09:16:12.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f2b2e20fc9a768d573f4c08935500cc403cd8e6d","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-05-20 13:25:50.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"a2550447887045ea4c89aed28e32a99ce0d35438","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-05-22 11:24:47.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b060a3cb05e4d9269136564d0a0bb8e57dcb4ca8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-05-22 15:23:02.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2b442492f39000e9357486db29184d6ec51e4287","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 15:34:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95c25308dcb3415d9c085ca8563ba90d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b52e7fed39354e5e867da1f606f93820 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f7bc0d89375400cbc8518a39719ee51 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":4},{"id":"f82f6001fda1412b64cea31293096986373f1619","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-05-22 15:55:45.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a7fc58aa0ea3603cb934ca7936e9f6301816076f","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-05-22 15:58:20.000000000","message":"Patch Set 4: Code-Review+1 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"d371ee8b20b68ac0fcb68b61d0299839cf7fd30f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-29 16:03:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6df6b287fbd12c0c0fe8f382c782037090fa4ba3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 16:03:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f6bb24db2f07e21f215734b698378f7d91100a6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 16:11:05.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bc663a2f81943e19cda0b4768e20e0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d4365c15b3f4e35a7ac7bf7e23eaaa2 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6898e240ec4434facc1f7488e07ec7e : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":4},{"id":"d9ed4a95d6c7135ed15082c6999bd4c02bd26c1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 16:11:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e2440aa9f8afe8adb89c3db96bb54184b38022d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 16:12:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/405fa340523045d5b2d9dda7966443e3\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/5de2135d8e3e45808669b115ff8bcf45 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7fcbff551658b5b1940fc5e0593f50a2a8907a35","revisions":{"897686e0ad91320a1d45b0902ebd071121bd66dd":{"kind":"REWORK","_number":1,"created":"2025-04-16 22:45:21.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/42/947542/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/42/947542/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/1"}}},"commit":{"parents":[{"commit":"2b660022a0eba45416eed9030191fbc62b617b90","subject":"Merge \"tox: Drop envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b660022a0eba45416eed9030191fbc62b617b90"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2025-04-16 22:42:24.000000000","tz":-420},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2025-04-16 22:42:55.000000000","tz":-420},"subject":"Re-propose vTPM live migration","message":"Re-propose vTPM live migration\n\nPreviously-approved: 2025.1\n\nRelated to blueprint vtpm-live-migration\n\nChange-Id: Iaa6ddbbc583195d51df417d5e276e7e7832c3fc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/897686e0ad91320a1d45b0902ebd071121bd66dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/897686e0ad91320a1d45b0902ebd071121bd66dd"}]},"branch":"refs/heads/master"},"65c2e7c29deb6d57ba61b3ad05b7df7942b49c04":{"kind":"REWORK","_number":2,"created":"2025-05-01 00:39:26.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/42/947542/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/42/947542/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/2"}}},"commit":{"parents":[{"commit":"2b660022a0eba45416eed9030191fbc62b617b90","subject":"Merge \"tox: Drop envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b660022a0eba45416eed9030191fbc62b617b90"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2025-04-16 22:42:24.000000000","tz":-420},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2025-05-01 00:36:19.000000000","tz":-420},"subject":"Re-propose vTPM live migration","message":"Re-propose vTPM live migration\n\nPreviously-approved: 2025.1\n\nChanges from the previously approved version:\n\n* Changed :ref:`problem-description` to\n  :ref:`vtpm-live-migration-2025.2-problem-description` to prevent\n  Sphinx error:\n    \"WARNING: duplicate label problem-description, other instance in\n     doc/source/specs/2025.1/approved/vtpm-live-migration.rst\"\n* Typos fixed where noticed\n* Updated config options names s/_vtpm/_tpm/ to reflect agreement\n  during implementation to align with existing TPM image properties\n* Added a new section describing the security policy user confirmation\n  mechanism\n\nRelated to blueprint vtpm-live-migration\n\nChange-Id: Iaa6ddbbc583195d51df417d5e276e7e7832c3fc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/65c2e7c29deb6d57ba61b3ad05b7df7942b49c04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/65c2e7c29deb6d57ba61b3ad05b7df7942b49c04"}]},"branch":"refs/heads/master"},"dc39d0f679751e191c71f038638836599bf03c41":{"kind":"REWORK","_number":3,"created":"2025-05-01 22:09:56.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/42/947542/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/42/947542/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/3"}}},"commit":{"parents":[{"commit":"2b660022a0eba45416eed9030191fbc62b617b90","subject":"Merge \"tox: Drop envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b660022a0eba45416eed9030191fbc62b617b90"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2025-04-16 22:42:24.000000000","tz":-420},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2025-05-01 22:09:00.000000000","tz":-420},"subject":"Re-propose vTPM live migration","message":"Re-propose vTPM live migration\n\nPreviously-approved: 2025.1\n\nChanges from the previously approved version:\n\n* Changed :ref:`problem-description` to\n  :ref:`vtpm-live-migration-2025.2-problem-description` to prevent\n  Sphinx error:\n    \"WARNING: duplicate label problem-description, other instance in\n     doc/source/specs/2025.1/approved/vtpm-live-migration.rst\"\n* Typos fixed where noticed\n* Updated config options names s/_vtpm/_tpm/ to reflect agreement\n  during implementation to align with existing TPM image properties\n* Updated config options from [compute] to [libvirt] to reflect\n  agreement during implementation\n* Added a new section describing the security policy user confirmation\n  mechanism\n\nRelated to blueprint vtpm-live-migration\n\nChange-Id: Iaa6ddbbc583195d51df417d5e276e7e7832c3fc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dc39d0f679751e191c71f038638836599bf03c41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dc39d0f679751e191c71f038638836599bf03c41"}]},"branch":"refs/heads/master"},"7fcbff551658b5b1940fc5e0593f50a2a8907a35":{"kind":"REWORK","_number":4,"created":"2025-05-22 15:23:02.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/42/947542/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/42/947542/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/42/947542/4"}}},"commit":{"parents":[{"commit":"2b660022a0eba45416eed9030191fbc62b617b90","subject":"Merge \"tox: Drop envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b660022a0eba45416eed9030191fbc62b617b90"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2025-04-16 22:42:24.000000000","tz":-420},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2025-05-22 15:23:01.000000000","tz":0},"subject":"Re-propose vTPM live migration","message":"Re-propose vTPM live migration\n\nPreviously-approved: 2025.1\n\nChanges from the previously approved version:\n\n* Changed :ref:`problem-description` to\n  :ref:`vtpm-live-migration-2025.2-problem-description` to prevent\n  Sphinx error:\n    \"WARNING: duplicate label problem-description, other instance in\n     doc/source/specs/2025.1/approved/vtpm-live-migration.rst\"\n* Typos fixed where noticed\n* Updated config options names s/_vtpm/_tpm/ to reflect agreement\n  during implementation to align with existing TPM image properties\n* Updated config options from [compute] to [libvirt] to reflect\n  agreement during implementation\n* Added a new section describing the security policy user confirmation\n  mechanism\n\nRelated to blueprint vtpm-live-migration\n\nChange-Id: Iaa6ddbbc583195d51df417d5e276e7e7832c3fc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7fcbff551658b5b1940fc5e0593f50a2a8907a35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7fcbff551658b5b1940fc5e0593f50a2a8907a35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
