)]}'
{"id":"openstack%2Fnova-specs~951636","triplet_id":"openstack%2Fnova-specs~master~If0a3f26fe3efa9c64cba5f743a8e8cbaae78a9bc","project":"openstack/nova-specs","branch":"master","topic":"bp/add-virtio-blk-iothreads","attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-10-16 08:39:58.000000000","reason":"\u003cGERRIT_ACCOUNT_8878\u003e replied on the change","reason_account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-10-15 08:26:36.000000000","reason":"\u003cGERRIT_ACCOUNT_8878\u003e replied on the change","reason_account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-07-03 16:50:55.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"8878":{"account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"last_update":"2026-01-28 16:34:13.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2025-07-03 16:50:55.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-01-28 16:34:13.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2026-01-28 17:02:22.000000000","reason":"\u003cGERRIT_ACCOUNT_16207\u003e replied on the change","reason_account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}}},"hashtags":[],"change_id":"If0a3f26fe3efa9c64cba5f743a8e8cbaae78a9bc","subject":"Add spec for virtio-blk multiqueue extra_spec","status":"NEW","created":"2025-06-03 09:18:59.000000000","updated":"2026-01-28 17:27:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":89,"unresolved_comment_count":12,"has_review_started":true,"meta_rev_id":"a5953242ae0b290b8415db739ef587dd95e7cdec","_number":951636,"virtual_id_number":951636,"owner":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-16 09:46:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2025-12-17 10:15:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-03 09:26:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-20 16:27:29.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-07-02 10:11:42.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2025-07-02 10:31:23.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-07-02 16:40:33.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2025-07-25 09:33:52.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-09-26 13:08:06.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2025-12-17 03:20:52.000000000","updated_by":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"reviewer":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"state":"CC"}],"messages":[{"id":"90edaeca1357d1ebb9ec00d99b2facaa76d53e7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-06-03 09:18:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6a6fa1b567c4cfa278759c20668a28edeafe31b","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-06-03 09:21:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca01a1736422b76ef08e23cc1da47f53057c97f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 09:26:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/059ec0ef354a43cbb6b471d884089678\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0bd594f97a7475a8604a11f94c62eca : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c8112ae3d0e42c3bceaaacf00cf3221 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f718c0c9d2361dd062330ef632564d8376a6bfba","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-20 16:27:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":1},{"id":"839b67728f1657c5946b1241c527928f2079f216","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-07-02 09:12:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5316b08716dce338a3ccdff749493526fb9c04eb","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-07-02 09:13:09.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0dfc9f5fdd085bb1510173f58863207b9bad4cff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 09:24:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20eb6da7a5f640c8b933994aff73a622\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/986ec7b0a7014a7b8f37c04be674b2f0 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b347b666264473c83acd9cd2c7b3d0f : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":2},{"id":"b5d4cb97d3b687fb6aa07c6181121ca7306d01aa","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-07-02 10:11:42.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9ea76e3e87c16501529c9897817858a325398d2f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-07-02 10:31:23.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b01e902528035f129faecfec84729b9145d55613","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-02 11:15:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cef3062e7b16379a66d8ee1760bc02a2aeecbf7e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-02 11:39:50.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9b7b5e3eca5ba760cc7e06b57fe65c1ec7ac79a7","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-07-02 16:40:33.000000000","message":"Patch Set 2: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"91d8e105b402437f28db4a536a31c9d0bb87e1d0","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-07-03 07:58:46.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2673020eb89fbe1f5bfeee50c19fa696d1c63f99","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-03 10:48:27.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9a4946ab4fd2655d29cae047a78755ba103bd4d6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-03 10:49:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"feded73cd51aa2b2348fbd76c1f838f4a2099347","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-07-03 16:50:28.000000000","message":"Patch Set 2:\n\n(12 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d03f4146a40b4965ba9891dca3ce0551dfda58d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-07-03 16:50:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e4c74905b9ea4ded07dafe3b5e37f8a78346f31c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 16:56:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37b9e304ddf44f1bbf19ef76f61611aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc767c727f34c5796381daa95674239 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/299a38aef08e4c9083af1ebd3a1f2ffc : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":3},{"id":"39d92d1a40d37dc65519d20585ed1046d3c4ae8b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-03 16:58:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"34a35d18096117023df2bef256e27b93e265d937","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-03 17:11:34.000000000","message":"Patch Set 3: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"88e5a1800282412c58f965e2f6b5bc15f282725a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-03 17:14:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c52e964f69c138a5a384c345c679cd8118271f00","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-25 09:33:52.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a5a14ded0521df36a5f8e7095f78da279c765e12","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-26 13:08:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"85b7def89165a77c888cf3680129cb4a0ecf2118","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-10-15 08:25:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"877cb1f3431b262da675605bbb4dc836756a957a","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-10-15 08:26:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2f58bd55403b1204b642e6896744352b561d8f18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-10-16 08:38:58.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7ed4ecfc3e05d97f341ece7a7f4053386e4592f9","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-10-16 08:39:58.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4710f898ac7e8a54f259cc0bd017b4ebe2557289","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 08:44:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/172bfc105d084c97b6d74db5fb77f09f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7094136da8bb48ed9cd0e8dca9652e50 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af9dad05559340f093085ca03e52bcc0 : FAILURE in 1m 48s","accounts_in_message":[],"_revision_number":4},{"id":"9dfa20720339d5300c20416297a8d4cc82d1907b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-10-16 09:37:24.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e47c96ab93500d1b4c5766b176408f93cd94165a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 09:46:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/546a028d4946487c83bc75db1ddca965\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6466acc213442878c9060a2e5d4df21 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a461fdf248945679b1af59ded34ea55 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":5},{"id":"24d81dd85649668f85c4f44ab48adefcb237a55f","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2025-12-17 03:20:52.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"dc5beb3fcfb4e105074e0d4139f329c78d5d729b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-12-17 10:15:57.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"6371c6a507331992c17b0b122748b65a69e22ebd","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2025-12-18 01:57:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a19b7d73130054740e6343c750d32e0e0ca881a0","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2026-01-21 15:52:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8eb9359c310010f0df3d7d5ff526e40bc3f9be72","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-01-21 16:52:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d30b38a4ea262c1986cc9cf83d3152a1c2bb0c18","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-01-21 18:06:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ad56af13e3ece9ef91206f3506561e565682151e","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-01-28 16:26:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"73a2d9bdb392d9b50366f02d4846f898df8ff736","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-01-28 16:34:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f65028a9cc66ada2e8e02a2a920ed9b930d2e529","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2026-01-28 17:02:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a5953242ae0b290b8415db739ef587dd95e7cdec","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-01-28 17:27:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b3aea415d2eaca394affe0da9fe9edfc848b0213","revisions":{"7240fa319dd274c1cd6a235dbb1844c7f40685b1":{"kind":"REWORK","_number":1,"created":"2025-06-03 09:18:59.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/36/951636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/951636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/1"}}},"commit":{"parents":[{"commit":"4b8f26f46e7c5e8c714516719cff66df519d2d19","subject":"Merge \"Add remove-os-volumes_boot-api spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4b8f26f46e7c5e8c714516719cff66df519d2d19"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-06-03 09:13:30.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-06-03 09:13:30.000000000","tz":540},"subject":"Add spec for virtio-blk multiqueue and iothreads extra_spec","message":"Add spec for virtio-blk multiqueue and iothreads extra_spec\n\nCloud admins want to configure the virtio-blk\u0027s multiqueue\nand iothreads for server\u0027s performance improvement.\n\nThis spec enables cloud admin to configure virtio-blk\nconfiguration using flavor\u0027s extra_specs.\n\nAPIImpact\n\nblueprint: add-virtio-blk-iothreads\nChange-Id: If0a3f26fe3efa9c64cba5f743a8e8cbaae78a9bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7240fa319dd274c1cd6a235dbb1844c7f40685b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7240fa319dd274c1cd6a235dbb1844c7f40685b1"}]},"branch":"refs/heads/master"},"84877f1b4179c8b1d4c00e1893e9720dd28c8cd0":{"kind":"REWORK","_number":2,"created":"2025-07-02 09:12:30.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/36/951636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/951636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/2"}}},"commit":{"parents":[{"commit":"4b8f26f46e7c5e8c714516719cff66df519d2d19","subject":"Merge \"Add remove-os-volumes_boot-api spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4b8f26f46e7c5e8c714516719cff66df519d2d19"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-06-03 09:13:30.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-07-02 09:11:49.000000000","tz":540},"subject":"Add spec for virtio-blk multiqueue and iothreads extra_spec","message":"Add spec for virtio-blk multiqueue and iothreads extra_spec\n\nCloud admins want to configure the virtio-blk\u0027s multiqueue\nand iothreads for server\u0027s performance improvement.\n\nThis spec enables cloud admin to configure virtio-blk\nconfiguration using flavor\u0027s extra_specs.\n\nblueprint: add-virtio-blk-iothreads\nChange-Id: If0a3f26fe3efa9c64cba5f743a8e8cbaae78a9bc\nSigned-off-by: Masahito Muroi \u003cmasahito.muroi@linecorp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/84877f1b4179c8b1d4c00e1893e9720dd28c8cd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/84877f1b4179c8b1d4c00e1893e9720dd28c8cd0"}]},"branch":"refs/heads/master"},"b3532001fa8a6626de2f2bc4b425f75ce00de8b7":{"kind":"REWORK","_number":3,"created":"2025-07-03 16:50:55.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/36/951636/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/951636/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/3"}}},"commit":{"parents":[{"commit":"4b8f26f46e7c5e8c714516719cff66df519d2d19","subject":"Merge \"Add remove-os-volumes_boot-api spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4b8f26f46e7c5e8c714516719cff66df519d2d19"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-06-03 09:13:30.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-07-03 16:49:40.000000000","tz":540},"subject":"Add spec for virtio-blk multiqueue extra_spec","message":"Add spec for virtio-blk multiqueue extra_spec\n\nCloud admins want to configure the virtio-blk\u0027s multiqueue\nfor server\u0027s performance improvement.\n\nThis spec enables cloud admin to configure virtio-blk\nconfiguration using flavor\u0027s extra_specs.\n\nblueprint: add-virtio-blk-iothreads\nChange-Id: If0a3f26fe3efa9c64cba5f743a8e8cbaae78a9bc\nSigned-off-by: Masahito Muroi \u003cmasahito.muroi@linecorp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b3532001fa8a6626de2f2bc4b425f75ce00de8b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b3532001fa8a6626de2f2bc4b425f75ce00de8b7"}]},"branch":"refs/heads/master"},"581923fe878702782de47c6dfa59fef71c85b9ba":{"kind":"REWORK","_number":4,"created":"2025-10-16 08:38:58.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/36/951636/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/951636/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/4"}}},"commit":{"parents":[{"commit":"9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7","subject":"Move Flamingo implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-06-03 09:13:30.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-10-16 08:38:13.000000000","tz":540},"subject":"Add spec for virtio-blk multiqueue extra_spec","message":"Add spec for virtio-blk multiqueue extra_spec\n\nCloud admins want to configure the virtio-blk\u0027s multiqueue\nfor server\u0027s performance improvement.\n\nThis spec enables cloud admin to configure virtio-blk\nconfiguration using flavor\u0027s extra_specs.\n\nPartial implement blueprint add-virtio-blk-iothreads\nChange-Id: If0a3f26fe3efa9c64cba5f743a8e8cbaae78a9bc\nSigned-off-by: Masahito Muroi \u003cmasahito.muroi@linecorp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/581923fe878702782de47c6dfa59fef71c85b9ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/581923fe878702782de47c6dfa59fef71c85b9ba"}]},"branch":"refs/heads/master"},"b3aea415d2eaca394affe0da9fe9edfc848b0213":{"kind":"REWORK","_number":5,"created":"2025-10-16 09:37:24.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/36/951636/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/951636/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/951636/5"}}},"commit":{"parents":[{"commit":"9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7","subject":"Move Flamingo implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-06-03 09:13:30.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-10-16 09:37:15.000000000","tz":540},"subject":"Add spec for virtio-blk multiqueue extra_spec","message":"Add spec for virtio-blk multiqueue extra_spec\n\nCloud admins want to configure the virtio-blk\u0027s multiqueue\nfor server\u0027s performance improvement.\n\nThis spec enables cloud admin to configure virtio-blk\nconfiguration using flavor\u0027s extra_specs.\n\nPartial implement blueprint add-virtio-blk-iothreads\nChange-Id: If0a3f26fe3efa9c64cba5f743a8e8cbaae78a9bc\nSigned-off-by: Masahito Muroi \u003cmasahito.muroi@linecorp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b3aea415d2eaca394affe0da9fe9edfc848b0213"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b3aea415d2eaca394affe0da9fe9edfc848b0213"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
