)]}'
{"id":"openstack%2Fnova-specs~958900","triplet_id":"openstack%2Fnova-specs~master~I2b3c9cf2ff0206b7cb4edb57d2756603a34777d2","project":"openstack/nova-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-12-03 18:02:43.000000000","reason":"Change was submitted"},"26250":{"account":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"last_update":"2025-12-03 18:02:43.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-12-01 16:40:43.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-12-03 18:02:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2b3c9cf2ff0206b7cb4edb57d2756603a34777d2","subject":"Spec for asynchronous volume-attach API","status":"MERGED","created":"2025-08-29 15:10:22.000000000","updated":"2025-12-03 18:03:48.000000000","submitted":"2025-12-03 18:02:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":29,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"958900","meta_rev_id":"70bb0980b8d7dcaae1588b3e482561702559fea5","_number":958900,"virtual_id_number":958900,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-03 18:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-01 16:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2025-12-03 17:50:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2025-12-03 17:50:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"}],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-29 15:22:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-30 10:02:02.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2025-08-30 10:02:05.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-09-01 07:53:12.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-09-01 11:54:04.000000000","updated_by":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"reviewer":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"state":"CC"},{"updated":"2025-11-27 10:38:02.000000000","updated_by":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"reviewer":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"state":"REVIEWER"},{"updated":"2025-12-03 17:50:10.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"6a3c93297e785ac3b3fa98a801030520f17eb260","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-08-29 15:10:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"987e6ed3da5f0c087252dceae665f3bbdbe83eeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 15:22:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf7da1caa4614913930fa9732156c811\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f94b44859ef449a1bbfbd5fea9f138b7 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9bc9cdc126d473bbcda36d812761c73 : FAILURE in 2m 26s","accounts_in_message":[],"_revision_number":1},{"id":"98bb2f95d0a7ea4df8bc39ec564e2e134c94e10c","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-30 10:02:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"337af3b63d2001f8f93128abc2f973856411d032","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-30 10:02:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7839fb7543ac2ee801f18ed8f85b81dfbd651a95","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-09-01 07:53:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e94023697a8f7b9adf1c040d336826647f83ffa","author":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"date":"2025-09-01 11:54:04.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d398d2281e37417c783bf210b14a4a4f71956daa","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-09-02 08:32:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7bcb9bf2c1d89e0e342bf89ca005bda69c09a35","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-09-02 13:55:03.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ffd01227dafb464794b9856c5071bff705810f48","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-03 09:25:34.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"00042ddf395214edff1e83724f43045f5bd5fa94","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-09-03 13:30:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41305cf3c051d0589d72603ef2744753471e3245","author":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"date":"2025-09-04 09:18:59.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2a01ba805d1d23fba074c23d880fd77527be9daa","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-09-04 13:44:41.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ec0d12a6a7f4e84ae2624a47fe12c32db4dc6416","author":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"date":"2025-09-05 08:37:11.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"51c18162136521f44826c46ba2514fe3f6e1db9b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-10-20 16:16:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c825a51161bd62a26b19a8599a70903ab471693c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-11-14 14:49:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cc4c438178ac371dade6c787a6889046987be69a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-11-14 14:51:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e307eecb0825077a42f2955f31271930a821de6d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-11-14 14:51:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"212233882546600a02736e1850f9e217b8d1e411","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 15:01:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86dd43262341487492fcb48005e3944b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1353af3d15842bf9cf977a2d499936f : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a12e66edfd904dd19aa511783f2daf94 : FAILURE in 2m 54s","accounts_in_message":[],"_revision_number":3},{"id":"83f7c35e38ec912cc7adb17667a5db8731b15897","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-11-20 13:16:53.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7cd9b6e7ad85a3c9107f75977303a6ada0d77475","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"date":"2025-11-27 10:38:02.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2b3bc68432d024c92816a9d8a2321cc205b8c27f","author":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"date":"2025-11-27 10:40:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7ca99543836a1429585c0ebc2576520f0e5ec3af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 10:53:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/759ef5e379c9465292d118a871ca4f78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aefaf88d1894a529d03d20f961810de : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf49e29508094a21872caed613423db8 : FAILURE in 3m 14s","accounts_in_message":[],"_revision_number":4},{"id":"5f0d795678105dc7fb9882dbc980773b7967a224","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"date":"2025-11-27 11:15:57.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5ebe92e532e966552f5c3c74b119c6cf7ecbaef5","author":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"date":"2025-11-27 11:16:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"23a4cfbcf1e2cf6e05f90bd4c31bd05e96d5df6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 11:27:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce65d7fdfc524dfeaa19fe31790a8a61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43093b69fc6546a2bcad029b596f0c01 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad2e4296e7a14d75bb2806cdd114cb5d : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":5},{"id":"bad6c7c258e19185af80415a31234c9994f2b4f6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-12-01 16:40:43.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ffb50bc2f7f99ba72d0668167d75e93cc40c1949","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-12-03 17:50:10.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"801e50d55ffd5a0d287510092b4448148bccad52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 17:50:21.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8b697cfe7b99cd4d0771887732a955b40c4b5912","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 18:02:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9845aa88ac64ecea7ef96da9f244377\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8700320f85b74b5ab14bd5d061b7c774 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0610f831e0d4f65bfc0ac532303427b : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":5},{"id":"17ea3e0f5d9c4fa13564a03affa0a0f0e0dc582c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 18:02:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"70bb0980b8d7dcaae1588b3e482561702559fea5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 18:03:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8962f49f2b804d76977b15c034842232\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/9f561e4260514b728242a9f963f7a9d1 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"435fc7159733a103e6b62009355ebbb3a70acf3b","revisions":{"9135a8adbe874f32ccf5488efb274682d4c30df3":{"kind":"REWORK","_number":1,"created":"2025-08-29 15:10:22.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/00/958900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/958900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/1"}}},"commit":{"parents":[{"commit":"25619599f55bce1fd3ae63f58603c5b1e9e80a81","subject":"Merge \"Support for tracking traits removed from provider.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/25619599f55bce1fd3ae63f58603c5b1e9e80a81"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-08-29 15:09:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-08-29 15:10:25.000000000","tz":-420},"subject":"WIP: Spec for async volume attach","message":"WIP: Spec for async volume attach\n\nChange-Id: I2b3c9cf2ff0206b7cb4edb57d2756603a34777d2\nSigned-off-by: Dan Smith \u003cdansmith@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9135a8adbe874f32ccf5488efb274682d4c30df3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9135a8adbe874f32ccf5488efb274682d4c30df3"}]},"branch":"refs/heads/master"},"64cb9716b4d0b10f12c3cfb0f15b0cb8bd4e73a0":{"kind":"REWORK","_number":2,"created":"2025-11-14 14:49:28.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/00/958900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/958900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/2"}}},"commit":{"parents":[{"commit":"25619599f55bce1fd3ae63f58603c5b1e9e80a81","subject":"Merge \"Support for tracking traits removed from provider.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/25619599f55bce1fd3ae63f58603c5b1e9e80a81"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-08-29 15:09:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-11-14 14:49:23.000000000","tz":-480},"subject":"WIP: Spec for async volume attach","message":"WIP: Spec for async volume attach\n\nChange-Id: I2b3c9cf2ff0206b7cb4edb57d2756603a34777d2\nSigned-off-by: Dan Smith \u003cdansmith@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/64cb9716b4d0b10f12c3cfb0f15b0cb8bd4e73a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/64cb9716b4d0b10f12c3cfb0f15b0cb8bd4e73a0"}]},"branch":"refs/heads/master"},"39c2cc50d8ae9588b267a455b64f3e4d3f460054":{"kind":"REWORK","_number":3,"created":"2025-11-14 14:51:28.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/00/958900/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/958900/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/3"}}},"commit":{"parents":[{"commit":"25619599f55bce1fd3ae63f58603c5b1e9e80a81","subject":"Merge \"Support for tracking traits removed from provider.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/25619599f55bce1fd3ae63f58603c5b1e9e80a81"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-08-29 15:09:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-11-14 14:51:23.000000000","tz":-480},"subject":"WIP: Spec for async volume attach","message":"WIP: Spec for async volume attach\n\nChange-Id: I2b3c9cf2ff0206b7cb4edb57d2756603a34777d2\nSigned-off-by: Dan Smith \u003cdansmith@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/39c2cc50d8ae9588b267a455b64f3e4d3f460054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/39c2cc50d8ae9588b267a455b64f3e4d3f460054"}]},"branch":"refs/heads/master"},"8e149251075f83c3f5f5b71845e1efb9c0777164":{"kind":"REWORK","_number":4,"created":"2025-11-27 10:38:02.000000000","uploader":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"ref":"refs/changes/00/958900/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/958900/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/4"}}},"commit":{"parents":[{"commit":"25619599f55bce1fd3ae63f58603c5b1e9e80a81","subject":"Merge \"Support for tracking traits removed from provider.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/25619599f55bce1fd3ae63f58603c5b1e9e80a81"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-08-29 15:09:28.000000000","tz":-420},"committer":{"name":"Johannes Kulik","email":"johannes.kulik@sap.com","date":"2025-11-27 10:37:58.000000000","tz":60},"subject":"Spec for asynchronous volume-attach API","message":"Spec for asynchronous volume-attach API\n\nChange-Id: I2b3c9cf2ff0206b7cb4edb57d2756603a34777d2\nSigned-off-by: Dan Smith \u003cdansmith@redhat.com\u003e\nSigned-off-by: Johannes Kulik \u003cjohannes.kulik@sap.com\u003e\nCo-authored-By: Johannes Kulik \u003cjohannes.kulik@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8e149251075f83c3f5f5b71845e1efb9c0777164"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8e149251075f83c3f5f5b71845e1efb9c0777164"}]},"branch":"refs/heads/master"},"435fc7159733a103e6b62009355ebbb3a70acf3b":{"kind":"REWORK","_number":5,"created":"2025-11-27 11:15:57.000000000","uploader":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"ref":"refs/changes/00/958900/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/958900/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/958900/5"}}},"commit":{"parents":[{"commit":"25619599f55bce1fd3ae63f58603c5b1e9e80a81","subject":"Merge \"Support for tracking traits removed from provider.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/25619599f55bce1fd3ae63f58603c5b1e9e80a81"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-08-29 15:09:28.000000000","tz":-420},"committer":{"name":"Johannes Kulik","email":"johannes.kulik@sap.com","date":"2025-11-27 11:15:33.000000000","tz":60},"subject":"Spec for asynchronous volume-attach API","message":"Spec for asynchronous volume-attach API\n\nChange-Id: I2b3c9cf2ff0206b7cb4edb57d2756603a34777d2\nSigned-off-by: Dan Smith \u003cdansmith@redhat.com\u003e\nSigned-off-by: Johannes Kulik \u003cjohannes.kulik@sap.com\u003e\nCo-authored-By: Johannes Kulik \u003cjohannes.kulik@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/435fc7159733a103e6b62009355ebbb3a70acf3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/435fc7159733a103e6b62009355ebbb3a70acf3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
