)]}'
{"id":"openstack%2Fnova-specs~962578","triplet_id":"openstack%2Fnova-specs~master~Ic5eae827b5ba6c83490fba6aba22c82f1c7eaf99","project":"openstack/nova-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"35307":{"account":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"last_update":"2025-11-14 01:57:05.000000000","reason":"\u003cGERRIT_ACCOUNT_35307\u003e replied on the change","reason_account":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"}},"28271":{"account":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"last_update":"2025-11-27 00:13:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic5eae827b5ba6c83490fba6aba22c82f1c7eaf99","subject":"Adjust the SEV/SEV-ES code for multiple cc support","status":"ABANDONED","created":"2025-09-30 10:52:45.000000000","updated":"2025-11-27 00:13:53.000000000","total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"8ba6db0eb443df8168fa9e9f2d21550fcccb2fb5","_number":962578,"virtual_id_number":962578,"owner":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-22 06:02:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-30 11:03:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-12 13:58:47.000000000","updated_by":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"reviewer":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"state":"CC"}],"messages":[{"id":"8060974f11ca20074519b524b2bb751d090ec539","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"date":"2025-09-30 10:52:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"367f17e50f38972843788620f6c1cadf499eb01e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 11:03:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3a3fbc2c1cc472baeff82327b4388e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5e39a2734794e6f986f0a2f52cb75d3 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df5c6c440a6c433daf9c5b95cbd2dcda : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"70ea1034bccf49b45435234ceef4ec24413ec33e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"date":"2025-10-14 08:25:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5acba96198ea63c49994dcc51dc3d7c8953348d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 08:33:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/647c692f7256430cae3cd866165c3204\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f1bd9cbe8104f938bd685bff3f85240 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baae737921bb4facb1dc6450a41e8772 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"3d6fed731721ee972d704dc82c6c881f13b54a30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"date":"2025-10-22 05:51:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"134e54b8e9ae20a1bd7fa3e141bcd3e16082ea6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"date":"2025-10-22 05:55:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"84e767f4d4af6186348e941dc98c4500aa83ae2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 06:02:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fee47eb769043ec895f617161981e4a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d8f34bc777a4848a18e19c4192317d7 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b0e0ca105b1446fa4ac2daaf50813a9 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":4},{"id":"86007533980824d56a838fab978cfe6d860cf6d3","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2025-11-12 13:58:47.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"64862f97a4a1cc807b20b7b97808c4395d4ab3f6","author":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"date":"2025-11-14 01:57:05.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"89cbc3c13e0ef7413ff2d2975b1e3a4bcab7294e","author":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"date":"2025-11-27 00:12:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8ba6db0eb443df8168fa9e9f2d21550fcccb2fb5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"date":"2025-11-27 00:13:53.000000000","message":"Abandoned\n\nIt was decided at PTG to make this development specless","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9d6a1ed6c6460f255d6bf26cbce8fafc302d1b91","revisions":{"2b3fa89d11b08fa65535b05a2455401c53731210":{"kind":"REWORK","_number":1,"created":"2025-09-30 10:52:45.000000000","uploader":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"ref":"refs/changes/78/962578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/78/962578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/1"}}},"commit":{"parents":[{"commit":"9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7","subject":"Move Flamingo implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7"}]}],"author":{"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","date":"2025-09-30 10:38:06.000000000","tz":0},"committer":{"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","date":"2025-09-30 10:38:06.000000000","tz":0},"subject":"Adjust the SEV/SEV-ES code for multiple cc support","message":"Adjust the SEV/SEV-ES code for multiple cc support\n\nThe current SEV/SEV-ES code contains certain data structures\nand functions that are specific to SEV/SEV-ES implementation,\nleaving insufficient room for integrating Confidential Computing\nimplementations from other architectures(e.g., Intel TDX, Arm CCA).\n\nThis proposal aims to improve the code to\nfacilitate the addition of such implementations.\n\nImplements: blueprint generalize-sev-code\nChange-Id: Ic5eae827b5ba6c83490fba6aba22c82f1c7eaf99\nSigned-off-by: Taketani Ryo \u003ctaketani.ryo@fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b3fa89d11b08fa65535b05a2455401c53731210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b3fa89d11b08fa65535b05a2455401c53731210"}]},"branch":"refs/heads/master"},"534a77f24138c62017b232a5570e144fae1ca83c":{"kind":"REWORK","_number":2,"created":"2025-10-14 08:25:51.000000000","uploader":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"ref":"refs/changes/78/962578/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/78/962578/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/2"}}},"commit":{"parents":[{"commit":"9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7","subject":"Move Flamingo implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7"}]}],"author":{"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","date":"2025-09-30 10:38:06.000000000","tz":0},"committer":{"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","date":"2025-10-14 08:04:13.000000000","tz":0},"subject":"Adjust the SEV/SEV-ES code for multiple cc support","message":"Adjust the SEV/SEV-ES code for multiple cc support\n\nThe current SEV/SEV-ES code contains certain data structures\nand functions that are specific to SEV/SEV-ES implementation,\nleaving insufficient room for integrating Confidential Computing\nimplementations from other architectures(e.g., Intel TDX, Arm CCA).\n\nThis proposal aims to improve the code to\nfacilitate the addition of such implementations.\n\nImplements: blueprint generalize-sev-code\nChange-Id: Ic5eae827b5ba6c83490fba6aba22c82f1c7eaf99\nSigned-off-by: Taketani Ryo \u003ctaketani.ryo@fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/534a77f24138c62017b232a5570e144fae1ca83c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/534a77f24138c62017b232a5570e144fae1ca83c"}]},"branch":"refs/heads/master"},"fac36888e976afac4077cc42bf24b50789edd0f9":{"kind":"REWORK","_number":3,"created":"2025-10-22 05:51:45.000000000","uploader":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"ref":"refs/changes/78/962578/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/78/962578/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/3"}}},"commit":{"parents":[{"commit":"9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7","subject":"Move Flamingo implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7"}]}],"author":{"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","date":"2025-09-30 10:38:06.000000000","tz":0},"committer":{"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","date":"2025-10-22 05:51:47.000000000","tz":0},"subject":"Adjust the SEV/SEV-ES code for multiple cc support","message":"Adjust the SEV/SEV-ES code for multiple cc support\n\nThe current SEV/SEV-ES code contains certain data structures\nand functions that are specific to SEV/SEV-ES implementation,\nleaving insufficient room for integrating Confidential Computing\nimplementations from other architectures(e.g., Intel TDX, Arm CCA).\n\nThis proposal aims to improve the code to\nfacilitate the addition of such implementations.\n\nImplements: blueprint generalize-sev-code\nChange-Id: Ic5eae827b5ba6c83490fba6aba22c82f1c7eaf99\nSigned-off-by: Taketani Ryo \u003ctaketani.ryo@fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fac36888e976afac4077cc42bf24b50789edd0f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fac36888e976afac4077cc42bf24b50789edd0f9"}]},"branch":"refs/heads/master"},"9d6a1ed6c6460f255d6bf26cbce8fafc302d1b91":{"kind":"REWORK","_number":4,"created":"2025-10-22 05:55:13.000000000","uploader":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"ref":"refs/changes/78/962578/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/78/962578/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/78/962578/4"}}},"commit":{"parents":[{"commit":"9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7","subject":"Move Flamingo implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7"}]}],"author":{"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","date":"2025-09-30 10:38:06.000000000","tz":0},"committer":{"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","date":"2025-10-22 05:56:55.000000000","tz":0},"subject":"Adjust the SEV/SEV-ES code for multiple cc support","message":"Adjust the SEV/SEV-ES code for multiple cc support\n\nThe current SEV/SEV-ES code contains certain data structures\nand functions that are specific to SEV/SEV-ES implementation,\nleaving insufficient room for integrating Confidential Computing\nimplementations from other architectures(e.g., Intel TDX, Arm CCA).\n\nThis proposal aims to improve the code to\nfacilitate the addition of such implementations.\n\nImplements: blueprint generalize-sev-code\nChange-Id: Ic5eae827b5ba6c83490fba6aba22c82f1c7eaf99\nSigned-off-by: Taketani Ryo \u003ctaketani.ryo@fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d6a1ed6c6460f255d6bf26cbce8fafc302d1b91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d6a1ed6c6460f255d6bf26cbce8fafc302d1b91"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
