)]}'
{"id":"openstack%2Fnova-specs~964444","triplet_id":"openstack%2Fnova-specs~master~Ibc6fb081ffe03224f45942762ff6816ec5399785","project":"openstack/nova-specs","branch":"master","topic":"offline-vm-migration","attention_set":{"35761":{"account":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"last_update":"2025-10-21 12:00:09.000000000","reason":"\u003cGERRIT_ACCOUNT_37697\u003e replied on the change","reason_account":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-07-22 07:50:58.000000000","reason":"\u003cGERRIT_ACCOUNT_37697\u003e mentioned you in a comment","reason_account":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2026-03-19 08:55:39.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"37697":{"account":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"last_update":"2026-07-22 07:50:58.000000000","reason":"\u003cGERRIT_ACCOUNT_37697\u003e replied on the change","reason_account":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"}}},"hashtags":[],"change_id":"Ibc6fb081ffe03224f45942762ff6816ec5399785","subject":"Cold VM migration using Webdav","status":"NEW","created":"2025-10-21 06:18:24.000000000","updated":"2026-07-22 09:04:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":66,"unresolved_comment_count":15,"has_review_started":true,"meta_rev_id":"ac5286bba684cbc49bc75641d6e84d468b95f342","_number":964444,"virtual_id_number":964444,"owner":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-22 07:59:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-21 06:29:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-21 07:50:58.000000000","updated_by":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"reviewer":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"state":"CC"},{"updated":"2025-10-21 12:42:18.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-12-04 09:17:43.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"cbf2228cbac901eaa509b07dfd9a07f322d7370d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2025-10-21 06:18:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4fe61470d3b02f7de98dc3f71a1374e9c29a24a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 06:29:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf168cffe0c94ee883f9a17887f1da48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8f365f63a9d42679cc04e3e9bb79b5d : FAILURE in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db5b1e91b26140839a15b4931ee2a07d : FAILURE in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"8ff57fdf6825e9b2b3feb84f0c96b0039aa8549d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2025-10-21 06:33:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ac981a53220817a9a0371206bdb377db5e5f6a0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 06:47:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dcd32ac9eb14896876b0df2461b4e23\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dfa9044af744cda859c04aacc27c2a1 : FAILURE in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfa8fb74235741f7be6da0d82c7480cf : FAILURE in 2m 50s","accounts_in_message":[],"_revision_number":2},{"id":"9f5cc5522a56102180622ddf98d8dfdc477948c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2025-10-21 06:55:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6a2072dc46a13421994a07b3d04b30d340cd6b23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 07:05:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/626f1ff0ed4c4b158579c1779cc3a5c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/435516f669554345ad5c78af6561162d : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dbd7224e7084e58881f40e7bb3a984f : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":3},{"id":"087bd25c42f67b546bfda594ad2bbd927b640a3d","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2025-10-21 07:50:58.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"665175e98c353ce32e2dd135a4d154ea2581e73c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2025-10-21 11:55:15.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a76c01f51f3a9f8d1b49dbe6f30cecde6c296922","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2025-10-21 12:00:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3a52df6c4b3bb3c77d589e5c2dbca376217b978a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 12:03:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2df04d62ed04f7fabe61ba5982e4c5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f725d72ed864a87a51e753ab50cd5df : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b7db5bf9af04cc28a6ab47fdd073533 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":4},{"id":"23f4d6a9ccfcce26a4096e7b99d14da761f89d5a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-10-21 12:42:18.000000000","message":"Patch Set 4: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bb457464d833417b18a901190d0525c533be9548","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2025-10-31 12:22:26.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"30574476d03726ae499f7a7d1afa076ffa670949","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-10-31 12:36:04.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"255d6b4a47cd8d247d208ab7360ae1ed86ba5be2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2025-12-02 08:02:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8ae8eb2c883e76f396a69bff2a1951535f323006","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2025-12-02 08:03:16.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"86be76f8b129f7c575e548a6e79ba4d49b5920fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 08:12:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1ac9f62238e4d9e989e718fa91af18e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/023e59d3fe78412eab3ece63bb520b91 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbf41a34d5d6492598fdf4d5343c446d : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":6},{"id":"1710f819ada3c776b5e09bec1bed35390020fb14","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2025-12-02 08:16:30.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8ece3c5046ba5d02640a877642289bc06a879859","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2025-12-02 08:18:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"438afe4d3b5f8770550d744ff0de8c7016af3ef7","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-12-04 09:17:43.000000000","message":"Patch Set 6: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":6},{"id":"cbfe5500bddf942a8c0480e3e490890ca5a506af","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-12-04 17:48:50.000000000","message":"Patch Set 6: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c5c10d9f19cf1626c7b2d17991cd5e1478a90ccf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2026-03-19 08:55:39.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"e9f7afba18344d60ddb564fa99a19c03a738d28f","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2026-03-19 08:58:48.000000000","message":"Patch Set 6:\n\n(8 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2f369b88e71029cdcd2066e53ff14f32af23b23a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 09:07:27.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6ad8364f37d49058e357e209659f53e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5847cbbf55e41ef9d35ce6c961aec10 : FAILURE in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb42c8d2fe7c40debc6b23de0923cc15 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":7},{"id":"fe3ec84b3fbfa183a9ef44fb707e599710f56b2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2026-04-01 05:16:31.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"00332a31333dcf89e73332259e39ce0c801a307e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 05:20:44.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c6fd5f7babd4d5bb05016c3b1785dcb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65c77071387647bd9e64aa410f608b77 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38142dfed82e4326af3b8c181c871913 : FAILURE in 2m 25s","accounts_in_message":[],"_revision_number":8},{"id":"8a6e65c1f08d282fbf3d0f4f136536b9cd5efe0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2026-04-01 06:10:25.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"33370dcd9614c5c82d30452333e1bcda2c28edd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 06:21:17.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f1ca4a484ae4a14be16de9c9188fe7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6048b8a7624d4c4dbf74fa981afa100c : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c49928576f34c71916e06a4203f7462 : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":9},{"id":"97dbc48047a58e9b482c39d8e9e44b45a0089180","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-05 12:27:52.000000000","message":"Patch Set 9: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":9},{"id":"0b586c4bb8623371f8afc4261b66ab61f4774e6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2026-07-22 07:49:28.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"bd03287db352c2f9b8574e3cf915b7bba77de683","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2026-07-22 07:50:58.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"3d1b4939088ce58a127e983a9a15db3aa3b2bffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 07:59:04.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f847dcafb47e4b868d45bfc92a911752\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e60b0263723e4a04b32a9bc09a475cde : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98425ef5efae40429b3684bd70fa7cb9 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":10},{"id":"ac5286bba684cbc49bc75641d6e84d468b95f342","author":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"date":"2026-07-22 09:04:31.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"157792ecc54c054c6301523edf24fe5fb5f9d750","revisions":{"3a6b35b35d4801db021a603c81a516da9738f48d":{"kind":"REWORK","_number":1,"created":"2025-10-21 06:18:24.000000000","uploader":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"ref":"refs/changes/44/964444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/964444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/1"}}},"commit":{"parents":[{"commit":"9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7","subject":"Move Flamingo implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7"}]}],"author":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-10-21 06:17:25.000000000","tz":300},"committer":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-10-21 06:17:25.000000000","tz":300},"subject":"Offline VM migration using https","message":"Offline VM migration using https\n\nImplements https driver and https file server\nfor migrating disk from one node to the other node.\n\nChange-Id: Ibc6fb081ffe03224f45942762ff6816ec5399785\nSigned-off-by: Ahmad Hassan \u003cahmad.hassan@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3a6b35b35d4801db021a603c81a516da9738f48d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3a6b35b35d4801db021a603c81a516da9738f48d"}]},"branch":"refs/heads/master"},"3039269713c8277db054170850bafff728951d8b":{"kind":"REWORK","_number":2,"created":"2025-10-21 06:33:32.000000000","uploader":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"ref":"refs/changes/44/964444/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/964444/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/2"}}},"commit":{"parents":[{"commit":"9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7","subject":"Move Flamingo implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7"}]}],"author":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-10-21 06:17:25.000000000","tz":300},"committer":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-10-21 06:32:32.000000000","tz":300},"subject":"Offline VM migration using https","message":"Offline VM migration using https\n\nImplements https driver and https file server\nfor migrating disk from one node to the other node.\n\nChange-Id: Ibc6fb081ffe03224f45942762ff6816ec5399785\nSigned-off-by: Ahmad Hassan \u003cahmad.hassan@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3039269713c8277db054170850bafff728951d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3039269713c8277db054170850bafff728951d8b"}]},"branch":"refs/heads/master"},"6b5c43d717b6158bfe42b692e156c685f6cce0d0":{"kind":"REWORK","_number":3,"created":"2025-10-21 06:55:16.000000000","uploader":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"ref":"refs/changes/44/964444/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/964444/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/3"}}},"commit":{"parents":[{"commit":"9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7","subject":"Move Flamingo implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7"}]}],"author":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-10-21 06:17:25.000000000","tz":300},"committer":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-10-21 06:54:08.000000000","tz":300},"subject":"Offline VM migration using https","message":"Offline VM migration using https\n\nImplements https driver and https file server\nfor migrating disk from one node to the other node.\n\nChange-Id: Ibc6fb081ffe03224f45942762ff6816ec5399785\nSigned-off-by: Ahmad Hassan \u003cahmad.hassan@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6b5c43d717b6158bfe42b692e156c685f6cce0d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6b5c43d717b6158bfe42b692e156c685f6cce0d0"}]},"branch":"refs/heads/master"},"dbb137a12075ebe00794ae4820e6f3480173e361":{"kind":"REWORK","_number":4,"created":"2025-10-21 11:55:15.000000000","uploader":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"ref":"refs/changes/44/964444/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/964444/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/4"}}},"commit":{"parents":[{"commit":"9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7","subject":"Move Flamingo implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7"}]}],"author":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-10-21 06:17:25.000000000","tz":300},"committer":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-10-21 11:52:48.000000000","tz":300},"subject":"Offline VM migration using https","message":"Offline VM migration using https\n\nImplements https driver and https file server\nfor migrating disk from one node to the other node.\n\nChange-Id: Ibc6fb081ffe03224f45942762ff6816ec5399785\nSigned-off-by: Ahmad Hassan \u003cahmad.hassan@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dbb137a12075ebe00794ae4820e6f3480173e361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dbb137a12075ebe00794ae4820e6f3480173e361"}]},"branch":"refs/heads/master"},"c7f23175118a880e859ef8dd40b0e85c75ff4f73":{"kind":"REWORK","_number":5,"created":"2025-12-02 08:02:14.000000000","uploader":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"ref":"refs/changes/44/964444/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/964444/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/5"}}},"commit":{"parents":[{"commit":"9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7","subject":"Move Flamingo implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7"}]}],"author":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-10-21 06:17:25.000000000","tz":300},"committer":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-12-02 08:01:48.000000000","tz":300},"subject":"Offline VM migration using https","message":"Offline VM migration using https\n\nImplements https driver and https file server\nfor migrating disk from one node to the other node.\n\nChange-Id: Ibc6fb081ffe03224f45942762ff6816ec5399785\nSigned-off-by: Ahmad Hassan \u003cahmad.hassan@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c7f23175118a880e859ef8dd40b0e85c75ff4f73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c7f23175118a880e859ef8dd40b0e85c75ff4f73"}]},"branch":"refs/heads/master"},"41dc6b8f5bc64afa7fe7c19360271a28c9cb1325":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-12-02 08:03:16.000000000","uploader":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"ref":"refs/changes/44/964444/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/964444/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/6"}}},"commit":{"parents":[{"commit":"9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7","subject":"Move Flamingo implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7"}]}],"author":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-10-21 06:17:25.000000000","tz":300},"committer":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-12-02 08:03:16.000000000","tz":0},"subject":"Cold VM migration using Webdav https","message":"Cold VM migration using Webdav https\n\nImplements https driver for cold migrating disk from one node to the other node.\n\nChange-Id: Ibc6fb081ffe03224f45942762ff6816ec5399785\nSigned-off-by: Ahmad Hassan \u003cahmad.hassan@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/41dc6b8f5bc64afa7fe7c19360271a28c9cb1325"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/41dc6b8f5bc64afa7fe7c19360271a28c9cb1325"}]},"branch":"refs/heads/master","description":"Edit commit message"},"123edea2d0e429773b7084186b681d653395c63d":{"kind":"REWORK","_number":7,"created":"2026-03-19 08:55:39.000000000","uploader":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"ref":"refs/changes/44/964444/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/964444/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/7"}}},"commit":{"parents":[{"commit":"9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7","subject":"Move Flamingo implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7"}]}],"author":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-10-21 06:17:25.000000000","tz":300},"committer":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2026-03-19 08:53:54.000000000","tz":300},"subject":"Cold VM migration using https","message":"Cold VM migration using https\n\nCold VM migration using Webdav Nova Client\n\nChange-Id: Ibc6fb081ffe03224f45942762ff6816ec5399785\nSigned-off-by: Ahmad Hassan \u003cahmad.hassan@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/123edea2d0e429773b7084186b681d653395c63d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/123edea2d0e429773b7084186b681d653395c63d"}]},"branch":"refs/heads/master"},"7a596d42a67fe8eca2c80494e305c30af162c37b":{"kind":"REWORK","_number":8,"created":"2026-04-01 05:16:31.000000000","uploader":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"ref":"refs/changes/44/964444/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/964444/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/8"}}},"commit":{"parents":[{"commit":"9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7","subject":"Move Flamingo implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7"}]}],"author":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-10-21 06:17:25.000000000","tz":300},"committer":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2026-04-01 05:16:10.000000000","tz":300},"subject":"Cold VM migration using https","message":"Cold VM migration using https\n\nCold VM migration using Webdav Nova Client\n\nChange-Id: Ibc6fb081ffe03224f45942762ff6816ec5399785\nSigned-off-by: Ahmad Hassan \u003cahmad.hassan@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7a596d42a67fe8eca2c80494e305c30af162c37b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7a596d42a67fe8eca2c80494e305c30af162c37b"}]},"branch":"refs/heads/master"},"dc470f7737cda59f15570b8c6245b0a1ae2f5e88":{"kind":"REWORK","_number":9,"created":"2026-04-01 06:10:25.000000000","uploader":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"ref":"refs/changes/44/964444/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/964444/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/9"}}},"commit":{"parents":[{"commit":"9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7","subject":"Move Flamingo implemented specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d135db4bdcd53f0a6d71ad54b8e232ff7dbf0d7"}]}],"author":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-10-21 06:17:25.000000000","tz":300},"committer":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2026-04-01 06:10:07.000000000","tz":300},"subject":"Cold VM migration using https","message":"Cold VM migration using https\n\nCold VM migration using Webdav Nova Client\n\nChange-Id: Ibc6fb081ffe03224f45942762ff6816ec5399785\nSigned-off-by: Ahmad Hassan \u003cahmad.hassan@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dc470f7737cda59f15570b8c6245b0a1ae2f5e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dc470f7737cda59f15570b8c6245b0a1ae2f5e88"}]},"branch":"refs/heads/master"},"157792ecc54c054c6301523edf24fe5fb5f9d750":{"kind":"REWORK","_number":10,"created":"2026-07-22 07:49:28.000000000","uploader":{"_account_id":37697,"name":"Ahmad Hassan","display_name":"Ahmad","email":"ahmad.hassan@canonical.com","username":"hassahma"},"ref":"refs/changes/44/964444/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/964444/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/964444/10"}}},"commit":{"parents":[{"commit":"e8b24e18df8001564b39cafb8958a76a585fea31","subject":"Use domain capability to detect SEV-SNP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e8b24e18df8001564b39cafb8958a76a585fea31"}]}],"author":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2025-10-21 06:17:25.000000000","tz":300},"committer":{"name":"Ahmad Hassan","email":"ahmad.hassan@canonical.com","date":"2026-07-22 07:48:48.000000000","tz":300},"subject":"Cold VM migration using Webdav","message":"Cold VM migration using Webdav\n\nCold VM migration using Webdav Nova Client\n\nChange-Id: Ibc6fb081ffe03224f45942762ff6816ec5399785\nSigned-off-by: Ahmad Hassan \u003cahmad.hassan@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/157792ecc54c054c6301523edf24fe5fb5f9d750"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/157792ecc54c054c6301523edf24fe5fb5f9d750"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
