)]}'
{"id":"openstack%2Fnova-specs~968065","triplet_id":"openstack%2Fnova-specs~master~I75d776bca8e41e483e2253413975ada7a832e89c","project":"openstack/nova-specs","branch":"master","topic":"improve-image-cache-behavior","attention_set":{"11302":{"account":{"_account_id":11302,"name":"Phillip Toohill","email":"phillip.toohill@rackspace.com","username":"ptoohill1"},"last_update":"2025-12-08 18:40:37.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-11-25 15:28:57.000000000","reason":"\u003cGERRIT_ACCOUNT_11302\u003e replied on the change","reason_account":{"_account_id":11302,"name":"Phillip Toohill","email":"phillip.toohill@rackspace.com","username":"ptoohill1"}}},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-12-08 18:40:37.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"hashtags":[],"change_id":"I75d776bca8e41e483e2253413975ada7a832e89c","subject":"WIP: Improve image caching behavior","status":"NEW","created":"2025-11-21 19:46:08.000000000","updated":"2025-12-08 18:40:37.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":15,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"eff7bf862a875a5d875936e1f73342dbf2b85783","_number":968065,"virtual_id_number":968065,"owner":{"_account_id":11302,"name":"Phillip Toohill","email":"phillip.toohill@rackspace.com","username":"ptoohill1"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-11-24 17:26:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2025-11-24 17:39:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-21 19:59:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-24 17:39:23.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-11-25 12:41:07.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"}],"messages":[{"id":"65fa80acad3acf4e25a18c7e30ffb34e1f594144","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11302,"name":"Phillip Toohill","email":"phillip.toohill@rackspace.com","username":"ptoohill1"},"date":"2025-11-21 19:46:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9660ab283308ce574be399cbecc3eafc23cc1479","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 19:59:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32ec385eb1b64971bf62ebf5da0b45a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9c6a1b755a74a7492fccb3427c7c131 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/565db774ff02476a89e8c722e37faa75 : FAILURE in 2m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6838c8a7870218305f9485897ec2cdc7fde76886","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11302,"name":"Phillip Toohill","email":"phillip.toohill@rackspace.com","username":"ptoohill1"},"date":"2025-11-21 20:07:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4525ca20a88f77f138bf341030f51f3e6203b7c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 20:12:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b325fa012fc43439967d73e93dc542a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d3c6169dd244c6b92de6c72a29cef0 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0263478c5d7448b5b7c3617b2fe9fd1a : FAILURE in 2m 19s","accounts_in_message":[],"_revision_number":2},{"id":"50a619e53ac2a49eea5af1d3f01f14c36779c795","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11302,"name":"Phillip Toohill","email":"phillip.toohill@rackspace.com","username":"ptoohill1"},"date":"2025-11-24 17:17:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9186f684666974f3635ebc52eac64b8b8efb745c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 17:26:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c69eb14f34a4cc297ca36e90d3f8858\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/502d23e692fa4a3b9dc0592c01f1b737 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3fe29b0ae8346eebb7e6620926e9b3b : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":3},{"id":"15674670e6b7a101fe9369c45a4b769bdf3236d4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-24 17:39:23.000000000","message":"Patch Set 3: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fc36119e65274aa3d45a99a88ad15ca0e91375b6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-11-25 12:41:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e5f8a80843d8a29a0678bf9ada13773d76a32c01","author":{"_account_id":11302,"name":"Phillip Toohill","email":"phillip.toohill@rackspace.com","username":"ptoohill1"},"date":"2025-11-25 15:28:57.000000000","message":"Patch Set 3:\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"eff7bf862a875a5d875936e1f73342dbf2b85783","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-12-08 18:40:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2a95966b88782ffbe5a096e4c56c5b2cf2ad345a","revisions":{"8c8651d7f26d57e90623bd465b10a7288a5e333b":{"kind":"REWORK","_number":1,"created":"2025-11-21 19:46:08.000000000","uploader":{"_account_id":11302,"name":"Phillip Toohill","email":"phillip.toohill@rackspace.com","username":"ptoohill1"},"ref":"refs/changes/65/968065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/65/968065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/65/968065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/65/968065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/65/968065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/65/968065/1"}}},"commit":{"parents":[{"commit":"8b5dce3c7b5ad1859893b02bc801552c4417f8f6","subject":"Merge \"Repropose remove-os-volumes_boot-api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b5dce3c7b5ad1859893b02bc801552c4417f8f6"}]}],"author":{"name":"phillipt","email":"phillip.toohill@rackspace.com","date":"2025-11-21 19:26:33.000000000","tz":-360},"committer":{"name":"phillipt","email":"phillip.toohill@rackspace.com","date":"2025-11-21 19:42:52.000000000","tz":-360},"subject":"WIP: Improve image caching behavior","message":"WIP: Improve image caching behavior\n\nChange-Id: I75d776bca8e41e483e2253413975ada7a832e89c\nSigned-off-by: phillipt \u003cphillip.toohill@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8c8651d7f26d57e90623bd465b10a7288a5e333b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8c8651d7f26d57e90623bd465b10a7288a5e333b"}]},"branch":"refs/heads/master"},"09c73352e9ddf430f028b75293c309de9b69438f":{"kind":"REWORK","_number":2,"created":"2025-11-21 20:07:45.000000000","uploader":{"_account_id":11302,"name":"Phillip Toohill","email":"phillip.toohill@rackspace.com","username":"ptoohill1"},"ref":"refs/changes/65/968065/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/65/968065/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/65/968065/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/65/968065/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/65/968065/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/65/968065/2"}}},"commit":{"parents":[{"commit":"8b5dce3c7b5ad1859893b02bc801552c4417f8f6","subject":"Merge \"Repropose remove-os-volumes_boot-api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b5dce3c7b5ad1859893b02bc801552c4417f8f6"}]}],"author":{"name":"phillipt","email":"phillip.toohill@rackspace.com","date":"2025-11-21 19:26:33.000000000","tz":-360},"committer":{"name":"phillipt","email":"phillip.toohill@rackspace.com","date":"2025-11-21 20:07:24.000000000","tz":-360},"subject":"WIP: Improve image caching behavior","message":"WIP: Improve image caching behavior\n\nChange-Id: I75d776bca8e41e483e2253413975ada7a832e89c\nSigned-off-by: phillipt \u003cphillip.toohill@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/09c73352e9ddf430f028b75293c309de9b69438f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/09c73352e9ddf430f028b75293c309de9b69438f"}]},"branch":"refs/heads/master"},"2a95966b88782ffbe5a096e4c56c5b2cf2ad345a":{"kind":"REWORK","_number":3,"created":"2025-11-24 17:17:16.000000000","uploader":{"_account_id":11302,"name":"Phillip Toohill","email":"phillip.toohill@rackspace.com","username":"ptoohill1"},"ref":"refs/changes/65/968065/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/65/968065/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/65/968065/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/65/968065/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/65/968065/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/65/968065/3"}}},"commit":{"parents":[{"commit":"8b5dce3c7b5ad1859893b02bc801552c4417f8f6","subject":"Merge \"Repropose remove-os-volumes_boot-api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b5dce3c7b5ad1859893b02bc801552c4417f8f6"}]}],"author":{"name":"phillipt","email":"phillip.toohill@rackspace.com","date":"2025-11-21 19:26:33.000000000","tz":-360},"committer":{"name":"phillipt","email":"phillip.toohill@rackspace.com","date":"2025-11-24 17:15:24.000000000","tz":-360},"subject":"WIP: Improve image caching behavior","message":"WIP: Improve image caching behavior\n\nChange-Id: I75d776bca8e41e483e2253413975ada7a832e89c\nSigned-off-by: phillipt \u003cphillip.toohill@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2a95966b88782ffbe5a096e4c56c5b2cf2ad345a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2a95966b88782ffbe5a096e4c56c5b2cf2ad345a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
