)]}'
{"id":"openstack%2Fnova-specs~975635","triplet_id":"openstack%2Fnova-specs~master~Ib96dc949328c672ba0056a30d8882de1ec009927","project":"openstack/nova-specs","branch":"master","topic":"pci-passthrough-groups","attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2026-05-11 15:11:41.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-05-12 09:50:29.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-05-12 09:50:29.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"hashtags":[],"change_id":"Ib96dc949328c672ba0056a30d8882de1ec009927","subject":"add pci passthrough groups spec for 2026.2","status":"NEW","created":"2026-02-04 13:27:40.000000000","updated":"2026-05-30 10:08:27.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":195,"unresolved_comment_count":9,"has_review_started":true,"meta_rev_id":"39120192af115a4246cc38d595fc5a228f4ba48b","_number":975635,"virtual_id_number":975635,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-11 15:23:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-05-30 10:08:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-04 13:53:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-05 14:23:52.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-02-05 15:43:27.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2026-02-17 14:19:43.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"}],"messages":[{"id":"0da05628f5b545e8cc97c259544464c2917823f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-04 13:27:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce05d9dea57330e26bee7b0b9ca080235c6960fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-04 13:36:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"253fc07a99450c423ac6cceafb874cce6260a313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 13:53:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3351ef2b29e5443da62e8b31f4b11147\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f28b8a38849f4124a10eb8c3d357e08a : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/544ac537fb40408e9e276c8545b8576a : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f0f750ce8d7b02c412195b5294d44a0f7ab128b1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-02-05 14:23:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"49196a3c5bcc21bd7a3e5aa77b403e3596a9d578","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-02-05 15:43:27.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"47b7d76fdaa9811d6ec29d52d2efc4c5b7b2b1de","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-05 17:16:57.000000000","message":"Patch Set 2:\n\n(11 comments)","accounts_in_message":[],"_revision_number":2},{"id":"da03409a5e264a0aa8d421547ca5216d73e4e13f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-02-10 12:10:25.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1eff459d5bab7d67ea321a7b0c6b2fc5d85edf4a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-04-17 14:37:25.000000000","message":"Patch Set 2:\n\n(18 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4ed806da121e5c620c8c0fe5ddae79618b67e681","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-04-17 15:25:11.000000000","message":"Patch Set 2: Workflow-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9607833603cda3df579ebe4390b09569506d2c33","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-04-17 15:31:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9f59dcb5e45b3886b3ecbfdd1d27e8ef8938e466","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-04-17 17:44:02.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c2f3193219e548a7cd27fdaf262ff7500190d7da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-04-28 14:31:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"60af217d4850a63d14db98e598fa2b6f86bfaa47","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-04-28 14:32:12.000000000","message":"Patch Set 3:\n\n(25 comments)","accounts_in_message":[],"_revision_number":3},{"id":"388bfc86590bbe9d039c3792282606333f6984bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 14:46:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f693707317bd4164b1cc04370c5f3af3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f8602b66e8c40c6a2488392aba200b9 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff3544f5cc784dd1889454de35e6136c : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":3},{"id":"67de49e704d137c102b0813d6e9263306dd31bf5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-04-28 15:49:11.000000000","message":"Patch Set 3:\n\n(14 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0f9c0c4b7a2b2a09b8cbe2bc076c6587fbc2c8a1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-04-29 09:46:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"959bbf4222c06061dff80b29fb8d4b344a1c8bfc","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-04-29 20:24:41.000000000","message":"Patch Set 3:\n\n(16 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e39add709fc27f819b17d2af689f4c32ce1fad90","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-04-29 20:25:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9a7dc652f91956ba8a7907d6a74d293d252ca88a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-04-30 14:41:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"476fb50666ac853c4f242adca5bcf7d3178305c6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-04-30 14:45:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d14b97b5ca6112aeff40a07b8165e4bb00f92ebd","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-04-30 14:59:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0f59414d796aeb62b67080359657445e5145dcc1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-04-30 15:38:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"16babbaec43509e62e3c5aa1ef32abace4c6c55f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-05-04 16:14:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cd5052ac64ea46c63d254f7d610e482bfdcc2151","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-05-07 17:45:50.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6a220381f62a173b4df374288604952cbaae6e8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-05-08 14:46:11.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"95c595c59632445fc6af09993617db3aa10df548","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-05-08 14:46:29.000000000","message":"Patch Set 3:\n\n(18 comments)","accounts_in_message":[],"_revision_number":3},{"id":"92eb464f22e7a52508e37066ff3929be297a98df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 14:54:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29cfe97dc49045329d1fe165548efc97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f5afc3de7fa41b8bdaeaca9dbc610db : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ff989010a9f48949c1cd650fea7b7bc : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":4},{"id":"0e1df34b7844a7ab6663a019cede0a2ff2dc236e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-05-08 15:18:45.000000000","message":"Patch Set 4:\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ba8ce10cfc4b1aa7266a5ca80bb19a2f095be20a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-05-08 16:47:58.000000000","message":"Patch Set 4: Code-Review+1\n\n(18 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8c8e98be53987a6c5c94d1120e223407ce56ef86","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-05-11 08:26:29.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5ce9d6bf9fa1bc9c311affbba37cede23f97ee14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-05-11 15:11:26.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"23f61d4bb8cd42f5aed8183082593a1eb325d3e2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-05-11 15:11:41.000000000","message":"Patch Set 4: -Workflow\n\n(17 comments)","accounts_in_message":[],"_revision_number":4},{"id":"066e4666879ba3473e4f39119c2035f4d734cc4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 15:23:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f84518d03e104d62a3a0e162be588fb1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe34492ffa674080aa5baa86bd91671b : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1881b93d40c6407e9f319c78b3889022 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":5},{"id":"306724d72e287768a84fdfeaf34c087021e9e1bd","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-05-11 15:35:04.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d51533a48bfe6a7709b559322b7314bb2a1877ac","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-05-12 08:14:34.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"539cf16b70445d9642ab536c222f9614cecbbc8f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-05-12 08:21:40.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ea54d9033dea492db8084cf51237af4724dd6011","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-05-12 09:50:29.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"39120192af115a4246cc38d595fc5a228f4ba48b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-05-30 10:08:27.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"793f271b0a471ec32371aecc6cc7298af9d38d8e","revisions":{"557532bd233e76c31409f92949bc55366a768c34":{"kind":"REWORK","_number":1,"created":"2026-02-04 13:27:40.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/35/975635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/35/975635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/1"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-02-04 13:25:30.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-02-04 13:27:31.000000000","tz":0},"subject":"add pci passthrough groups spec for 2026.2","message":"add pci passthrough groups spec for 2026.2\n\nThis change re proposes the prior appoved\nspec with some addtional detials from the\npoc.\n\nChange-Id: Ib96dc949328c672ba0056a30d8882de1ec009927\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/557532bd233e76c31409f92949bc55366a768c34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/557532bd233e76c31409f92949bc55366a768c34"}]},"branch":"refs/heads/master"},"f6ca110cad80c17846cfc1a3ed765bb068fb74f3":{"kind":"REWORK","_number":2,"created":"2026-02-04 13:36:41.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/35/975635/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/35/975635/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/2"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-02-04 13:25:30.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-02-04 13:36:32.000000000","tz":0},"subject":"add pci passthrough groups spec for 2026.2","message":"add pci passthrough groups spec for 2026.2\n\nThis change re proposes the prior appoved\nspec with some addtional detials from the\npoc.\n\nChange-Id: Ib96dc949328c672ba0056a30d8882de1ec009927\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f6ca110cad80c17846cfc1a3ed765bb068fb74f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f6ca110cad80c17846cfc1a3ed765bb068fb74f3"}]},"branch":"refs/heads/master"},"9012288d3b9c82e299b1f053b232fa1e40f94243":{"kind":"REWORK","_number":3,"created":"2026-04-28 14:31:48.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/35/975635/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/35/975635/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/3"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-02-04 13:25:30.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-04-28 14:30:44.000000000","tz":120},"subject":"add pci passthrough groups spec for 2026.2","message":"add pci passthrough groups spec for 2026.2\n\nThis change re proposes the prior appoved\nspec with some addtional detials from the\npoc.\n\nCo-authored-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: Ib96dc949328c672ba0056a30d8882de1ec009927\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9012288d3b9c82e299b1f053b232fa1e40f94243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9012288d3b9c82e299b1f053b232fa1e40f94243"}]},"branch":"refs/heads/master"},"519f90594561411547f6bf114bb5a01ba9c85232":{"kind":"REWORK","_number":4,"created":"2026-05-08 14:46:11.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/35/975635/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/35/975635/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/4"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-02-04 13:25:30.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-05-08 14:45:10.000000000","tz":120},"subject":"add pci passthrough groups spec for 2026.2","message":"add pci passthrough groups spec for 2026.2\n\nThis change re proposes the prior appoved\nspec with some addtional detials from the\npoc.\n\nCo-authored-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: Ib96dc949328c672ba0056a30d8882de1ec009927\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/519f90594561411547f6bf114bb5a01ba9c85232"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/519f90594561411547f6bf114bb5a01ba9c85232"}]},"branch":"refs/heads/master"},"793f271b0a471ec32371aecc6cc7298af9d38d8e":{"kind":"REWORK","_number":5,"created":"2026-05-11 15:11:26.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/35/975635/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/35/975635/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/35/975635/5"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-02-04 13:25:30.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-05-11 15:10:34.000000000","tz":120},"subject":"add pci passthrough groups spec for 2026.2","message":"add pci passthrough groups spec for 2026.2\n\nThis change re proposes the prior approved\nspec with some additional details from the\npoc.\n\nCo-authored-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: Ib96dc949328c672ba0056a30d8882de1ec009927\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/793f271b0a471ec32371aecc6cc7298af9d38d8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/793f271b0a471ec32371aecc6cc7298af9d38d8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
