)]}'
{"id":"openstack%2Fnova-specs~977339","triplet_id":"openstack%2Fnova-specs~master~Ic2cc959541d9fe371e46c6083eaec487d9aa90c7","project":"openstack/nova-specs","branch":"master","topic":"bp/instance-metadata-protection","attention_set":{"37432":{"account":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"last_update":"2026-08-06 04:48:00.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic2cc959541d9fe371e46c6083eaec487d9aa90c7","subject":"Repropose newly instance metadata/tag protection feature","status":"NEW","created":"2026-02-19 16:30:10.000000000","updated":"2026-08-06 04:48:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"2b795a1fa185afd4838db02c9eff7607fa230c96","_number":977339,"virtual_id_number":977339,"owner":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-24 05:50:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-19 16:43:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-06 04:48:00.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"f4a4a86a054594c6d8dc83cbd029c229d1d4703b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2026-02-19 16:30:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d33bc2b5ae62db26cbc14e2ca117312097b1384d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2026-02-19 16:31:15.000000000","message":"Topic set to bp/instance-metadata-protection","accounts_in_message":[],"_revision_number":1},{"id":"c3f36b2a9c2e5d78888611cb635ea621c0b80b76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2026-02-19 16:34:19.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b5fba08256c54ccaa15c0b34b048ad1055b79712","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2026-02-19 16:36:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"73f4838db679707b1625630d1f80e2f297fd3c79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 16:43:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f30f303534644fc88a81d38c53979ff6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/122bb9eef65b42fd9e21c9745fb8a177 : FAILURE in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdf8c2adc7ea49cb9be540a964618240 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"b08e9a3c34679e3ef3ac1b80fcd3f5ef5b9fee15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2026-04-13 11:33:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"0a836c8815c57c2d6b17e878a188ca3dda99d29f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 11:43:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45a597ec1fd34ead9dc32720e28c88d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a6f7e96b6074f0888b00ad86700f893 : FAILURE in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8975b292e48d493c99ec3d7f1faf465e : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":3},{"id":"f696f059c21c9078df8ec86d8c1a30e7b1d8146a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2026-04-13 13:22:58.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"f59cf8743667c862761e70777352802a4ab9ac36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 13:41:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8ca827c1e194e28aab3dc9b2437c7f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38bdc1530dc6412389b7bbf41e97d631 : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be8814d1f2124791a03c42afcf907bc7 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":4},{"id":"5ea19aa25f21d5afeb264150a746d55545f94cda","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2026-04-15 02:17:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3cb8c693816d0b9ec426cda0f923772e609d027c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2026-04-24 05:36:27.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"5bb98b628bd3a41b406ab10b5a40a54d2fb70cf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 05:50:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08e3c0e5e7c94b3eb48db32c98acfa51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df983aafdc842edbce91ba92f9ae94f : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/790aa21a4bda46aa8e5ecd589c371cc8 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":5},{"id":"2b795a1fa185afd4838db02c9eff7607fa230c96","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-06 04:48:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0d317e9c2062bc51014ddeeac5bada68e14dba6c","revisions":{"d45fbaab7dd932b07a74f7fb580ee90e158a2f4f":{"kind":"REWORK","_number":1,"created":"2026-02-19 16:30:10.000000000","uploader":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"ref":"refs/changes/39/977339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/39/977339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/1"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2026-02-19 16:27:44.000000000","tz":540},"committer":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2026-02-19 16:29:35.000000000","tz":540},"subject":"Repropose instance-metadata-tag-protection","message":"Repropose instance-metadata-tag-protection\n\nThis patch re-proposes the spec for the instance\nmetadata/tag protection feature by implementing\nthe policy enforcement using oslo.policy.\n\nChange-Id: Ic2cc959541d9fe371e46c6083eaec487d9aa90c7\nSigned-off-by: Koya Watanabe \u003ckoywatan@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d45fbaab7dd932b07a74f7fb580ee90e158a2f4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d45fbaab7dd932b07a74f7fb580ee90e158a2f4f"}]},"branch":"refs/heads/master"},"fc0e42fbea1b20ff4d568af3030003eaaf0028c6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-02-19 16:34:19.000000000","uploader":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"ref":"refs/changes/39/977339/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/39/977339/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/2"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2026-02-19 16:34:05.000000000","tz":540},"committer":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2026-02-19 16:34:05.000000000","tz":540},"subject":"Repropose newly instance metadata/tag protection feature","message":"Repropose newly instance metadata/tag protection feature\n\nThis patch re-proposes the spec for the instance\nmetadata/tag protection feature by implementing\nthe policy enforcement using oslo.policy.\n\nChange-Id: Ic2cc959541d9fe371e46c6083eaec487d9aa90c7\nSigned-off-by: Koya Watanabe \u003ckoywatan@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fc0e42fbea1b20ff4d568af3030003eaaf0028c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fc0e42fbea1b20ff4d568af3030003eaaf0028c6"}]},"branch":"refs/heads/master"},"d6463d4cf2df970b051ed307e7690e9a20322462":{"kind":"REWORK","_number":3,"created":"2026-04-13 11:33:26.000000000","uploader":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"ref":"refs/changes/39/977339/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/39/977339/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/3"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2026-02-19 16:34:05.000000000","tz":540},"committer":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2026-04-13 11:33:07.000000000","tz":540},"subject":"Repropose newly instance metadata/tag protection feature","message":"Repropose newly instance metadata/tag protection feature\n\nThis patch re-proposes the spec for the instance\nmetadata/tag protection feature by implementing\nthe policy enforcement using oslo.policy.\n\nChange-Id: Ic2cc959541d9fe371e46c6083eaec487d9aa90c7\nSigned-off-by: Koya Watanabe \u003ckoywatan@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d6463d4cf2df970b051ed307e7690e9a20322462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d6463d4cf2df970b051ed307e7690e9a20322462"}]},"branch":"refs/heads/master"},"4f9d24127b5953696a4dad513c071602c2aff817":{"kind":"REWORK","_number":4,"created":"2026-04-13 13:22:58.000000000","uploader":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"ref":"refs/changes/39/977339/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/39/977339/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/4"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2026-02-19 16:34:05.000000000","tz":540},"committer":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2026-04-13 13:22:48.000000000","tz":540},"subject":"Repropose newly instance metadata/tag protection feature","message":"Repropose newly instance metadata/tag protection feature\n\nThis patch re-proposes the spec for the instance\nmetadata/tag protection feature by implementing\nthe policy enforcement using oslo.policy.\n\nChange-Id: Ic2cc959541d9fe371e46c6083eaec487d9aa90c7\nSigned-off-by: Koya Watanabe \u003ckoywatan@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4f9d24127b5953696a4dad513c071602c2aff817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4f9d24127b5953696a4dad513c071602c2aff817"}]},"branch":"refs/heads/master"},"0d317e9c2062bc51014ddeeac5bada68e14dba6c":{"kind":"REWORK","_number":5,"created":"2026-04-24 05:36:27.000000000","uploader":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"ref":"refs/changes/39/977339/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/39/977339/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/39/977339/5"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2026-02-19 16:34:05.000000000","tz":540},"committer":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2026-04-24 05:36:00.000000000","tz":540},"subject":"Repropose newly instance metadata/tag protection feature","message":"Repropose newly instance metadata/tag protection feature\n\nThis patch re-proposes the spec for the instance\nmetadata/tag protection feature by implementing\nthe policy enforcement using oslo.policy.\n\nChange-Id: Ic2cc959541d9fe371e46c6083eaec487d9aa90c7\nSigned-off-by: Koya Watanabe \u003ckoywatan@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d317e9c2062bc51014ddeeac5bada68e14dba6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d317e9c2062bc51014ddeeac5bada68e14dba6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
