)]}'
{"id":"openstack%2Fnova-specs~979696","triplet_id":"openstack%2Fnova-specs~master~Ib2bd6a5d781724a6f42be5295f2acf69a226a158","project":"openstack/nova-specs","branch":"master","topic":"bp/default-ssd-disk-rotation-rate","attention_set":{"38909":{"account":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"last_update":"2026-06-08 15:04:41.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-06-08 15:27:43.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"hashtags":[],"change_id":"Ib2bd6a5d781724a6f42be5295f2acf69a226a158","subject":"Spec for default SSD rotation rate on instance disks","status":"NEW","created":"2026-03-09 16:22:41.000000000","updated":"2026-07-03 13:41:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":39,"unresolved_comment_count":8,"has_review_started":true,"meta_rev_id":"0d92db499a454008d38ebca37d8c16f37a71dffc","_number":979696,"virtual_id_number":979696,"owner":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-08 15:04:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 16:54:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-09 16:54:50.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"5631827621a6bef36667005c7ba55176b988c70a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"date":"2026-03-09 16:22:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"564e3c3e5a27f819b2438e0885ff497a8d61ebd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"date":"2026-03-09 16:25:17.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7543f93f90f5ecda5cade04224ac890acaaaa9a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"date":"2026-03-09 16:42:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f86ef4ae8380bb1abc17c16704d1cfa5e3d0935e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 16:54:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/70f3e095ea5c4a97a37ab8728075d43f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d47a2d6d01c74063bc319b1301948f57 : FAILURE in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/735afb68f0f3478bb8ac7581320498b7 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":3},{"id":"f9d2eb59452f42a0a7f81bb226ec8db02b72592c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-03-09 16:54:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"732e355689fefad262f52be2f3aaec968d2741e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"date":"2026-03-09 17:01:50.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9dc6fb07a7ed13de4962040a5adf368daaeecf4f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"date":"2026-03-09 17:10:31.000000000","message":"Abandoned\n\nI think, is not the best path.","accounts_in_message":[],"_revision_number":4},{"id":"86a2f705b1e12f9dea368a3e66ee198b685222c1","tag":"autogenerated:gerrit:restore","author":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"date":"2026-06-04 14:03:26.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"a80db3057c4b4c9a54e57b36f78dc5956243d4bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 14:16:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd5e906a2452437aacf94aca5a05b1d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/762258a2eeda4c679bc534b11b145376 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed1c9c0b1f8d46c5a96a06006f2d6e20 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":4},{"id":"0bb06958b098caa8b262d17ec34ed6771cd0420f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"date":"2026-06-05 10:53:42.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"8abadacf967dd1ee3a0fb07238d97d4a6e0fde2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 11:05:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e54e1ac39ded44e0bac52d117170e002\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7916fb0153ce448faf007a780e435b8a : FAILURE in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e29b4c1c4aba4091934f40ac15c44b6a : FAILURE in 2m 19s","accounts_in_message":[],"_revision_number":5},{"id":"2c7244afcf779b21f766377074382965bb8635dd","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-05 11:43:57.000000000","message":"Patch Set 5: Code-Review-1\n\n(17 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7856691a315e684ab64f7aadfc50907b194af849","author":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"date":"2026-06-08 14:23:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7599c853acdb48093d555000a81faf7073884162","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-08 14:29:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d4637af545c3bdc2f056e41b00c6d52e991e25da","author":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"date":"2026-06-08 14:34:20.000000000","message":"Patch Set 5:\n\n(8 comments)","accounts_in_message":[],"_revision_number":5},{"id":"830e939a019250402bc6992f22b27d3ff11eb4b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"date":"2026-06-08 14:47:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"de3a7ed934ad173654012c195900322f07204691","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"date":"2026-06-08 14:49:33.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a39b100b4e7f8ea238fd02b3f3b64da5aba72711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 15:04:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e84b9b32d59a491ab528a7d83f0a4865\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61128aeed81a4549a1a9d536e5587226 : FAILURE in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93705b102d754188ab92ba29cec3b1e2 : FAILURE in 3m 51s","accounts_in_message":[],"_revision_number":7},{"id":"23ab0268b091fc715ab23f2df749350b23365121","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-08 15:27:43.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"0d92db499a454008d38ebca37d8c16f37a71dffc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-03 13:41:03.000000000","message":"Topic set to bp/default-ssd-disk-rotation-rate","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"1e2658f0dec174b0d5a3952f15c6496879edcc53","revisions":{"35fd4e6927ff2859cbcd321aecc8c3c443c59348":{"kind":"REWORK","_number":1,"created":"2026-03-09 16:22:41.000000000","uploader":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"ref":"refs/changes/96/979696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/96/979696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/1"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Thibaut Demaret","email":"thidem@worteks.com","date":"2026-03-09 16:10:56.000000000","tz":60},"committer":{"name":"Thibaut Demaret","email":"thidem@worteks.com","date":"2026-03-09 16:22:33.000000000","tz":60},"subject":"Spec for default SSD rotation rate on instance disks","message":"Spec for default SSD rotation rate on\ninstance disks\n\nChange-Id: Ib2bd6a5d781724a6f42be5295f2acf69a226a158\nSigned-off-by: Thibaut Demaret \u003cthidem@worteks.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/35fd4e6927ff2859cbcd321aecc8c3c443c59348"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/35fd4e6927ff2859cbcd321aecc8c3c443c59348"}]},"branch":"refs/heads/master"},"30060c107a34fa5ceeed5a57ea720b546e56c660":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-03-09 16:25:17.000000000","uploader":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"ref":"refs/changes/96/979696/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/96/979696/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/2"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Thibaut Demaret","email":"thidem@worteks.com","date":"2026-03-09 16:10:56.000000000","tz":60},"committer":{"name":"Thibaut Demaret","email":"thidem@worteks.com","date":"2026-03-09 16:23:50.000000000","tz":60},"subject":"Spec for default SSD rotation rate on instance disks","message":"Spec for default SSD rotation rate on\ninstance disks\n\nBlueprint: default-ssd-disk-rotation-rate\nChange-Id: Ib2bd6a5d781724a6f42be5295f2acf69a226a158\nSigned-off-by: Thibaut Demaret \u003cthidem@worteks.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/30060c107a34fa5ceeed5a57ea720b546e56c660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/30060c107a34fa5ceeed5a57ea720b546e56c660"}]},"branch":"refs/heads/master"},"9f78c8743e55ed596660ceac7ea427dbb5cfc558":{"kind":"REWORK","_number":3,"created":"2026-03-09 16:42:48.000000000","uploader":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"ref":"refs/changes/96/979696/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/96/979696/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/3"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Thibaut Demaret","email":"thidem@worteks.com","date":"2026-03-09 16:10:56.000000000","tz":60},"committer":{"name":"Thibaut Demaret","email":"thidem@worteks.com","date":"2026-03-09 16:42:37.000000000","tz":60},"subject":"Spec for default SSD rotation rate on instance disks","message":"Spec for default SSD rotation rate on\ninstance disks\n\nBlueprint: default-ssd-disk-rotation-rate\nChange-Id: Ib2bd6a5d781724a6f42be5295f2acf69a226a158\nSigned-off-by: Thibaut Demaret \u003cthidem@worteks.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9f78c8743e55ed596660ceac7ea427dbb5cfc558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9f78c8743e55ed596660ceac7ea427dbb5cfc558"}]},"branch":"refs/heads/master"},"95ebbec2d08384cba0c4c0e8461db4ff69f54589":{"kind":"REWORK","_number":4,"created":"2026-03-09 17:01:50.000000000","uploader":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"ref":"refs/changes/96/979696/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/96/979696/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/4"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Thibaut Demaret","email":"thidem@worteks.com","date":"2026-03-09 16:10:56.000000000","tz":60},"committer":{"name":"Thibaut Demaret","email":"thidem@worteks.com","date":"2026-03-09 17:01:41.000000000","tz":60},"subject":"Spec for default SSD rotation rate on instance disks","message":"Spec for default SSD rotation rate on\ninstance disks\n\nBlueprint: default-ssd-disk-rotation-rate\nChange-Id: Ib2bd6a5d781724a6f42be5295f2acf69a226a158\nSigned-off-by: Thibaut Demaret \u003cthidem@worteks.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/95ebbec2d08384cba0c4c0e8461db4ff69f54589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/95ebbec2d08384cba0c4c0e8461db4ff69f54589"}]},"branch":"refs/heads/master"},"3b531264117e03847032bb62af5c2bbd8ec5082a":{"kind":"REWORK","_number":5,"created":"2026-06-05 10:53:42.000000000","uploader":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"ref":"refs/changes/96/979696/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/96/979696/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/5"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Thibaut Demaret","email":"thidem@worteks.com","date":"2026-03-09 16:10:56.000000000","tz":60},"committer":{"name":"Thibaut Demaret","email":"thidem@worteks.com","date":"2026-06-05 10:53:29.000000000","tz":120},"subject":"Spec for default SSD rotation rate on instance disks","message":"Spec for default SSD rotation rate on\ninstance disks\n\nBlueprint: default-ssd-disk-rotation-rate\nChange-Id: Ib2bd6a5d781724a6f42be5295f2acf69a226a158\nSigned-off-by: Thibaut Demaret \u003cthidem@worteks.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3b531264117e03847032bb62af5c2bbd8ec5082a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3b531264117e03847032bb62af5c2bbd8ec5082a"}]},"branch":"refs/heads/master"},"867079d5e93f70eda2816a23f497c07167fbe716":{"kind":"REWORK","_number":6,"created":"2026-06-08 14:47:27.000000000","uploader":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"ref":"refs/changes/96/979696/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/96/979696/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/6"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Thibaut Demaret","email":"thidem@worteks.com","date":"2026-03-09 16:10:56.000000000","tz":60},"committer":{"name":"Thibaut Demaret","email":"thidem@worteks.com","date":"2026-06-08 14:46:22.000000000","tz":120},"subject":"Spec for default SSD rotation rate on instance disks","message":"Spec for default SSD rotation rate on\ninstance disks\n\nBlueprint: default-ssd-disk-rotation-rate\nChange-Id: Ib2bd6a5d781724a6f42be5295f2acf69a226a158\nSigned-off-by: Thibaut Demaret \u003cthidem@worteks.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/867079d5e93f70eda2816a23f497c07167fbe716"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/867079d5e93f70eda2816a23f497c07167fbe716"}]},"branch":"refs/heads/master"},"1e2658f0dec174b0d5a3952f15c6496879edcc53":{"kind":"REWORK","_number":7,"created":"2026-06-08 14:49:33.000000000","uploader":{"_account_id":38909,"name":"Thibaut Démaret","display_name":"Thibaut Démaret","email":"thidem@worteks.com","username":"worteks-thidem"},"ref":"refs/changes/96/979696/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/96/979696/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/96/979696/7"}}},"commit":{"parents":[{"commit":"6ef4b94da1e992f42ffa9915ed9582cde6257439","subject":"Create specs directory for 2026.2 Hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6ef4b94da1e992f42ffa9915ed9582cde6257439"}]}],"author":{"name":"Thibaut Demaret","email":"thidem@worteks.com","date":"2026-03-09 16:10:56.000000000","tz":60},"committer":{"name":"Thibaut Demaret","email":"thidem@worteks.com","date":"2026-06-08 14:49:21.000000000","tz":120},"subject":"Spec for default SSD rotation rate on instance disks","message":"Spec for default SSD rotation rate on\ninstance disks\n\nBlueprint: default-ssd-disk-rotation-rate\nChange-Id: Ib2bd6a5d781724a6f42be5295f2acf69a226a158\nSigned-off-by: Thibaut Demaret \u003cthidem@worteks.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1e2658f0dec174b0d5a3952f15c6496879edcc53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1e2658f0dec174b0d5a3952f15c6496879edcc53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
