)]}'
{"id":"openstack%2Fnova-specs~980166","triplet_id":"openstack%2Fnova-specs~master~Ia97b1bd84530857a7f45bfe4f988daf05eb5c002","project":"openstack/nova-specs","branch":"master","topic":"bp/vtpm-live-migration","hashtags":[],"change_id":"Ia97b1bd84530857a7f45bfe4f988daf05eb5c002","subject":"Propose vTPM live migration: \u0027deployment\u0027 mode","status":"ABANDONED","created":"2026-03-11 21:28:34.000000000","updated":"2026-05-19 17:39:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a5da95ec62821828815fc04e4cc6e8e50b5d1210","_number":980166,"virtual_id_number":980166,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-05 03:07:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-11 21:39:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"01d79cff0a4c444ed54ef1f2bee9ae19b2415e01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-03-11 21:28:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f77207801599761053c83d3480f63bbf80052fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-03-11 21:31:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ba915954bfaccb093c77b615bc71495e3049caea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 21:39:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c3a0dee39ae421fb13d95732aa1fe39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9f29bd8786e440582ef7ccfd95cbf3b : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a6f8dc709774597bae9ff5f588788d0 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2},{"id":"049098b74a08788821ac40bf55ed5607f573de7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-05-05 02:33:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"1d0f9c86958c9d2f5893f511e0a78bb3270c05e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 03:07:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0de004dcd1748f8aacea9edb5567239\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eaf05c8798544af862afdf46f6db67e : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f823ceb328f04ca9a50e192756de4b9b : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":3},{"id":"a5da95ec62821828815fc04e4cc6e8e50b5d1210","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-05-19 17:39:57.000000000","message":"Abandoned\n\nDecided as not a priority at this time, to be reconsidered if there is interest from cloud operators.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"92c374770e53dc5457d05840cad95af1490aa02b","revisions":{"fbdffc2657c203a87b373c78e2a436bd08f934c0":{"kind":"REWORK","_number":1,"created":"2026-03-11 21:28:34.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/66/980166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/66/980166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/980166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/980166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/980166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/66/980166/1"}}},"commit":{"parents":[{"commit":"e50d07dd2f9c3df610cbb432e82abf5711cef8d4","subject":"Amend vTPM live migration spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e50d07dd2f9c3df610cbb432e82abf5711cef8d4"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-03-11 21:24:08.000000000","tz":-420},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-03-11 21:28:01.000000000","tz":-420},"subject":"Propose vTPM live migration: \u0027deployment\u0027 mode","message":"Propose vTPM live migration: \u0027deployment\u0027 mode\n\nLast cycle 2026.1 Gazpacho we landed the \u0027host\u0027 mode for vTPM live\nmigration, so this spec is mostly a copy of the same background content\nwith the \u0027deployment\u0027 mode pieces added in to work on in 2026.2\nHibiscus.\n\nRelated to blueprint vtpm-live-migration-deployment-mode\n\nChange-Id: Ia97b1bd84530857a7f45bfe4f988daf05eb5c002\nSigned-off-by: melanie witt \u003cmelwittt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fbdffc2657c203a87b373c78e2a436bd08f934c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fbdffc2657c203a87b373c78e2a436bd08f934c0"}]},"branch":"refs/heads/master"},"feb99531aff524ae9fefaff7e81b01bdc2765e47":{"kind":"REWORK","_number":2,"created":"2026-03-11 21:31:52.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/66/980166/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/66/980166/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/980166/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/980166/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/980166/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/66/980166/2"}}},"commit":{"parents":[{"commit":"e50d07dd2f9c3df610cbb432e82abf5711cef8d4","subject":"Amend vTPM live migration spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e50d07dd2f9c3df610cbb432e82abf5711cef8d4"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-03-11 21:24:08.000000000","tz":-420},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-03-11 21:31:12.000000000","tz":-420},"subject":"Propose vTPM live migration: \u0027deployment\u0027 mode","message":"Propose vTPM live migration: \u0027deployment\u0027 mode\n\nLast cycle 2026.1 Gazpacho we landed the \u0027host\u0027 mode for vTPM live\nmigration, so this spec is mostly a copy of the same background content\nwith the \u0027deployment\u0027 mode pieces added in to work on in 2026.2\nHibiscus.\n\nRelated to blueprint vtpm-live-migration-deployment-mode\n\nChange-Id: Ia97b1bd84530857a7f45bfe4f988daf05eb5c002\nSigned-off-by: melanie witt \u003cmelwittt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/feb99531aff524ae9fefaff7e81b01bdc2765e47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/feb99531aff524ae9fefaff7e81b01bdc2765e47"}]},"branch":"refs/heads/master"},"92c374770e53dc5457d05840cad95af1490aa02b":{"kind":"REWORK","_number":3,"created":"2026-05-05 02:33:44.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/66/980166/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/66/980166/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/980166/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/980166/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/66/980166/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/66/980166/3"}}},"commit":{"parents":[{"commit":"4ec109fb8b33316943f4697759951463aad2c0f0","subject":"Propose vTPM live migration: \u0027user\u0027 mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4ec109fb8b33316943f4697759951463aad2c0f0"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-05-05 00:42:21.000000000","tz":-420},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-05-05 02:33:04.000000000","tz":-420},"subject":"Propose vTPM live migration: \u0027deployment\u0027 mode","message":"Propose vTPM live migration: \u0027deployment\u0027 mode\n\nThis spec proposes a third TPM secret security mode, \u0027deployment\u0027, in\nwhich the Nova service user owns the key manager secret. This enables\nlive migration and restart after host reboot without requiring the\ninstance owner\u0027s involvement or credentials.\n\nThis is a continuation of the vTPM live migration work that landed in\nthe 33.0.0 (2026.1 Gazpacho) release.\n\nRelated to blueprint vtpm-live-migration-deployment-mode\n\nAssisted-by: Claude claude-opus-4-6 claude-code-2.1.92\nChange-Id: Ia97b1bd84530857a7f45bfe4f988daf05eb5c002\nSigned-off-by: melanie witt \u003cmelwittt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/92c374770e53dc5457d05840cad95af1490aa02b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/92c374770e53dc5457d05840cad95af1490aa02b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
