)]}'
{"id":"openstack%2Fnova-specs~991636","triplet_id":"openstack%2Fnova-specs~master~I919a19890dedae7461d116774bb9290d63d2192a","project":"openstack/nova-specs","branch":"master","attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-06-11 23:22:55.000000000","reason":"\u003cGERRIT_ACCOUNT_35674\u003e replied on the change","reason_account":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-06-08 15:33:08.000000000","reason":"\u003cGERRIT_ACCOUNT_35674\u003e replied on the change","reason_account":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"}}},"removed_from_attention_set":{"35674":{"account":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"},"last_update":"2026-06-11 23:22:55.000000000","reason":"\u003cGERRIT_ACCOUNT_35674\u003e replied on the change","reason_account":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"}}},"hashtags":[],"change_id":"I919a19890dedae7461d116774bb9290d63d2192a","subject":"Spec for reduce failed builds value by time","status":"NEW","created":"2026-06-04 13:31:51.000000000","updated":"2026-06-11 23:22:55.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":19,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"eb1eee996f4f4aab54b1838d090a015ccadad2eb","_number":991636,"virtual_id_number":991636,"owner":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-05 10:55:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":1,"date":"2026-06-05 11:09:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":-1,"date":"2026-06-08 13:20:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-04 13:44:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-04 14:31:52.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-06-08 13:20:59.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"4ab94b110cd5b94bb6583a570e045289113a2ff5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"},"date":"2026-06-04 13:31:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8052642e2121e9a855a3999f2febf02bad7bd642","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"},"date":"2026-06-04 13:37:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d26894f7daa5c80d4e2f98b75dfbb167a22f0be9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 13:44:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0962d90a0f3948ea99d69ed2926acbaf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/083becfc9bc04a838b174f0798965fea : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62574bfc5a9948e2ab3eec612e14cd39 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"aafe2bd2999fcccc1574c3f325bcce89e5478659","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-04 14:31:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"66ec09775040f76446383bd500f358794ace6ae3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-04 14:32:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5d109444ca47f417f7a75f331f6e5c53cb4ac982","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"},"date":"2026-06-05 10:21:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"96bde7cd141c90e75d9859801f8c3c4c7e13f3f9","author":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"},"date":"2026-06-05 10:27:38.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"962a0ecd8ffa49e7c3b8d34c95c5a60b59c51b55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 10:41:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4ee15a4dbf349f6b64e8c227a54adaa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d4eaa240bbe4bad84ad3d26fc0947e6 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af0a59072dac4afd92e689bd3b74016b : FAILURE in 2m 29s","accounts_in_message":[],"_revision_number":3},{"id":"f0bd144fb4eab8d4d72af436b2b8119b29240d3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"},"date":"2026-06-05 10:49:29.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"da49c0c8f3123bb3340cbd19a03d8cbb6588c0da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 10:55:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c20a4197dd504906933a950804100bfc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e8d7b9dfd1477f95698e289c5e7e33 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f63a0186e6b464883fc1b703926e43f : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":4},{"id":"0cad97405216f2cb6c77dd8661ba997ac5d93f41","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-05 11:09:23.000000000","message":"Patch Set 4: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4595caa611db91aea4fbf8aea4a2b39d217ec41e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-06-08 13:20:59.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"871098596409ac67dbb30a2f1cec56857813e45f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-08 14:27:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b99df2453013fc87a58ce982c9a04579f9b41452","author":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"},"date":"2026-06-08 15:33:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b1b1368e2622d67dfe91d9014eec6a911e272222","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-11 09:40:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"486ca4d73c9237d2af94bdefef5b4229e4197fac","author":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"},"date":"2026-06-11 09:58:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5439b6ee636d13d366ce9056a306fc2dd4ab590f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-11 15:36:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eb1eee996f4f4aab54b1838d090a015ccadad2eb","author":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"},"date":"2026-06-11 23:22:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f8fdcafb23def136b805c9b2016a3e51d11fceba","revisions":{"779cdfec68ad5f4bd449bae8a798ab2400f70c6b":{"kind":"REWORK","_number":1,"created":"2026-06-04 13:31:51.000000000","uploader":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"},"ref":"refs/changes/36/991636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/991636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/1"}}},"commit":{"parents":[{"commit":"4f1cdc096c49400fd1e6579559fe5dae8da5d202","subject":"Merge \"libvirt: AMD SEV-SNP support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4f1cdc096c49400fd1e6579559fe5dae8da5d202"}]}],"author":{"name":"cw0306-lee","email":"cw0306.lee@samsung.com","date":"2026-06-04 13:25:37.000000000","tz":540},"committer":{"name":"cw0306-lee","email":"cw0306.lee@samsung.com","date":"2026-06-04 13:31:37.000000000","tz":540},"subject":"Spec for reduce failed builds value by time","message":"Spec for reduce failed builds value by time\n\nChange-Id: I919a19890dedae7461d116774bb9290d63d2192a\nSigned-off-by: cw0306-lee \u003ccw0306.lee@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/779cdfec68ad5f4bd449bae8a798ab2400f70c6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/779cdfec68ad5f4bd449bae8a798ab2400f70c6b"}]},"branch":"refs/heads/master"},"61fc2c2bdc4e4f79c97e34f44c01d737c164b4d8":{"kind":"REWORK","_number":2,"created":"2026-06-04 13:37:57.000000000","uploader":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"},"ref":"refs/changes/36/991636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/991636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/2"}}},"commit":{"parents":[{"commit":"4f1cdc096c49400fd1e6579559fe5dae8da5d202","subject":"Merge \"libvirt: AMD SEV-SNP support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4f1cdc096c49400fd1e6579559fe5dae8da5d202"}]}],"author":{"name":"cw0306-lee","email":"cw0306.lee@samsung.com","date":"2026-06-04 13:25:37.000000000","tz":540},"committer":{"name":"cw0306-lee","email":"cw0306.lee@samsung.com","date":"2026-06-04 13:37:45.000000000","tz":540},"subject":"Spec for reduce failed builds value by time","message":"Spec for reduce failed builds value by time\n\nChange-Id: I919a19890dedae7461d116774bb9290d63d2192a\nSigned-off-by: cw0306-lee \u003ccw0306.lee@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/61fc2c2bdc4e4f79c97e34f44c01d737c164b4d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/61fc2c2bdc4e4f79c97e34f44c01d737c164b4d8"}]},"branch":"refs/heads/master"},"8f276d015defeaed17bfc1784690235ea5a894ef":{"kind":"REWORK","_number":3,"created":"2026-06-05 10:21:34.000000000","uploader":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"},"ref":"refs/changes/36/991636/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/991636/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/3"}}},"commit":{"parents":[{"commit":"4f1cdc096c49400fd1e6579559fe5dae8da5d202","subject":"Merge \"libvirt: AMD SEV-SNP support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4f1cdc096c49400fd1e6579559fe5dae8da5d202"}]}],"author":{"name":"cw0306-lee","email":"cw0306.lee@samsung.com","date":"2026-06-04 13:25:37.000000000","tz":540},"committer":{"name":"cw0306-lee","email":"cw0306.lee@samsung.com","date":"2026-06-05 10:21:06.000000000","tz":540},"subject":"Spec for reduce failed builds value by time","message":"Spec for reduce failed builds value by time\n\nChange-Id: I919a19890dedae7461d116774bb9290d63d2192a\nSigned-off-by: cw0306-lee \u003ccw0306.lee@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8f276d015defeaed17bfc1784690235ea5a894ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8f276d015defeaed17bfc1784690235ea5a894ef"}]},"branch":"refs/heads/master"},"f8fdcafb23def136b805c9b2016a3e51d11fceba":{"kind":"REWORK","_number":4,"created":"2026-06-05 10:49:29.000000000","uploader":{"_account_id":35674,"name":"ChungWon Lee","display_name":"cw0306-lee","email":"cw0306.lee@samsung.com","username":"cw0306-lee"},"ref":"refs/changes/36/991636/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/991636/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/991636/4"}}},"commit":{"parents":[{"commit":"4f1cdc096c49400fd1e6579559fe5dae8da5d202","subject":"Merge \"libvirt: AMD SEV-SNP support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4f1cdc096c49400fd1e6579559fe5dae8da5d202"}]}],"author":{"name":"cw0306-lee","email":"cw0306.lee@samsung.com","date":"2026-06-04 13:25:37.000000000","tz":540},"committer":{"name":"cw0306-lee","email":"cw0306.lee@samsung.com","date":"2026-06-05 10:49:03.000000000","tz":540},"subject":"Spec for reduce failed builds value by time","message":"Spec for reduce failed builds value by time\n\nChange-Id: I919a19890dedae7461d116774bb9290d63d2192a\nSigned-off-by: cw0306-lee \u003ccw0306.lee@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f8fdcafb23def136b805c9b2016a3e51d11fceba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f8fdcafb23def136b805c9b2016a3e51d11fceba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
