)]}'
{"id":"openstack%2Fnova~1000081","triplet_id":"openstack%2Fnova~master~I1b3be45b42b41f28b1616834f9538fccff4e5fe6","project":"openstack/nova","branch":"master","topic":"bp/intel-tdx-libvirt-support","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-08-24 15:03:31.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"38744":{"account":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"last_update":"2026-08-26 17:31:00.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-08-26 13:40:14.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-08-22 03:03:03.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I1b3be45b42b41f28b1616834f9538fccff4e5fe6","subject":"doc: Add documentation for Intel TDX","status":"MERGED","created":"2026-08-07 10:51:39.000000000","updated":"2026-08-26 17:33:10.000000000","submitted":"2026-08-26 17:31:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":57,"unresolved_comment_count":10,"has_review_started":true,"submission_id":"1000081-bp/intel-tdx-libvirt-support","meta_rev_id":"e983569145f57ebd01c5b7e8796af20dee3d1f1d","_number":1000081,"virtual_id_number":1000081,"owner":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 17:31:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-08-24 15:03:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2026-08-22 03:03:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-08-24 15:03:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-07 11:07:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-07 11:25:22.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-07 13:22:31.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-08-11 14:32:29.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-08-18 22:48:59.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-08-21 03:19:18.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"ad0ff239359702f9937f0320bea49506e25eed93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-07 10:51:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c4fa746450738eee61bbfe6fee8c0ed347a5cc1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-07 10:52:15.000000000","message":"Topic set to bp/intel-tdx-libvirt-support","accounts_in_message":[],"_revision_number":1},{"id":"eec84b82901ca0cd222464ec599d3df7aed13951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 11:07:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ecc901db8ee74a10bc85cd98c39037a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/887787c049c74fe597cb246e3a6e02ff : SUCCESS in 8m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fdb795c15784d2e926b989a129570be : FAILURE in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7effb4f628b647bda2099c4a11d6cc43 : SUCCESS in 4m 58s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9d92798060ca4cff878d860da69542cc : SUCCESS in 6m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a154ca042276c0faf2d2a944097580140645338","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-07 11:25:22.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1b3be45b42b41f28b1616834f9538fccff4e5fe6 .","accounts_in_message":[],"_revision_number":1},{"id":"6f5932f01b52e17c1d02423c70ac744139631a34","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-07 13:22:31.000000000","message":"Patch Set 1: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"27ebec4042a0f2cd9fcc54f511ee6cd0319cd5fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-11 14:11:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"5634fc620c4f6da2e05d57bb3d1db67b2caa97ad","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-11 14:23:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed65b92ccc2873bf11aa5e92da9c3f93c413f7d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 14:32:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da87fd0a29bc4b6c9a369d67d3d2c7de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c951714a8c0d469884675f9d87846be5 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/257997207b3a4fda884e678db347f96e : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bfdbc8b7a884ed5b73d9287a30d3625 : SUCCESS in 8m 37s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bc0585b536be4408951650103fbff7cd : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"035165b43adadbeb04a714c737665f81a289280e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-11 14:42:52.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"f2a3db017531ce08497ba7f7679fd918e95c56dc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-11 15:09:23.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1b3be45b42b41f28b1616834f9538fccff4e5fe6 .","accounts_in_message":[],"_revision_number":2},{"id":"f4b9ace918a9c81ccb2b6da4764b3f075705fbe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-13 07:25:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"16c1705d65638fb3f0d9592db4240575693244e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 07:39:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b095c554eb942e6b74fcb6683dc7199\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fecf273aa8e14d4d85d92aee6724a409 : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3240e8b7db3c44d99f00ca67003eeb81 : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df0815e67494448080795e79b11a175e : SUCCESS in 8m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a58c6b5c07464493b12323b53e39c6db : SUCCESS in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1770e58ec1281108a27ed0ccf1e8a735e60f354a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-13 07:50:24.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"fcb45486c4f97dec786b89f85d6432f56d7df770","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-13 08:26:20.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1b3be45b42b41f28b1616834f9538fccff4e5fe6 .","accounts_in_message":[],"_revision_number":3},{"id":"eedd204522582c2fabd0dd881ac35b378b75b61f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-13 13:07:33.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7147d0b6958b30c2c186498edcb7c2518600881d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-13 13:50:03.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c2d75c6ede7d0f60384cc7266d44fe4c5ed18ac9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-13 14:29:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"a77e4fa0c4d9c8e64548668ac4c1e3f9b32eaa1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 14:54:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a67da2f942dd418caa5e72a18817fd4a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/740928ff155043e5bdf89d9a5c17a06c : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ef280d54dd48c4adf4584df8717a6e : SUCCESS in 13m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/120415127bf84da1ab2c07f431324a18 : SUCCESS in 4m 23s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/edfadad769d6426ba0b568b4b3543cf7 : SUCCESS in 5m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2c5489adecdfc41cb4c658477aa8552ce4e74c94","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-13 15:04:57.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"d0baadc44dc1e5520caed432e6adcc6c87e70d8c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-13 15:46:26.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1b3be45b42b41f28b1616834f9538fccff4e5fe6 .","accounts_in_message":[],"_revision_number":4},{"id":"63bbeed50e4883be06c8c2ac8dc08ef5b432009f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-14 08:34:23.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f6b7811ef97b5953d27c6a652243033c71d0f9cf","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-18 22:48:59.000000000","message":"Patch Set 4: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d36e20b925972d3079a59c7cd9d310c254d6b09b","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-19 08:27:16.000000000","message":"Patch Set 4:\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"13df8492b326f2e3ecc74da5cc651171fdb7bc5b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-21 03:19:18.000000000","message":"Patch Set 4: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d54df6a3512fd497c1d14ed6f62d37091eca5d9f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-21 13:00:51.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4f58a15a5c5be1373d2adc195831275602ee724d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-21 13:41:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"884edaffa550c6f534a41839ec3c344cddd5dab7","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-21 13:44:46.000000000","message":"Patch Set 5:\n\n(9 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c941466de7aefb003ad355f52c1ce30400f6c3d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 13:52:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8283e38c38fc4b9083f94e07b3932292\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ad8aff4b37041888ef394a1e2b36990 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecec201c38eb47f5ba53cefb65706d60 : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2095a480dbdd41fc9a00ecd4af52ef62 : SUCCESS in 5m 10s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/0b486d51395d4e1dadebc1ca90d1f7d8 : SUCCESS in 4m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b09a8ab0fb2411422c9dc29cbfce7472c2631bab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-21 14:02:57.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"b8bbaf52033b943fa6e5ea38922a701c2737a303","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 14:47:30.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1b3be45b42b41f28b1616834f9538fccff4e5fe6 .","accounts_in_message":[],"_revision_number":5},{"id":"acedfbba3924666ae63b5598e34b9b3d0c21b6d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-21 14:56:39.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"542046038da9f50bab69dc4ca6e07dd9c0a017de","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-21 14:59:49.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7f5a7ebf155dfa72d65f3a1f9d07d1b9c993f459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 15:16:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94bafe362e0049a4abd54ff085dadf2a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb1e42965fbc442c92e4602064c87557 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d84072a454d44219c7ced1fc53c79aa : SUCCESS in 13m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4bf29c6605c4189934be26a809b2609 : SUCCESS in 8m 15s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/56fd1b7859794428a225c050115b6712 : SUCCESS in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fcb4205f6abc94ab69a498fd33370361da2a94ac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-21 15:27:05.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"77293b6a5f189e0c39749277fc3695de957ea621","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 16:43:12.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1b3be45b42b41f28b1616834f9538fccff4e5fe6 .","accounts_in_message":[],"_revision_number":6},{"id":"d9618f0d7116ec31daa1a31ab5118d5d3fb4e4ad","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-22 03:03:03.000000000","message":"Patch Set 6: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"a18b2cc1e942ee40bc997b62ad6e3a64a59bef2c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-24 15:03:31.000000000","message":"Patch Set 6: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"5add2f5bd6eb13e9a145ce9d7b56f12a053fd353","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-24 15:03:36.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"515bd929f8a43096d7ab8e993e7639e00d94f168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 20:08:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"68f988a2c21bdc43eb87a1cae35c50f5ac72acb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 21:45:40.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/999249 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"74281d2039eacf479450e7393ed77aeaa9818525","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-26 13:40:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3e68e3715e8122169e3575d3d34ca11315d7c3fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 14:00:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfa3df910ed54b95a4a56a209e323602\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b42f64b60a84966a8d536b1b170dd52 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3ce892b8f7348dbaf2718106b61706a : SUCCESS in 13m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69bcadcad0ea412c91207b84f1407579 : SUCCESS in 8m 51s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/58307e3e382f4f72834a08aa36c27ca7 : SUCCESS in 5m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dbd4fa5e56de9261bedf36de3db5a393c9118930","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-26 14:11:33.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"cd0c3dc4e3bfcc1a97e347a2133f9d25211f2db8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 15:39:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2a721652c17068482956decf4963a03f2308863e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 17:31:00.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8b7260bb1b44cf3917add10a5018140\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aab40f185dd14433b3d925ef3bd009ec : SUCCESS in 9m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93a449620acc490899cfe8526cd579bf : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4da4a419c9a4fd28a0ccd3fe45d121f : SUCCESS in 9m 08s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6f113b08ac66422897b28b0565ea3585 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":6},{"id":"9ccec517b3f2c8b825d8e1238c6e18efbc34b703","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 17:31:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"e983569145f57ebd01c5b7e8796af20dee3d1f1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 17:33:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a3c37170c2a4bec86e80b01298ed589\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ebfeb35e3947619bbb3a5da46fd621 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12aac7c3c63c4a1f8867b11068174b0c : SUCCESS in 47s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"51928db00915238e60df98b2cb0eb4c296a6f273","revisions":{"f7b9ec3119015d6b1dea8ac1ebb3c61435a62d08":{"kind":"REWORK","_number":1,"created":"2026-08-07 10:51:39.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/81/1000081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/1000081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/1000081/1"}}},"commit":{"parents":[{"commit":"60586a4872b8b3877fdeeb9baaac860ee6e80eb4","subject":"libvirt: Launch security Intel TDX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/60586a4872b8b3877fdeeb9baaac860ee6e80eb4"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-04 13:57:46.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-07 10:48:47.000000000","tz":120},"subject":"doc: Add documentation for Intel TDX","message":"doc: Add documentation for Intel TDX\n\nDocumentation for Intel TDX including operator and user instructions, as\nwell as limitations and a section on attestation. Release notes are also\nadded.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: I1b3be45b42b41f28b1616834f9538fccff4e5fe6\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f7b9ec3119015d6b1dea8ac1ebb3c61435a62d08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f7b9ec3119015d6b1dea8ac1ebb3c61435a62d08"}]},"branch":"refs/heads/master"},"6b8a363924551fb070bbaa674d08d78094360714":{"kind":"REWORK","_number":2,"created":"2026-08-11 14:11:22.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/81/1000081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/1000081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/1000081/2"}}},"commit":{"parents":[{"commit":"60586a4872b8b3877fdeeb9baaac860ee6e80eb4","subject":"libvirt: Launch security Intel TDX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/60586a4872b8b3877fdeeb9baaac860ee6e80eb4"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-04 13:57:46.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-11 09:12:24.000000000","tz":-420},"subject":"doc: Add documentation for Intel TDX","message":"doc: Add documentation for Intel TDX\n\nDocumentation for Intel TDX including operator and user instructions, as\nwell as limitations and a section on attestation. Release notes are also\nadded.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: I1b3be45b42b41f28b1616834f9538fccff4e5fe6\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6b8a363924551fb070bbaa674d08d78094360714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6b8a363924551fb070bbaa674d08d78094360714"}]},"branch":"refs/heads/master"},"b050013ded02e626067b438221c19c78cc95a85a":{"kind":"REWORK","_number":3,"created":"2026-08-13 07:25:43.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/81/1000081/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/1000081/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/1000081/3"}}},"commit":{"parents":[{"commit":"2a44070da4a9111d01d5de11c06f62f477dcd72f","subject":"libvirt: Launch security Intel TDX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a44070da4a9111d01d5de11c06f62f477dcd72f"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-04 13:57:46.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-12 20:16:02.000000000","tz":-420},"subject":"doc: Add documentation for Intel TDX","message":"doc: Add documentation for Intel TDX\n\nDocumentation for Intel TDX including operator and user instructions, as\nwell as limitations and a section on attestation. Release notes are also\nadded.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: I1b3be45b42b41f28b1616834f9538fccff4e5fe6\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b050013ded02e626067b438221c19c78cc95a85a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b050013ded02e626067b438221c19c78cc95a85a"}]},"branch":"refs/heads/master"},"432ec45fd18f090d0b2ff8a29fba937b026b058a":{"kind":"REWORK","_number":4,"created":"2026-08-13 14:29:05.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/81/1000081/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/1000081/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/1000081/4"}}},"commit":{"parents":[{"commit":"2a44070da4a9111d01d5de11c06f62f477dcd72f","subject":"libvirt: Launch security Intel TDX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a44070da4a9111d01d5de11c06f62f477dcd72f"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-04 13:57:46.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-13 14:25:48.000000000","tz":-420},"subject":"doc: Add documentation for Intel TDX","message":"doc: Add documentation for Intel TDX\n\nDocumentation for Intel TDX including operator and user instructions, as\nwell as limitations and a section on attestation. Release notes are also\nadded.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: I1b3be45b42b41f28b1616834f9538fccff4e5fe6\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/432ec45fd18f090d0b2ff8a29fba937b026b058a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/432ec45fd18f090d0b2ff8a29fba937b026b058a"}]},"branch":"refs/heads/master"},"7bf4f38c8aa719cca7eb12d859873d8c510deb1b":{"kind":"REWORK","_number":5,"created":"2026-08-21 13:41:14.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/81/1000081/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/1000081/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/1000081/5"}}},"commit":{"parents":[{"commit":"5ff7d15e4716c2baa8abc831a0ee2910fd4bed75","subject":"libvirt: Launch security Intel TDX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ff7d15e4716c2baa8abc831a0ee2910fd4bed75"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-04 13:57:46.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-21 13:38:05.000000000","tz":-420},"subject":"doc: Add documentation for Intel TDX","message":"doc: Add documentation for Intel TDX\n\nDocumentation for Intel TDX including operator and user instructions, as\nwell as limitations and a section on attestation. Release notes are also\nadded.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: I1b3be45b42b41f28b1616834f9538fccff4e5fe6\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7bf4f38c8aa719cca7eb12d859873d8c510deb1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7bf4f38c8aa719cca7eb12d859873d8c510deb1b"}]},"branch":"refs/heads/master"},"51928db00915238e60df98b2cb0eb4c296a6f273":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-08-21 14:56:39.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/81/1000081/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/1000081/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1000081/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/1000081/6"}}},"commit":{"parents":[{"commit":"6f91fb725264fc8b328d3ccf66c088b236397e74","subject":"libvirt: Launch security Intel TDX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6f91fb725264fc8b328d3ccf66c088b236397e74"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-04 13:57:46.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-21 14:48:25.000000000","tz":-420},"subject":"doc: Add documentation for Intel TDX","message":"doc: Add documentation for Intel TDX\n\nDocumentation for Intel TDX including operator and user instructions, as\nwell as limitations and a section on attestation. Release notes are also\nadded.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: I1b3be45b42b41f28b1616834f9538fccff4e5fe6\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/51928db00915238e60df98b2cb0eb4c296a6f273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/51928db00915238e60df98b2cb0eb4c296a6f273"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
