)]}'
{"id":"openstack%2Fnova~1000211","triplet_id":"openstack%2Fnova~master~I59a3091d53e3e6ab990e80ed3974b2153569b2bf","project":"openstack/nova","branch":"master","topic":"bp/nova-services-graceful-shutdown-part2-task-tracking","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2026-08-28 18:15:30.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-08-28 18:15:30.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-08-27 08:22:56.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}},"hashtags":[],"change_id":"I59a3091d53e3e6ab990e80ed3974b2153569b2bf","subject":"Shutdown thread pool executors during graceful shutdown","status":"MERGED","created":"2026-08-08 20:02:22.000000000","updated":"2026-08-28 18:16:41.000000000","submitted":"2026-08-28 18:15:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"1000211-bp/nova-services-graceful-shutdown-part2-task-tracking","meta_rev_id":"af7fc47bd185b875cf12fae09668a58c17b7cb27","_number":1000211,"virtual_id_number":1000211,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-28 18:15:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2026-08-27 03:02:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2026-08-26 00:24:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2026-08-26 12:30:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2026-08-27 03:02:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2026-08-28 14:48:31.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-08 20:47:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-08 21:00:23.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-08 22:17:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-09 01:05:14.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-08-14 15:35:37.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2026-08-21 14:57:38.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2026-08-25 14:08:06.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-08-26 00:24:44.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-08-27 03:02:35.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"}],"messages":[{"id":"441150b4cd543a5c5eaeb544ab8ff9ec598a980c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-08 20:02:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8814067bb6508abfa8cd55849a6c04c7f32b551","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-08 20:02:34.000000000","message":"Topic gs5 removed","accounts_in_message":[],"_revision_number":1},{"id":"16f816a6db58f14a0eb365be1e54f03a74d569d4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-08 20:02:41.000000000","message":"Topic set to bp/nova-services-graceful-shutdown-part2-task-tracking","accounts_in_message":[],"_revision_number":1},{"id":"0a02c65e15c767d6fc00d967ebf19b36f289b19a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 20:47:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d45a1585c8584bc986e8309408f2b684\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/98cc25a605994f64b575f04782c7d623 : SUCCESS in 42m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3bd2f12ec0c14c058c8926ac716a17a1 : SUCCESS in 24m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fa7536bf86da4a71b9b80b9b3ec6e97b : SUCCESS in 28m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a6021a6dee168547f2336dd61915bd7835b3112","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-08 21:00:23.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I59a3091d53e3e6ab990e80ed3974b2153569b2bf .","accounts_in_message":[],"_revision_number":1},{"id":"c41bba63a3f2fe523cc77503378589629d9379bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 22:17:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d9a28e7ce7f4b3794daa5ffe539bbbc\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cfd3423037ee4e7294d60401679b8d86 : SUCCESS in 58m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0623c236ea1843b7859f7339bb1f961a : SUCCESS in 1h 22m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/203e7cf400b949cfa9e686222a5010ce : FAILURE in 58m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b67b245b5fed426b8cdf3031dbfbaade : SUCCESS in 1h 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44ce98bad530409083f3821dbc8a0074 : SUCCESS in 7m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02a540b2e1b2453c85ded2dda1d7da21 : SUCCESS in 13m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c699ae062f294e299e80b267f4eea435 : SUCCESS in 13m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a300ac6c483a4d0db95576e4e98740cb : SUCCESS in 14m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87a824454a7d4aff995a839e5dde1234 : SUCCESS in 10m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1776848f1c5d43c98bc1bb7272363827 : SUCCESS in 8m 01s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/f8e8573b0b9d4fbcb98963e5472197ee : SUCCESS in 1h 00m 56s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/15d0429a5f50444d8966d52446b578ee : SUCCESS in 1h 21m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b7d446a0173f45a5908b170ad62718f9 : SUCCESS in 47m 50s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/08b8c0af49184fd1ba9c97546d3671c0 : SUCCESS in 35m 44s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/941c034a87eb402ca8e401ff64395d0a : SUCCESS in 58m 57s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b6edac3fbc504f488cfee1477c93f2f8 : SUCCESS in 48m 09s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9723546142cf4e80a84d13a97191f996 : SUCCESS in 1h 44m 25s\n- nova-next https://zuul.opendev.org/t/openstack/build/746bd5be79e848f18bcb13392a83e265 : SUCCESS in 2h 07m 47s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/c6e139506cdd4aafa349507d55b9ccf2 : SUCCESS in 48m 12s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f7b3baec77694c1698db439e4626d586 : SUCCESS in 6m 19s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/75fed233e4104b01a8308f1b124c1db7 : SUCCESS in 27m 29s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f446edc83b0d4e3ab1d50acb14c8ce92 : SUCCESS in 24m 08s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/99af44e01fbf4bcfb0f4961dcc6b3ce3 : SUCCESS in 33m 03s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/21511df7c5e64e2d91977f673a507b05 : SUCCESS in 1h 23m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88c8269135b343dc931961079d105aa0 : SUCCESS in 56m 00s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/bab9eb9d12a44304b61ca4ade4708442 : SUCCESS in 39m 56s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/991d245e786548f3a6f3ff4132324c4d : SUCCESS in 34m 16s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d1fc5940d02d426388590f0709c14b0b : SUCCESS in 11m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5d6e36f552f4483de993ee4dd3a97f5e8becfe76","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-08 23:04:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7bf36aba192653049ea9a8140673320e6151fd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 23:47:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddddc888e8014ee7815685efdd62c7cc\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8e49a082cd8e4ff98cde5f674b7a2bb7 : SUCCESS in 18m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/916c8f976253457fa0d0f9463dc519dc : SUCCESS in 24m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7c6df6de575e4b82978059491fd775d3 : SUCCESS in 40m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13a20707f8c706cb57b60084012b119d8ed928a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-09 01:05:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12eb9354b6024f9ab372d170e41a6d67\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb06c817a31c4b5e92b1e7f50f62ab81 : SUCCESS in 45m 45s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f04683d1d95f4b85b524558ecea12931 : SUCCESS in 1h 03m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a361ee4fe98e496cae83e21eab3e3cad : SUCCESS in 41m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbaecaf5732843b0834f7fe139de8e4e : SUCCESS in 1h 23m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efe1f12153ff449eabada2c6ba89b15c : SUCCESS in 6m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/289289a559284e4c996388997e0094e3 : SUCCESS in 13m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c493eb7582744570b7abd3d531eb1e45 : SUCCESS in 13m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/92a4385da0f44b3cb33de4586b9b6f06 : SUCCESS in 18m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f2bcb5ddab46848310415875549bd5 : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0b54c59ca454220ae364bc70d3a57a2 : SUCCESS in 7m 44s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/9e09896fa85a4f80b0e5abc011a5d055 : SUCCESS in 53m 26s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/53de5f65072c46179ed0c5c04a6e8bdd : SUCCESS in 1h 38m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/23191ebe47a145a49162279abe9727e8 : SUCCESS in 50m 18s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/15d0b1cae4de4653b7162daaa41edc18 : SUCCESS in 35m 52s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/3efd39a463e74a9aa49362caf2d66b51 : SUCCESS in 47m 55s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f49578333d124a059fec71b2a30d14b4 : SUCCESS in 48m 01s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/de137e90d989431688cd8dc80825d6dd : SUCCESS in 1h 46m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/fb7961e190334f45b2a601b6624a8fce : SUCCESS in 1h 54m 15s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/c06aa4e191394b238a31e02c176aef6c : SUCCESS in 32m 57s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b94e70511bb5413d935cf39d31f67257 : SUCCESS in 5m 05s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/87e00d4d841b4455969d73ec40c1d871 : SUCCESS in 25m 18s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f51bdcc66d194debb7ef0475985f8ca9 : SUCCESS in 25m 40s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b3e514ad73c443f3a37dbbf2b86a78c4 : FAILURE in 28m 14s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/55e772df84ae41d8b692c7d0570515a1 : SUCCESS in 1h 21m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e64a3b5f2f9b425f974bf37f4327abc9 : SUCCESS in 59m 24s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/852bde5ac7dd4286a0004c345dfc10c9 : SUCCESS in 36m 41s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/71f37f95864f4114b55245af916a1b8f : SUCCESS in 31m 13s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/627e2656c2794186bfd6dd9744b4a484 : SUCCESS in 12m 53s","accounts_in_message":[],"_revision_number":1},{"id":"eaf5656df53e18de9bf6cb08da68858a33941514","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-09 01:15:36.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"9996b42909861660850a8d978b23329751d7db79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-11 04:14:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"08ade467de9619c354a03b285f29cb25c5fc392b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-11 04:27:24.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c51804d932c78554068fbecaf4f062fddd5c9acc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-11 05:19:29.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I59a3091d53e3e6ab990e80ed3974b2153569b2bf .","accounts_in_message":[],"_revision_number":2},{"id":"22ff951e2247fe659cc3b454fca4f317bec30e69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 05:42:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9564d836fd254ddab8d0a0eb19f1c3dc\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/283b97265c6149cd8ef37cc3ae7fe4ed : SUCCESS in 25m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/89bfa8375a394199842db53966a4731d : SUCCESS in 21m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b3c903776b5e49e8b5097667600dcc90 : SUCCESS in 35m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"29b6e945eb4f6bd7e12363a39fcd4ab9def2bcd1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-11 05:51:08.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I59a3091d53e3e6ab990e80ed3974b2153569b2bf .","accounts_in_message":[],"_revision_number":3},{"id":"99c115a106257359e48524fa3ff9296d04d9882a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 07:01:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54413dbe29d54e5d861d4a203e34314c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9d6aec5665ef471388efb3811a672f27 : SUCCESS in 1h 04m 16s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/03a19fab2be847b4a20cbf422c81a9e4 : SUCCESS in 1h 45m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/add98c6815594b2cb916b866928d2cdf : SUCCESS in 28m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80c2703de903414a9cf6a1ee3374bc45 : SUCCESS in 1h 18m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3efd6477d3214bc7b9cb8391f1b3fc98 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25444722d47b447083a33d9be93b1c87 : SUCCESS in 13m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5297b545e0c440dc891e450115c1f9ac : SUCCESS in 8m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e7eb2bf333d847c0aa00375cce77eedf : SUCCESS in 16m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30dd0456bbe2495ba3298139f8f7f6ba : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b697d109f35d46eb864e5090490a4c24 : SUCCESS in 4m 32s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/3418ee1cd78b464d9502ab8dcaf4db91 : SUCCESS in 35m 36s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e08314dd92d84894863b1dcf2119804d : SUCCESS in 1h 24m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b6a0b744e9cf4cf79a5f983977f7666e : SUCCESS in 52m 35s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/1927aeee5ed04cbd91136ded43ac1626 : SUCCESS in 34m 53s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/45b3142ea15b4e4e8e7bd6d6eb2761ca : SUCCESS in 51m 08s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0a462bde25964cc0ad6c43751b171b6a : SUCCESS in 50m 21s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b4d2852b39b44b608cd18aa4bba1e157 : SUCCESS in 1h 41m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/0b298257883841829f250f83ceced79d : SUCCESS in 2h 27m 08s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/cf5d421021564b0db2de08d8deb5ae32 : SUCCESS in 26m 15s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c0c0524fc3d646df8f8036d5cd8d012e : SUCCESS in 5m 29s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1d1600aeb0e5477691b063256ff7c883 : SUCCESS in 27m 41s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b7aa235d407c4998976acf111f7fbf17 : SUCCESS in 31m 49s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/ec6c95c46cb84d9db2950f86d91831dc : SUCCESS in 30m 53s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5664572a74f04ed290b98c90ce23e631 : SUCCESS in 1h 28m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a68f5f196de47d58a1fcde2449c673f : SUCCESS in 57m 59s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0c3fd4aea6a64023abd99139d39b0912 : SUCCESS in 51m 09s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/28b37ebd26174fcbb55c2159e21c8cd6 : SUCCESS in 27m 23s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/2da6bafef3464f82ad18dcd1b01748fc : SUCCESS in 13m 21s","accounts_in_message":[],"_revision_number":3},{"id":"3e229aa90c3c989fe835fda251faed9fe7c876dd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-11 09:04:22.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"c2751e1d79a80437b33094da88bbb8d6a7d92fe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-13 04:22:05.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"b3479b98681589fac4ad527e5de6316679022c54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-13 05:07:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"72cacab93ee7f778df5481f932bddb2d82878bc8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-13 05:21:37.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I59a3091d53e3e6ab990e80ed3974b2153569b2bf .","accounts_in_message":[],"_revision_number":4},{"id":"def95cb379031e2f593fd6b0a8a847c867f58fa2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 06:06:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dcdca29c5d74ae3b5e0514d5b5cdc26\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b30065ba10dc4239893a436627d87a0d : SUCCESS in 12m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4d3961c3b7b341ed87dbec6307245849 : SUCCESS in 27m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/87060f4cc6154d9da4f42ec4b4ce840f : SUCCESS in 42m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"71087f071f54d217385ed6a5db8106e5cc9b310b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-13 06:06:46.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I59a3091d53e3e6ab990e80ed3974b2153569b2bf .","accounts_in_message":[],"_revision_number":5},{"id":"c1a7652405bdebdb17c7e3009e4686e161a58006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 07:11:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7a8a2c73fd34eb79b92b917aa14c23b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5773bdd523304efe9dc963dbc0abcf9a : SUCCESS in 30m 16s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/689311e5cdca42798f7493a3ecb0bef9 : SUCCESS in 55m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab991437fa744626991b5053673a6150 : SUCCESS in 32m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6eedc37481b14e5fb43650b6d40321b2 : FAILURE in 59m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7da0cd3dca374b8495851a5800596496 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/48045124ff6f4a77a964b0abb2001548 : SUCCESS in 10m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb2be7170c1e408383945bd63b23f2a7 : SUCCESS in 13m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eec407b1228049d59ef4eaefaf95db36 : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf73e7f872254c9783aea7f564d32175 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d74b455571534c57b08a99337b13928a : SUCCESS in 10m 06s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/f8f816fe9937485884177bfcde984c7c : SUCCESS in 48m 50s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f23e7b4da1ae4f41911024320f6a28a5 : SUCCESS in 1h 18m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/082c13b784fc44839a4320f278834df8 : SUCCESS in 28m 10s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/f096a290675846e6ba1d9e5b6b7d941a : SUCCESS in 37m 13s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/411d323770cd48cd8d6fe567279e9270 : SUCCESS in 37m 05s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/963c32f9c22440a5892ec7870a7acefe : SUCCESS in 32m 04s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a460921face84e96a1431073f3d5820a : SUCCESS in 1h 51m 42s\n- nova-next https://zuul.opendev.org/t/openstack/build/1cd3a728af2f42e2b8f19b9b2bd474ec : POST_FAILURE in 1h 24m 56s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/a23667753f8d4f2d89c67d175a49c822 : SUCCESS in 27m 11s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/2df605b480a1462e9d8e38eab93e161e : SUCCESS in 5m 13s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/13d60102ee6e4e83aeb9765e7d4c5f7f : FAILURE in 27m 05s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/755c8a4ea8614ca09b50d5eb319b96cd : FAILURE in 25m 31s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3a2b5b624bdb42ce9dfb1e5a229fbff0 : TIMED_OUT in 2h 01m 21s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4547faa98e1e480e9ab406a68237cd73 : SUCCESS in 1h 28m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd8bc5be7a3a40bd8cfe175326d8cf50 : SUCCESS in 59m 14s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/b51a49defd6848e8a310e4f73417cfca : SUCCESS in 47m 16s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2d740820a6c543a1b0f323f9b21dadd4 : SUCCESS in 16m 24s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/6f31062b914c4cf7946582c6ac28d13d : SUCCESS in 14m 04s","accounts_in_message":[],"_revision_number":5},{"id":"db9dbdd25059a7d2b1c0779ee30b8cd6a2b5656f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-13 19:40:09.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"0e01f543809b960041f2ff78c699acc111a2e267","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 20:17:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6e30d6711e14e3ab0117c43e4f8f729\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3fecfb6011d8446cac2526ef74cd0de7 : SUCCESS in 24m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5b97f7321a53470ea5750cb34b867997 : SUCCESS in 20m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9e9119c29d624bcc9731bf6a9d953f11 : SUCCESS in 35m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7114b0964e9abdf36e2a6c80684a15b7bb67a6d0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-13 20:37:19.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I59a3091d53e3e6ab990e80ed3974b2153569b2bf .","accounts_in_message":[],"_revision_number":6},{"id":"62c1ca23ee951833b9ba5378b19ac3a107bfcb4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 21:48:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32465e137f044319ab0e09d205862fc4\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bd720d66f6dc4ee1b24519b54f3c9679 : SUCCESS in 1h 01m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0059d2f9c4a1420886f2e925ab59c805 : SUCCESS in 48m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea1072d28c2b494f8678809f16c16c82 : SUCCESS in 58m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c4eaa866dc5499ca506883095f6eb1e : SUCCESS in 1h 00m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19e7be45f7ed409b9acee9bfdf218352 : SUCCESS in 9m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3943ca4145eb4dabb9c1eef29d2d35a0 : SUCCESS in 13m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55569fdbc8ce416e9f79c4db520ae617 : SUCCESS in 13m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/776077ce3bde4d8cab2c60d94aa45265 : SUCCESS in 19m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/470a3f06aae54ab0830cc072f3ea7679 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9f371f4da4d443280765e3fa048e0eb : SUCCESS in 7m 46s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/60ec321b34e04f6d9d0c0246b07ce791 : SUCCESS in 1h 03m 57s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/19f1d10e6ad041279895ca2e2d5f492a : SUCCESS in 1h 35m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5abe9a27eef24fa391c6aa91ad163283 : SUCCESS in 37m 32s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/b8d92e3c5d644e6889ba47948c4e4bfa : SUCCESS in 36m 43s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/76356689ac3e46699fed04d5de6a1b9e : SUCCESS in 59m 42s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/202db8117c214d1a9c9f160999559621 : SUCCESS in 39m 10s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/031ad09c14894b9898f682c3a02802fc : SUCCESS in 1h 49m 15s\n- nova-next https://zuul.opendev.org/t/openstack/build/7c92a3c326804e5ea0bb55ebd31f60b9 : SUCCESS in 54m 04s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/58d0d8e6937d4a6aaed3bba8ebc0576d : SUCCESS in 48m 09s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5822c0e185a34aaab532f7aac2708b89 : SUCCESS in 3m 49s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8128a5d9af264c4dbaba071d31b13585 : SUCCESS in 28m 22s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/763cae4af18849b6a086d8057f5a12d4 : SUCCESS in 25m 16s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/fd7efc3b855a418e8c6b2abac870f589 : TIMED_OUT in 2h 01m 13s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/48d5f7a611ae45e5a5251c9061125ac4 : SUCCESS in 1h 27m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c89b417d37d1497c9bbbb0463c43b6bf : SUCCESS in 52m 45s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/572ae49e70734150b2cdb1e3be2d4eb5 : SUCCESS in 56m 10s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d75612484a0c45ba9693acdb9b7737ab : SUCCESS in 35m 22s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/875ff71da0484bd59b1e03c991f2c405 : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":6},{"id":"21d253963cc0b67c1dac4c168e00bc9adcb3b47e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-13 21:58:54.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"95373c7d33cc5ce6ba453d6ce25f27a06172903c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-14 14:01:49.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"533436f5e00016385cb63c4b541b52222594f801","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 14:04:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90c70668a63142dfa2c4b8c000f5fb8a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/56d80c7bb6b0475dba5b5e27ec78e878 : NODE_FAILURE Node(set) request 6f0f237392714b3d8b5ce7922c8f814d failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/17ddc27bf4124847b5860cc97612441d : NODE_FAILURE Node(set) request a931619af0134aa4b6d570d5642e2ed7 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/016796677bb64dd08b83e4d49c1ac874 : NODE_FAILURE Node(set) request 3fd64c4d568d417eab1324735fa26d16 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b75ba6fb781d39651430012ce94a1f09f49bee01","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-14 14:58:55.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I59a3091d53e3e6ab990e80ed3974b2153569b2bf .","accounts_in_message":[],"_revision_number":7},{"id":"b74b2fb0991fb8b4cdd5bb73eb8da2b84c1320b1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-14 15:35:37.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ee42e02d085acaaed15ddcdec1c35e7dcf46f9fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 16:19:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a9adfd9ba064bee958c97d32cd2043d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b5c9dc5686684688b6cec89137823e43 : SUCCESS in 57m 04s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d3c998b48c644803910a89d69d4c3647 : SUCCESS in 49m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0836cd99eba48199ae15bc758833d26 : SUCCESS in 57m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/476b2b6484614689a7a56078144c39e1 : SUCCESS in 58m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c400462d27b44dbe8e994527c5268bbd : SUCCESS in 7m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2af1f6c84bb94df6b40c12e19091f04a : SUCCESS in 15m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f33c9900f5440da97d1dea91cdac7c7 : SUCCESS in 9m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8af0bf2f15aa45b2ad5345acfc985668 : SUCCESS in 18m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc2c8af52dc4044949107eacde30912 : SUCCESS in 13m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1e6af24cef3428a95a1610e30b84ff0 : SUCCESS in 7m 39s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/df5deab319784208b974f7000bfbc132 : SUCCESS in 30m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/01fc129ab8984b869dc13af3461e603d : SUCCESS in 1h 10m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/015a8e07558540059913c09d72df7894 : SUCCESS in 48m 51s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/50357bb500d040a189caedb906771b7b : SUCCESS in 37m 11s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/3da8a174875c4048b40bc3064367a6bc : SUCCESS in 1h 00m 22s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/893fa88c1cc7423ba87501f4a002ff48 : SUCCESS in 50m 21s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0770ff40864e4736b0e8290c8e53a1f9 : SUCCESS in 1h 32m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/2d2148bddddf4fe8bf592fa0f64e1df9 : SUCCESS in 2h 08m 02s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/4d2d6b2bc0634db4a29588a3412a64b9 : SUCCESS in 51m 33s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5fa17883bec7488499c11709a261bf54 : SUCCESS in 5m 29s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3db6b9ca71214b8488bfdb6c959e065c : SUCCESS in 32m 23s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2a261c8eda2b475f8a1596935c9bfea0 : SUCCESS in 28m 43s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3db2fb7486cb4ec6ba4736d8085becd7 : SUCCESS in 31m 54s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/02e4c21839f0463f81b1f80d64e633b4 : SUCCESS in 1h 26m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1057c8de5284d5093eb5c9fe8385973 : SUCCESS in 1h 02m 28s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/f682483a455f40afbbdc20412112872c : SUCCESS in 54m 02s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/12ed76ea1cb94b32b51d367391fe0619 : SUCCESS in 25m 10s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b9864d02529c4bb3872fbd9dea8435cc : SUCCESS in 11m 19s","accounts_in_message":[],"_revision_number":7},{"id":"b30df6cf16e14e60fa99eeacf593368e238f5b93","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-14 16:30:15.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"503b6132ffd70650172b038d8713f4c1329c4aaa","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-14 16:40:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5ecaef57d0b68d4b1139febb24ff74fd652dbde8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-14 17:58:04.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"eeb069a3c746f4d945d5533ad73808651d3192cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-14 18:30:15.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"ca59ead68aa5298d4711f9cb44c8a81c0343f7b2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-14 18:54:37.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I59a3091d53e3e6ab990e80ed3974b2153569b2bf .","accounts_in_message":[],"_revision_number":8},{"id":"1a83674b02d5be5934254dcc09275caa7560e8a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 19:14:16.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32e97f7284bd4ffdbac9ebdd06c5119f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7439b18b4fe94b0f8130d850e6428bca : SUCCESS in 26m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/871e5de530c34ce3a98c1294c4f21070 : SUCCESS in 20m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d77941c484854792b4b9ca6c94f4526b : SUCCESS in 41m 36s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1b1294158fa4609b7ba61c6cb1f2c2ebaf4a5cbb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-14 19:28:34.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I59a3091d53e3e6ab990e80ed3974b2153569b2bf .","accounts_in_message":[],"_revision_number":9},{"id":"ebb004ca478807ef82471f418de783f8328a8647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 20:13:12.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c84a45992c2641d98b67ca85946489fc\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6edf624b649147c38078acedd2a69fce : SUCCESS in 1h 06m 49s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e58cde911556432490dd58c08c371f9b : SUCCESS in 1h 40m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ae342560e4445399d2284d03e4181c9 : SUCCESS in 33m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc9b9efc43c7456aa61d66974c67fdbd : SUCCESS in 1h 04m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53cbc591188b4ccfa17c9856b5ea44aa : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93f4adc328214edda6cce64d4b5e01e0 : SUCCESS in 13m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff4fa18cac6a48bea4c7008c28f4eba4 : SUCCESS in 9m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3c30472674df4614b768a52d581586ae : SUCCESS in 13m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcd8daac67ab454880ed4b0c6de8b95f : SUCCESS in 12m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29f1fe6beb534b5da8bece1a25b51695 : SUCCESS in 4m 14s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/4e4a2c77fe9b48279ae4c506e121d6f5 : SUCCESS in 44m 22s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/14cef6ccf4b84afeb495f0d3c710f61a : SUCCESS in 1h 10m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2c7abff9e6f9438eb275471ef409d598 : SUCCESS in 44m 35s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/1773b079d73947d9b955e270887a979c : SUCCESS in 35m 43s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/b92036ab9b8b4d9f93dc4e05fee26233 : FAILURE in 59m 41s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f33d4eca36924e0eb7ae3817ee9fa224 : SUCCESS in 56m 25s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0148ff11b1354a179ea1c96055d50959 : SUCCESS in 58m 44s\n- nova-next https://zuul.opendev.org/t/openstack/build/685f720462bd4fe6ad3eb14ae2f05b89 : SUCCESS in 57m 38s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/99f7e42a7f434f11b119a92f71be6cd4 : SUCCESS in 38m 00s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/30e9a9997b594da580d898cb0285151b : SUCCESS in 5m 01s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/06168035db60478daa3f69fcfa270929 : SUCCESS in 21m 25s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/edf907dde011496c8244d357766a6133 : SUCCESS in 20m 37s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/58d749e8bd904c2a96ce35c253fe9846 : FAILURE in 27m 11s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8d9c1d608b742edb2b100b50c9feae1 : SUCCESS in 1h 26m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/840aab7711414ee68574c02d7bb5661b : SUCCESS in 25m 49s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/2d8f2a465d364cfbbc13e1871eb678cf : SUCCESS in 40m 52s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d030ec08e2704aae949dbfd8b77007d5 : SUCCESS in 36m 31s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/8f38836dd74246fc8f981e44374d76d8 : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":9},{"id":"f47e540c58e80013322996a0b54c06e8226b1f94","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-15 00:59:55.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d7382aa68852101a8ff66687c499bc41be8f8bda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-15 01:40:15.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98456b4cb66849028940adce550c3b34\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1e2efb6920b14cf3ac43ac2357333f41 : SUCCESS in 25m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0b49971a8ec6440e9c054eab82f9c608 : SUCCESS in 23m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/627d244fa3f1462bb951ee1129b9f7de : SUCCESS in 38m 16s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0531bdaf42f3e1de6b4762b8a2fcc7d7f73cb10f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-15 03:02:16.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f58f963e019d4336b096be83714b2248\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c24394e6b59d4209a8b1bd00e819ff09 : SUCCESS in 48m 30s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5f87f57ef0674afba880979ec2b0471e : SUCCESS in 1h 34m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0bb806679b44b27a604af8b17e5949d : SUCCESS in 31m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca869273f630485bb348de5b82df79b6 : SUCCESS in 50m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7d0ef6c36a44d96b0b8f514d73a2e28 : SUCCESS in 7m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c810339ec6bd4c70b15ac7c4d1177096 : SUCCESS in 15m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5038b2ebf14f4be8a25b426bc0068501 : SUCCESS in 11m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/28cb1f5b26914d2c993a65f9f1157908 : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c924652a75e437ab45dce59c80a16d2 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ca05c24f48f4155ba5cc72b4bc069eb : SUCCESS in 5m 45s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/70a6bab0952d486d8e5aa8d544fd5b1f : SUCCESS in 26m 22s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8ae4c5d369e04ebc9279aa6ccfb82db0 : SUCCESS in 1h 32m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c5a23278a4164ba9a22a78f068ebf8e5 : SUCCESS in 49m 40s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/c99fb2abd24d41058d458f03b1ae9ae0 : SUCCESS in 35m 38s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/6eb1bc76b99d423b881a83740ec91437 : SUCCESS in 57m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f8803325c01e4c70affa54c60850dd99 : SUCCESS in 47m 46s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/92efc63084aa4c9c8a8197e8ae57cd49 : FAILURE in 1h 48m 24s\n- nova-next https://zuul.opendev.org/t/openstack/build/098ed184ae4a4c8fbde92c79b854c5c4 : SUCCESS in 1h 51m 55s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/e1e6a4b684e44d8c974b3a47b74677db : SUCCESS in 49m 04s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/354f46f2f9ec40388cbec7feeefa30db : SUCCESS in 5m 58s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9ee828871359479f8d6df92dc8e44127 : SUCCESS in 25m 52s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8c4d788c1491421aa944ba85b4aa3c64 : SUCCESS in 25m 33s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f42bbc2dd38c41e29040938317f3dbeb : SUCCESS in 29m 29s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc92ea5276a1437d896b907ec98a01c7 : SUCCESS in 1h 30m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9703e9d743946a1814c4e6b2907787e : SUCCESS in 57m 12s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/a1ac0a59b4fd4ff0a2fbad5e99daea4f : SUCCESS in 51m 13s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/096aadc8a3c04a2ca16c3d3f0016603e : SUCCESS in 23m 37s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c9bd3f7cafde4d089f9ee2556727f2ea : SUCCESS in 9m 58s","accounts_in_message":[],"_revision_number":9},{"id":"cf0695c21614288e2ab4bd27669a541cd85ed9e8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-15 19:49:44.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ce271b9a352cc207c9cd06ae2c8b83fb09bff149","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-15 20:24:52.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce5fea2300174927a7d658f0b1fd56e6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f036a13cf47d45c080d72df54d6d594f : FAILURE in 4m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd875c0c8d2141aeb8cc894e674fd336 : SUCCESS in 16m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/610ff4c596b14f86abbb2f8ad920ceed : SUCCESS in 32m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"31921999300f4925166bb2607a46d7608a91ce20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-15 21:49:12.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c2378f52d96427eb030d480b8502c88\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/86e333b6d13c4e3b89940209253ef349 : SUCCESS in 1h 01m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e61014d59a5b40a18a17f024f6411249 : SUCCESS in 1h 37m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76c21989f164445ca9c0051b9668fc96 : SUCCESS in 55m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97fd98eaf93a40c7b2ef288494a182a5 : SUCCESS in 1h 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49a0d7ab6d474229acc3b58bc76c8668 : SUCCESS in 7m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/242ecfd2856346909006cbafa9415b8a : SUCCESS in 14m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b2c0e5203f04334aaee0b79897ba85f : SUCCESS in 9m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5962c54f2a5d4e0e9ea319f92dece79e : SUCCESS in 19m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04a03e432dc44e6fb45b51a694cb1683 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44cdbfc35d5f40558c4371bc051b33d4 : SUCCESS in 8m 50s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a1d74332e99546629d4a2edd510ac7bd : SUCCESS in 1h 06m 57s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/37b207044eac41ff90ab21d19f5d4f4f : SUCCESS in 1h 35m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cfc26166175e48968cbdd7b3fb30270d : SUCCESS in 52m 48s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/c7326aec109b49799c4755900befaac6 : SUCCESS in 34m 02s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/375a81cfed5f45d088252195e5ebe948 : SUCCESS in 58m 23s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e54da0ea85604b3daf3db5d48efd0efb : SUCCESS in 51m 47s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0384f984feaf48be9b4f1aa6130e6909 : SUCCESS in 1h 20m 14s\n- nova-next https://zuul.opendev.org/t/openstack/build/9ac73dbbf9234751ac048dfaffd509d0 : SUCCESS in 1h 51m 49s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/fa1f4c523d714cd18d111c9962ec88db : SUCCESS in 45m 35s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/855281c484d14bb3b6d5d485545fcf1b : SUCCESS in 2m 58s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/171ae17c9c3b4b8ebacfb54d5f08db7c : SUCCESS in 26m 58s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/97da66540ca5421198da2d5eaec12f50 : SUCCESS in 21m 40s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c1adf2e2852549789cf5339012590f59 : FAILURE in 31m 10s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/39d3c5fa01034b80b25653cc7ae7db16 : SUCCESS in 1h 31m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f7404c741514749a935b398493a7c02 : SUCCESS in 35m 29s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/33892070e3944f418c1ac3fda10d8695 : SUCCESS in 55m 12s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1d4cb26d0e55431ebfe7ac8dff332e54 : SUCCESS in 17m 44s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/841b09c42580493298dedc0b754118ad : SUCCESS in 8m 27s","accounts_in_message":[],"_revision_number":9},{"id":"fd6a5baf6773c1364d5fd67c83356aab38be6223","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-15 21:59:32.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"0363439c969cb62313321f4ae74db0dace9bd8b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-19 02:50:21.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"7abd6de2d1dc8c88dfbdcac5563caf62bf9b8e92","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-19 03:46:57.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I59a3091d53e3e6ab990e80ed3974b2153569b2bf .","accounts_in_message":[],"_revision_number":10},{"id":"90bc27e396ccde92494d2daf5b69238ebec76154","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 04:47:38.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/265ee69f5a78481db8db4545c61d3f88\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/78efadeb01c44dd8962bb5d8fb54a538 : FAILURE in 7m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3d148b885c2f449da813004368f0a762 : FAILURE in 6m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/61b5c901751644b6b3f626ddaaa9a988 : FAILURE in 34m 23s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"486d19dba138a0a64ab8b082576b47e0762ba881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 05:27:25.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eda4191124a3464d913818c840062347\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f61424228a474f03b2a04ad2ac71c441 : SUCCESS in 45m 28s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e359ce9a373c4e1fa448d78c7ed9fd6e : SUCCESS in 1h 07m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/696b19a5f5f0421b8efdb7374d852b52 : SUCCESS in 41m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4350a264d0c439c8f2df1ae15b30a86 : FAILURE in 32m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a407bb3d102f4fb3bf7fa047f81b7718 : FAILURE in 6m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1046d613607941ef8c8c0f7901a43aaa : FAILURE in 13m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ecda7919a7a4b2cbcc55a8cb7183352 : FAILURE in 13m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b1189ee7dc4c4df188b5b9c3483ad541 : FAILURE in 11m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8655b8066814b809640bfa52eb97462 : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/768461a3a3ea47d594439e692cfadd9c : SUCCESS in 10m 17s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/1fa1dde5a00b4287ba5598ecc2f95ad1 : SUCCESS in 1h 04m 36s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d946bcd8a0c047be97e4ebbe1fc8a741 : SUCCESS in 46m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/163b208c5cfa44c08d0f1fe2f3cc1adb : SUCCESS in 34m 47s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/6a5e731b7e6a457ab4f7ae0dc0ac748a : SUCCESS in 35m 44s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/9c8b267b3d4844589b5328be36e1d287 : SUCCESS in 38m 31s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e530a9fd40524871aaa72872143531a6 : SUCCESS in 27m 24s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/915a941a3eb44ae788678c46ad248416 : SUCCESS in 1h 55m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/335330204001421182bf8ec47bcc64fb : FAILURE in 2h 19m 18s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/979641f0cd3640cbbf7496b62e82a0f7 : SUCCESS in 51m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c6e9c6e4570f47d5a6c6b42ba4192d14 : SUCCESS in 5m 39s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/58a2f3cf337c4d918ba400765da67a9e : SUCCESS in 25m 14s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b6d2c44f8f484a218a447828644c309b : SUCCESS in 24m 24s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/afc8e0e8b0ee416f8038278ac16ac953 : FAILURE in 28m 31s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/84741972f63842a1b14778322d446846 : SUCCESS in 59m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b09317963d7c4bf8a228976bc5b74fbd : SUCCESS in 57m 14s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/07acd6ac5d9543b5b0385e5828b318e0 : SUCCESS in 52m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1788ffc2e49c4250ad6866b3fc0a5a42 : SUCCESS in 35m 02s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/88c0cbf96df34de4b3f487d13a3944cd : FAILURE in 9m 26s","accounts_in_message":[],"_revision_number":10},{"id":"6d7153e63417f52fcc0cdec987f6a7af128c405a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-19 16:50:56.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"743854723211d07d015c09a9f21e33ccbcaad8f7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-19 17:11:02.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"cfb46d26e433d9066fe6f0e4b1025379c1216ee1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-19 17:11:09.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6656ac2572779e520b0da16f24ecabb0fd59c6a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 17:38:45.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99bc03ef30a84201b2f99f880af8e3b3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8ad69e97d6c6499993ef245bab65af3d : SUCCESS in 23m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e44046d080074a10850db6d7c80c64af : SUCCESS in 22m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dbb7294c381e4615bccf85cbb3be595f : SUCCESS in 24m 11s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"fc3bbf38b389a30fc7cc46644097a4d0a3b94a98","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-19 17:47:47.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I59a3091d53e3e6ab990e80ed3974b2153569b2bf .","accounts_in_message":[],"_revision_number":11},{"id":"4150f320317e87fcb6f7dc9044b24c24de75b23c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-19 18:07:41.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I59a3091d53e3e6ab990e80ed3974b2153569b2bf .","accounts_in_message":[],"_revision_number":12},{"id":"291cf2329d5365d4b6e37b529f8e15a5e680568d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 19:18:34.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f32faa1e6dec49a980cdbe31c73adb39\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2802a0a4aa494156a9d6fc1ed042aa03 : SUCCESS in 1h 01m 11s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/cd87b7398e9244ccaaf947457fe20a08 : SUCCESS in 50m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83b94c27c7254a9e8f003410dd26d1ca : SUCCESS in 57m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65589b21f11944228258db86ddfde653 : SUCCESS in 55m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d470834cddc40ec9a7e830678157a4d : SUCCESS in 9m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b56a18df5c1f4abebe5df2e100ec0414 : SUCCESS in 14m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c99a9e4754fa44dd86f04e4de0c8bf48 : SUCCESS in 9m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/41e4e3c5ecf34b86869affd42d8fddc1 : SUCCESS in 19m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90f713bb3ce746c5aa834814d1a304b0 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec72e142a96049e9bcda2a9c29064e49 : SUCCESS in 9m 09s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/cd1cd3766e2847228d77c8293e6fd340 : SUCCESS in 33m 44s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d411ef97b8c54909b2b57370995a9ca5 : FAILURE in 28m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/28e1597bc02e4bd3810ed4e6496d93e0 : SUCCESS in 49m 44s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/8db8e8530ddf4299972d97c669e8259d : SUCCESS in 34m 30s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/607cc77cbb374ac6b1cd513c35518b5c : SUCCESS in 44m 33s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/150b6a0a11c840f89f0dd98305a5fd85 : POST_FAILURE in 39m 26s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9d71d88f92fd4013b64400d1b5a1fd4f : SUCCESS in 2h 00m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/426be993b2814e51b06041773eeef104 : SUCCESS in 1h 56m 41s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/61e0ab494a0d456ca95d16ee7383eec2 : SUCCESS in 46m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/773bc7134bc24826a48557aae9865135 : SUCCESS in 4m 38s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7c795057ef7940568fdc52340f5fc442 : SUCCESS in 28m 13s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b4e9aa45d08f4b0aa96768ed29d9bc5e : SUCCESS in 32m 19s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c69a1eaf10e14efaabc170f4836d63f6 : FAILURE in 29m 00s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b647282eacf4417182e9f6c294de6d07 : SUCCESS in 1h 08m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1036929ca0bc4b1dbfe4e1114e017b0a : SUCCESS in 42m 54s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/a2f8b21380bf40eb83b1d64d78926516 : SUCCESS in 31m 48s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5751c9d8cf064aaf9c1b735449e4742a : SUCCESS in 41m 31s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/1f6923520e674addacf3c39393842f65 : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":12},{"id":"b265f3140e798bf5059a3cac2b3f8ec5249192d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-19 21:20:15.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"4b39d6c463e365d9321e519a85229e4b57f14e74","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-19 21:20:41.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"56babde52bd219fc0d3dcdb47bab02bd8712eae2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 21:55:42.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/055c8b8ba64e4826ad0df0e0334791b2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d8a297a81f914770ad3e37857b8dfa24 : SUCCESS in 21m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1dcba0b63c054d868de68c9976402905 : SUCCESS in 17m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e6f3c4a82e464331b107921ea6da7941 : NODE_FAILURE Node(set) request 51dba71e9fb54f299d3ec67c02ba37ad failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"2c1edc79269895d137ae6ebec34acce732790ad2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-19 22:20:27.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I59a3091d53e3e6ab990e80ed3974b2153569b2bf .","accounts_in_message":[],"_revision_number":13},{"id":"61a4eef81e63312ed2cce02b94ad99a3da90eded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 23:21:01.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5dd3d371847f403aba74fc46b178c45a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4fc3c9269251417fb28a10323b30c745 : SUCCESS in 44m 12s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0739a05aa2de4be682a352ce9db60766 : SUCCESS in 1h 00m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62b3e03691774bf6bb0e09d6c1d0f223 : SUCCESS in 1h 01m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff4ac8e801e2407487f4cb1ea3e380bb : SUCCESS in 1h 02m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b6ea4d60ae46f2936862355189afed : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c081e3691534c1b85292ba4c24bc8a0 : SUCCESS in 11m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bbb06d0df3754c818b42328241db5767 : SUCCESS in 8m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/869c2481f8244516ba2516ea9c7c31eb : SUCCESS in 14m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/355ca2042b844dc2b838d8b34a483ee0 : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2119e51b16043a0b0d3d934aff1254d : SUCCESS in 8m 27s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/5468a8659a3e4805b638f56e179be7b1 : SUCCESS in 1h 06m 33s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c29ccb024f0e49fba646ee1b28f03986 : FAILURE in 49m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a541e698b04e4ac9a780139259573037 : SUCCESS in 53m 50s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/661e951f4ab440bda2f0ea8235cc7474 : SUCCESS in 34m 43s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/69b51b6df2264d20a3b7ee5c5bd766ca : SUCCESS in 1h 03m 25s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f8cc798644284132afff276f0688ba73 : POST_FAILURE in 37m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f83d8e17f230421f8e330f59988692cd : SUCCESS in 1h 54m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/a482ae7d80fe4984b5ea426c44e80c68 : SUCCESS in 1h 46m 28s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/1b6c1fe77f5b4583b33ebf0b1ebbefaf : SUCCESS in 50m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bd2ef0546d6042a2846b73ad904769bc : SUCCESS in 5m 00s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/549b7c85f0b248bf9c8c0abccd4141ff : SUCCESS in 25m 53s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/67f42ff4a41e4b7d887428a84042d754 : SUCCESS in 26m 50s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f98fd85d9b444615b6963693157f3ea3 : FAILURE in 37m 28s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d0c46a514474d3fa003354db03507fb : SUCCESS in 1h 30m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c26e6392a07e4514930ea76f03f34ae9 : SUCCESS in 1h 00m 46s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/2e86e44c4c6e435b8f0fffe1e9333080 : SUCCESS in 52m 35s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8b283679bab34638b69da785e978f655 : SUCCESS in 35m 49s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/609394d947364a159e176b21b541e556 : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":13},{"id":"7f79cb2bb686b25a791c7dd3eb8c15a1658455e6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-20 01:28:01.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"154f7d58f54a933ca30832f54b30934f8ecf8256","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-20 03:03:00.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"d6bceab5884b11457646d825143cf481d82b3dcc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-20 03:58:34.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I59a3091d53e3e6ab990e80ed3974b2153569b2bf .","accounts_in_message":[],"_revision_number":14},{"id":"412cbe235199be73e93adff93cbc245587c39e40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 05:24:59.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ce7179e81744e7eb75ac7f6f22ffbcd\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/13bd6040011f43d0b2c3993a08a57c89 : SUCCESS in 1h 03m 12s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e1fdd455ea1246e0b5318043e9ed532a : SUCCESS in 1h 28m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed28c82843be45a0a19d9c2a002863f0 : SUCCESS in 30m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74495189009d4f47a69d32a00b31f120 : SUCCESS in 58m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f846d72c0640809e59c0270f16d45c : SUCCESS in 7m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/24ddc23538ce45f9804399bbe76b5c5b : SUCCESS in 13m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9de3791a440c44908a70e7ec4281940a : SUCCESS in 12m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d1c8455d5fcb4b0e87b63b02ed864a70 : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28a23ca91beb40e6b432d8318dd85370 : SUCCESS in 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8df4fbb70d2b41918df47ed5caa37bfc : SUCCESS in 7m 48s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a457dfb267074098b91edd0cb4308c84 : SUCCESS in 29m 38s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/04bc0366765e4b4c86342466eca1322d : FAILURE in 46m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c5f0d18bcbc14c9195c5627d2d2ba917 : SUCCESS in 49m 46s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/0f6b8b4f8bca47eea4d69bac5370a982 : SUCCESS in 39m 37s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/96276843cb93474a84c9e78008d5507c : SUCCESS in 1h 01m 11s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ec5b0d5e2ef84dd5b91598bfd08444ee : POST_FAILURE in 36m 21s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/06229c3589624f2fbde11e4d5e5091b3 : SUCCESS in 1h 53m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/8ac6a774d578425a9e78b560ec3592bd : SUCCESS in 2h 08m 57s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/1dcc1e3b14f247a5a421776cc5f24a39 : SUCCESS in 47m 13s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/01dc1a4280b3452ca41ece6c0d528fb2 : SUCCESS in 6m 53s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c101f0fe5934499aa1cc825cce58411b : SUCCESS in 24m 02s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/95112144feeb4f3a99698fd529429ef8 : SUCCESS in 27m 27s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a1c2262af3994cf3ba54b387141eb980 : FAILURE in 31m 28s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/70c203ff07524a749384b2267830d67a : SUCCESS in 1h 27m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ac02cfefa1b43008ff1035a6b6ff354 : SUCCESS in 27m 42s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0f43610e972b4a03bb5a4e5c3d6a361b : SUCCESS in 57m 03s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/58b69ca64018478282df982d496ffcd7 : SUCCESS in 32m 26s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/ada40868655d4f2eb116976925f85663 : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":14},{"id":"5c91512187f4f328696b55380e3abf483ebbd210","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 06:31:34.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aac4c69f208f4ceb9db114eee6e6979f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8e0c31bc8f3e4adc8e3c0eeb0a94ec72 : SUCCESS in 27m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/544b896a39034be3b33dfb4b58f2942d : SUCCESS in 25m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/300332c2805942b5a78967b856f57d16 : SUCCESS in 41m 22s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c63a7fa6b31b3ae78e1ae97218e85da84feb21c5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-20 14:51:44.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"a7782291105b87144f0a95e1bdfd95c437b829fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 15:46:20.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c30c63a36dee42559f12794c6c44aa45\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/15660d11c816432fa32115d0122d6e68 : SUCCESS in 22m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/147b90731e6548fcbce287c3538470d7 : SUCCESS in 32m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e004ee6e13f0443abc3094b61b2c62e0 : SUCCESS in 34m 47s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"9b3b97558cb55e7be9b0668096021546502a7691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 17:06:09.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f39347b9e51b4b74847272ae2d64cba6\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b1c7519ce3954c5e83b8fdd527b38c03 : SUCCESS in 54m 17s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a2f6464b63654ab7aa3c7d8ab3c4f40c : SUCCESS in 1h 45m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2162a48f8294a5a9042c51879c883cd : SUCCESS in 1h 00m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4bc73d38f204370ae05ec12627ddca4 : SUCCESS in 54m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/529840a200fa43f19649e1621b7b075b : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d36fa407242848c498334bd19de18380 : SUCCESS in 15m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3efa90cecd9a4696a4014259a9d8f80a : SUCCESS in 12m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1ca1adc64e5144a5ae7f97cec56e924a : SUCCESS in 19m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afd0ff0cd9f1400f9ae62c7ebafb1fdb : SUCCESS in 12m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe4b8bc583a043d6bb85b7741d46a3d1 : SUCCESS in 5m 16s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/b02bf57e788843aa9255548455deac30 : SUCCESS in 1h 04m 00s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5905ded4c25d4c56abfbf80ba7fe46b7 : SUCCESS in 1h 31m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6d6e25bcedb948f3b22dde8eca8b6cb6 : SUCCESS in 28m 02s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/c3626b5d85744e2885a8450f93e13298 : SUCCESS in 37m 41s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/15a6cecaff2b4990aa8f5684f606374a : SUCCESS in 1h 03m 08s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/dd1ce1f9890b419d8648f8269f6629a5 : SUCCESS in 27m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/49cedc5db03f4b878e3076bd6f7e86ea : SUCCESS in 59m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/340a568207e149aaa043e6c5b2154b33 : SUCCESS in 2h 04m 00s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/09def1ea1b104c4387eb91e9d3a753b4 : SUCCESS in 41m 57s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/905d58d9896f43ecac50f7e33957150b : SUCCESS in 6m 22s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/51e6acdc36ed4be280c5772ba9d5b6f5 : SUCCESS in 27m 51s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d9248e167077497fb6c57e0f3c4a2d02 : SUCCESS in 23m 36s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a54a97a603404b16b360fda17015a9ee : SUCCESS in 29m 56s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/dab3aac869ec4f159ac155b2ba2169a7 : SUCCESS in 1h 35m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2967ab9f012b486caf5d4c8d2252d66a : SUCCESS in 49m 43s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/c14da6ad0a704eb68920392b0a2f303c : SUCCESS in 53m 06s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6656710744b141679ebc93b70d3dfaef : SUCCESS in 34m 15s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/378adba1522c4082a23351094f3bb2f3 : SUCCESS in 14m 45s","accounts_in_message":[],"_revision_number":14},{"id":"9ea8961c43d283898d47562b97f7743efeb105e1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-20 17:17:41.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":14},{"id":"acd662d46754f3748fdce5a44973056f983b016d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-21 14:57:38.000000000","message":"Patch Set 14: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"93ee1175f8e2b42bff049afa7f94dae22fc3237c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-21 15:09:44.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"c0b1d84263aac33b1426540280116f075c7924bd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-25 14:08:06.000000000","message":"Patch Set 14: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"89d9717562a0cb88ecf65a6d5ea5a3f1ee7b8937","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-25 14:14:19.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"9f7fc673dfbcbe69b243bd88ba8f351c94354594","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-25 19:00:05.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"4a79975db0fbfdbdcc80d9b3ae5050810dc13aad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-26 00:20:11.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"fbddd9b25697e5cab5ca3dd69e120cb0765f9fc4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-26 00:21:37.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"df3f523ad1d4cbd492384de87743fa113fce21ec","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 00:24:44.000000000","message":"Patch Set 15: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/6f6bc934078e42cca19b94bc0594fb1e\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/baaf4c0bce554c678bfb88d0c876f3c2 : FAILURE in 3m 23s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"79ea696bdf11c534ddd62c2a6ce028432048dd5a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-26 01:21:36.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I59a3091d53e3e6ab990e80ed3974b2153569b2bf .","accounts_in_message":[],"_revision_number":15},{"id":"1b97f42bbe6660c6e02be00c13789d1637584f93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 02:03:55.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f251f70d4abf491fab6bec5f09d95d2a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fe6d7412dd2e483281c3b68f177a2601 : SUCCESS in 14m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5d5bafe10ad34e4788f97a5d4b75261e : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c3b76de849da42d29f025816219a2519 : SUCCESS in 35m 32s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ffb90aa67fb5e8fda881cd5f681656111893983f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 02:41:50.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/321487f6555f4baaa8559721a349e341\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8c9593ccca2649628544922b2ac2b82b : SUCCESS in 1h 09m 22s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2e85e7edff70461b81de9157d1b8b573 : SUCCESS in 48m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45e9589a461e41a386ed54feafaca448 : SUCCESS in 49m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67567e3970394e91b3be8085233cfaf0 : SUCCESS in 55m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6df7b7883d864c0dbe528e874afd0c75 : SUCCESS in 6m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/091465775f614463810a0a588cfc0032 : SUCCESS in 9m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e10a1aa686da4737b4c8d46c4f3b5225 : SUCCESS in 12m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/872be16758d14a118ddaea0ddad788bf : SUCCESS in 18m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa8a91c22e604f42bfd78936c2e38d88 : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/745a38849ea94e83ae8f6ad442714df8 : SUCCESS in 8m 33s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/15c315cbfce74f1eafdb79cb0832be36 : SUCCESS in 1h 05m 49s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1b4f769cc7c546af878ea23e524e4ae4 : SUCCESS in 1h 01m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bc14213cb08d48c9a5b763b662382b2b : SUCCESS in 54m 18s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/0702cf52acac4806a4500431b2af4858 : SUCCESS in 37m 15s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/20d2d0a3ed3845cab709bb452bdb0626 : SUCCESS in 43m 06s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/84e624a986ba4e189b21b0e6925837b6 : SUCCESS in 25m 29s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/be114aa0dde54a9392563c2d954eb03e : SUCCESS in 1h 00m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/6e83b2e527ec42589f45f5a3e477d054 : SUCCESS in 2h 05m 24s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/ced71d1cb7634d449e70a8cc979ed1ef : SUCCESS in 32m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a800334177124bf7b53c4a5f663f6058 : SUCCESS in 5m 38s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/427b788bbffd48d2881ddb55cb60022e : SUCCESS in 38m 01s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/09be3ba7a792471c9aa8f1eac6760710 : SUCCESS in 26m 16s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/614b9ae53d874e659db7777e675b09aa : SUCCESS in 33m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/92b0e588c31548609231de9bfea755e9 : SUCCESS in 1h 02m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c66cbc9634fc4c8cb9af9e113125b9f2 : SUCCESS in 1h 01m 37s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0fdb4f271e984b029681888efa885109 : SUCCESS in 59m 59s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a38894a025344966a45cba9192e049d7 : SUCCESS in 20m 54s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/ff37506b45604263bfca9559d93e23fd : SUCCESS in 14m 27s","accounts_in_message":[],"_revision_number":15},{"id":"6c3ccff6915f20f1f5fc54c8466d6f62f1966b11","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-26 02:52:34.000000000","message":"Patch Set 15:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":15},{"id":"ac41e0c858e1f64448e287120591b02a0a83435c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-26 12:30:24.000000000","message":"Patch Set 15: Review-Priority+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"44d0d785e19794c8a4b4ef89679077f7567576eb","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-26 12:30:33.000000000","message":"Patch Set 15: -Review-Priority Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"a77506916eb38e29c3e63afd33242ce3f01f7a57","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-26 15:08:33.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"adb85b4068f7e2d73e5e361d5a2b8e337d3d9d0f","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-08-27 03:02:35.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"fdbfe18f303454fdcb2b6434499cae2a05b50ca6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 03:03:34.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"9c4322347c747a7b262eb85868c987644b7330b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 05:41:22.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9567c4935b4b4780856c19f7dd2e3db1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d930f262b3a444f2b27944f462084af4 : SUCCESS in 33m 40s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/971161baaf3f4654b50fec1398be8467 : SUCCESS in 50m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49b68cf4a1924433a10ee263692cb119 : SUCCESS in 1h 05m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b32671e006164d0aa1a6288452710801 : SUCCESS in 9m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/920a55ded08e46ec947faef779d45359 : SUCCESS in 9m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1a24a5e5881f4ed78cb0a21901d3ee32 : SUCCESS in 18m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2ec4950cf3f401aad15114a41dbfdf0 : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f88ef024bcff4e2884ef407762594f9a : SUCCESS in 10m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/49e1864eb2914a7f8fde9a0ab57412a0 : SUCCESS in 22m 51s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ef4d4e1c189e4898a57de8f58ca45c56 : SUCCESS in 24m 50s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d223a2d4f3e64b1da462102dde368a34 : SUCCESS in 28m 24s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0985323e5a8c4ef0831a172f1a77b827 : SUCCESS in 35m 47s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/45918f1d2dfe433881bb4c57558d7152 : SUCCESS in 2h 01m 09s\n- nova-next https://zuul.opendev.org/t/openstack/build/5fc96162595c4a09a5b73a673a07c9f1 : SUCCESS in 1h 00m 04s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d180eaf3d23646fb97fbe43932d14c7c : SUCCESS in 3m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d44a218a7d8f4210b20dcc0ecdf1467a : SUCCESS in 1h 16m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/855dce8949e44b108d412c3ad3211765 : FAILURE in 1h 20m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c14ca993563e40a1ae504daf98f49b21 : SUCCESS in 1h 07m 36s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7b04f6d863474380965d1ca31712da38 : SUCCESS in 16m 07s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/57cfcb21720e44b99071a28996fd0e8f : SUCCESS in 27m 27s","accounts_in_message":[],"_revision_number":15},{"id":"132171d8b7cacb27a951d1325ba7d808a0be2208","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-27 08:22:56.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"271a0dd918f78e5b1c6b112bec2e25feea70745f","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 08:31:39.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/8ead5fe4f87c495896c9422707a0b9ac\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/be2915bbf7b8422697cd729c3322e4e0 : SUCCESS in 7m 29s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b6992494445e16a78fbb99a3ebbec421bc45e15c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 10:08:45.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cdbcb8b74c940069febd873a7e9ea14\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/632785061d52435ba0ff1213e756b29c : SUCCESS in 13m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/517f364654d847b8a4daf48b7ccbaf68 : SUCCESS in 16m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6434ed58ff994d959e1359cab91e1c1c : SUCCESS in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0de544341c50c28ebbf5988ac26c637a8d10e024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 10:52:36.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a980652d35b472c96ad63bd44906f1a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5d4b1571854c4622949c22ff586541cc : SUCCESS in 43m 35s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a74d521c3602433990e08cac6c9ae8ce : SUCCESS in 51m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31f20f09c6004d04a08b8710316bb3a7 : SUCCESS in 56m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e25a892e3084350b88481845dadb6a5 : SUCCESS in 1h 15m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f566a97e7a7447599e0459543b15cef6 : SUCCESS in 9m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d807a56ebcf448fc9cd9b0b024bae301 : SUCCESS in 13m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3906d6ff70b40d8ad5f543eb403fc8c : SUCCESS in 14m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/23d9f37b2a83492db6cc6d1ccb984dc0 : SUCCESS in 17m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5668ae8a4f44eb99d29ff3e92f82f6e : SUCCESS in 13m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20fd8bd630bf4d7293429ba50b83d710 : SUCCESS in 8m 35s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/cf6f03af3dc04e9c9eb60791ab9bc43d : SUCCESS in 45m 18s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c4baf96b702743b08492806a596d42bf : SUCCESS in 1h 30m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dcb469f42eb142738328713c7520a743 : SUCCESS in 51m 32s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/d5ecc481f6344a1b84595ee09c893d95 : SUCCESS in 37m 10s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/7ebf3abd03c74398888b54cfacaa84c3 : SUCCESS in 49m 14s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5d715b83e1e9474698b075e1cad74490 : SUCCESS in 52m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d0236359840343988ae490040aa892ed : SUCCESS in 1h 58m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/2e46244d33e346099d7e7cb0474597ef : SUCCESS in 2h 14m 35s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/7531182057bb4f0aae2d5c1dd9d1fc66 : SUCCESS in 35m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/cfa5d785edb944318ccb339282dd6cc7 : SUCCESS in 6m 58s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/28c9c096938b407ba600c00990580c08 : SUCCESS in 27m 55s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a8c7c9a8a5434824a80ec47339d1d9f6 : SUCCESS in 23m 06s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/74de4a32a9e54993b83b414752daae49 : SUCCESS in 34m 04s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/715db0ab179c418fb8f521b49506744a : SUCCESS in 1h 34m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e523bbbd9264560aa1834e874bf126b : SUCCESS in 51m 57s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/a32c5f99cf2a4cb8b7c6b2c3978519c1 : SUCCESS in 39m 30s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b6efac3578dd4fbb904485fe554cf9f7 : SUCCESS in 38m 58s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/f179f8e7d328434bace5b41be4fd1eb0 : SUCCESS in 14m 19s","accounts_in_message":[],"_revision_number":15},{"id":"94670ceb932421e198d1fcc386306bda886bf41a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 10:53:28.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"1692067b85e03c60e4fa992e115a22f510118fb6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-27 11:03:21.000000000","message":"Patch Set 15:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":15},{"id":"cc1ad39836df7e3bf7986153fef4e9a084b5baf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 12:47:55.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f89e7a9f2cf1450ab08fe3ed36cdbaa1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0e7d95dd81e9420ab45d06b9be052c9d : SUCCESS in 1h 10m 49s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1b7cd60f3bb649beb88841dccfaa6f2f : SUCCESS in 1h 36m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b556d76af1f47ada5ead9545f09ad4c : SUCCESS in 51m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e68aa3ce0f14434997ad5f2550f9daf : SUCCESS in 8m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44ffe9aed7f54df7aa601e7bb690ece9 : SUCCESS in 14m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf4f79d201c841aba55374aa57995f71 : SUCCESS in 13m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1be1df9fa6de442198c7b701e80223c1 : SUCCESS in 11m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c995b59f07dd4e1d8018792db2c39de5 : SUCCESS in 8m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0753b6245244436d890dd4623cb173e2 : SUCCESS in 27m 54s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e359ccadbb134515a050914d20c2f83f : SUCCESS in 27m 39s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4a35768264284635ab732c26aaf44ff6 : SUCCESS in 23m 35s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/83f9cf4115674c34bbc2a12cf6b260e6 : FAILURE in 37m 59s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d6bbcb51c4c04afebd54bb3ed392d0f1 : SUCCESS in 1h 48m 11s\n- nova-next https://zuul.opendev.org/t/openstack/build/478850c57b9147778f7d0832535efbfb : SUCCESS in 1h 04m 00s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f9463acaa38b4ea396af335c4d168949 : SUCCESS in 4m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/84219f2de48c4b4f9b159daf19e19381 : SUCCESS in 1h 24m 35s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3b3677ba037458f8a270f948700d8fb : SUCCESS in 1h 34m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8c57fe82d3e4188bf17caa82ecbac59 : SUCCESS in 1h 03m 58s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4265a7bf568f440381cb4f5e6ff74a10 : SUCCESS in 10m 19s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/12bc8fcdc1af407897d3056e1883cf8d : SUCCESS in 51m 55s","accounts_in_message":[],"_revision_number":15},{"id":"85ebec888f05439d95ce40858236f404444cddbf","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-27 12:58:07.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"e28cf7b6df1e9ca9bdc4c470acbd032dbe353db4","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 13:05:46.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/086caadc9c5848d3a28a5aa319caed3c\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/3e571b3fb2c74f13a3dcaf505b2bedd9 : SUCCESS in 6m 57s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"37b3ca22dd9001e979c683a14d1841c349b953c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 15:04:51.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6622ea5522f4af797e28c8c87b3a411\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3e0d6fe6c93b475e8f52f7072e1867df : SUCCESS in 30m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8c27d13d44f3400c9c587affd12d9285 : SUCCESS in 1h 39m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddaf68f8c61f49f78421de6113f177e0 : SUCCESS in 53m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b886ea52191941a1a66dc47e4089cc58 : SUCCESS in 1h 00m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbb87a1061744fa6acf3d07c87fc7df2 : SUCCESS in 8m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18e8db9d1e7547ccba72e70d130da2aa : SUCCESS in 13m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f266252511a44468dd825aca25b14b9 : SUCCESS in 9m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/737d7ec0487e4d1c9d4fb590d754431f : SUCCESS in 15m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d930c64cbd04b388e6fba4cdbc0cfed : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/274090c653dd421d99cc654a940f525c : SUCCESS in 9m 59s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/7dbe2bdc0e8d4b10a59694a6867cdab8 : SUCCESS in 1h 00m 45s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f4c50885fa094b7cbaf5e5e9a3f7d02c : SUCCESS in 1h 16m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/72c9022016d949fa85a7604669f2a113 : SUCCESS in 26m 57s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/575c5e2a36724036b8e758ed969050e6 : SUCCESS in 39m 34s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/64cc9e7820f34c7cb03faaf5e7150355 : SUCCESS in 1h 00m 26s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/cd6350bd90984449aee2878abc9d771f : SUCCESS in 44m 31s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9e5d931a675b4d0c91bf74c0e47ae6d9 : SUCCESS in 49m 15s\n- nova-next https://zuul.opendev.org/t/openstack/build/1862ddc23d3f4a49872a5fcf93178caa : SUCCESS in 1h 58m 30s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/152e74897d17406ba7b0232b1c844b60 : SUCCESS in 49m 39s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/940373d0b2d04069adf834f6d8ee1e06 : SUCCESS in 4m 17s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/efb8256f70044b2889d5760c8fe4f32a : SUCCESS in 28m 45s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/57a37ffc537b487c92ec31565b950556 : SUCCESS in 28m 42s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/86964afe16be4746a9fe7fbb7f55c58c : SUCCESS in 31m 35s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6de4bd2803704eb3bba0b5a105b53c15 : SUCCESS in 1h 27m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9eadce5205e942c79addd78ff561d139 : SUCCESS in 26m 15s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/88000d7bcf394b749de35ffefe1fbe91 : SUCCESS in 52m 02s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/53403af95ce64285aa40dba0b400e890 : SUCCESS in 36m 03s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/8926033513d54ad7a882f3ba3e9c3263 : SUCCESS in 14m 26s","accounts_in_message":[],"_revision_number":15},{"id":"6847d8b02c6788bda5389e2f652714c0091c611b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 15:06:38.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"faca6e9f3411b93c8e5159a58c135c841ddf8718","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 15:10:39.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0168e84ef1e4a13a9116111acb40030\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/03f112b32f044b6185a9998fe6c7def9 : SUCCESS in 16m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/59673fbd367e4f8db8457a51cb6d573e : SUCCESS in 20m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5032031ae7564a91a3332baa1055cd4b : SUCCESS in 24m 36s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"43e7cd0ade38584f57b008c71a634db44dbe5d68","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-27 15:15:24.000000000","message":"Patch Set 15:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":15},{"id":"6537d270d9b9392d009ef4c426b659b19b28aad3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 20:05:58.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5849119e6d641dd818f1c2d89720f57\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/68d4b2801dfc4987b17731d52e868da3 : SUCCESS in 37m 04s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/14d9113ce38b44768be22bbf104bbceb : SUCCESS in 45m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c144fab8d50a49af8b39be82dcc20410 : SUCCESS in 32m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ba3ba3af2794d938203b3c30c68ee7b : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db5a6d62428b4ed3aa2ff890de401385 : SUCCESS in 8m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b44a7ec8e5ff4d568b1e099fa10e6f12 : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b16a82310bd48a992e8396da27c2283 : SUCCESS in 12m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaeca2d533e64244869495ccaaf09d69 : SUCCESS in 4m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6a13f8f78aa94ddc85cc7697fceec0d9 : SUCCESS in 52m 33s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/12f07b255b954b1baa06b256320a5fa1 : SUCCESS in 26m 36s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fd157f9df11345d6b43afaaaf774d73c : SUCCESS in 29m 10s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a78a3989858945cd8d651574f29d712f : FAILURE in 38m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/06ce66219d794f3690ff484d11e03942 : SUCCESS in 55m 40s\n- nova-next https://zuul.opendev.org/t/openstack/build/1b271540403c48449ac61492bbded1e8 : SUCCESS in 2h 09m 25s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9eeeb45ab3b5490b839f7218ae33a28e : SUCCESS in 3m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/02ff1118a4b4448fae0c0ed380ed4eba : SUCCESS in 48m 41s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/09a29af3ad67436ea23e9ccf51914bc1 : SUCCESS in 57m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe73ecd4dd874aada52552535d57d6b7 : SUCCESS in 27m 50s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/cd937fbd9a744a9aab6905992a4a5f77 : SUCCESS in 10m 08s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6564fee7c9fd4a8da65edd0588621e45 : SUCCESS in 29m 42s","accounts_in_message":[],"_revision_number":15},{"id":"19e112254c871df22ecb6895b9f5f0808c0094b5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-27 21:49:38.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"1356d915c8fb27fda83ce588950073c07b97b78b","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 21:57:12.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/9c6a07c55aaf41c58a394a95fdee4f48\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/950a7b4e0db84484bb4129595a54dbd5 : FAILURE in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"431e4882967ffbfd8852cbb361c077f3f924a89d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 22:31:28.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/777133c3c3f14efea60d47613057a872\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bf97cbd9f84f4e8aa3413d79b4c25ae2 : SUCCESS in 21m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ae938837b87d4151aa10ca3cdb38a5e3 : SUCCESS in 22m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1dcd07db99ae475097371f3ac2e0b4c2 : SUCCESS in 37m 55s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"985fd02cfb74a0346c997de70a1064425fafea52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 23:50:52.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4725285aa87342d2b6c6e7bea31c2fed\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1798c382c4684897b3e40ec21c103290 : SUCCESS in 45m 50s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f2535d43802947b0b85756858189e7da : SUCCESS in 1h 39m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cca0686e2f6d491a9d888071f21e8b5a : SUCCESS in 1h 01m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef966fcf5c2d4537afe7d37e3c5c975e : SUCCESS in 1h 16m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e6e3e4d625a4d1c9825a43f0661adda : SUCCESS in 7m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40f442e6ce0a4a96ab7cd23905ac5767 : SUCCESS in 14m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15daebfa65354b069d957edc4634150b : SUCCESS in 13m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/24e58b8bd5c34566a30e4fb1046fb9a7 : SUCCESS in 14m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/252984b48a57428ca78c74dcf1e8890a : SUCCESS in 14m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fe34e7490644bc6b7555af0c983ab66 : SUCCESS in 8m 45s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d9b1bc313424404d9ec72e9c43c0d053 : SUCCESS in 57m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/755b23f9d6c44875b805a4887ade3490 : SUCCESS in 1h 37m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e759473d14314518a4fc7bb6b84bb681 : SUCCESS in 52m 50s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/3e1a4fbe7a2c42c0beef9f508130981b : SUCCESS in 37m 36s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/3fe45430779c49f7888c14e78d90a584 : SUCCESS in 50m 56s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6b3cf3f2cef1424382a90ea68bf29bcc : SUCCESS in 54m 26s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3916a5637c8f40b6a8df1efaa498a18e : SUCCESS in 49m 55s\n- nova-next https://zuul.opendev.org/t/openstack/build/0df3aaec8b494a5997f8399e016a39d9 : SUCCESS in 1h 54m 45s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/3e9e130f0d054f12b3fe30b386e49c7a : SUCCESS in 37m 52s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/66ad4fbbe2854c899f640d4b1f7d60fb : SUCCESS in 2m 55s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/49bdc7ba2d7447bf83e98f01c7f7c34e : SUCCESS in 22m 03s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5fb451339fd647078ec0b3b1d94b1df7 : SUCCESS in 25m 36s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3187de41f27f48b2b094b1a12ef38c47 : SUCCESS in 41m 45s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6028200771714fc488f1721a687ae8eb : SUCCESS in 1h 28m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1257de5bc5d2485e8bd934795a4d7cc7 : SUCCESS in 24m 40s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/27de3c98a194424ca9f857bc8337064d : SUCCESS in 56m 48s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d56fbdc7cd194ba2b756c3fc127115d4 : SUCCESS in 28m 01s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/5e1a65f5f8e845bf8e129cbed72780e5 : SUCCESS in 14m 56s","accounts_in_message":[],"_revision_number":15},{"id":"59d761d97f81a0c1bc58f7293a575214562f44e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 23:51:26.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"bbc18de624cebaf2a9ec3560ec00482087ff4f73","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-28 00:01:32.000000000","message":"Patch Set 15:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":15},{"id":"59606d7212211cd002a72e6c7ab2562fbd13c187","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 01:49:17.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8856e4e663d44e8b6e271a23d013c12\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2758b62c5c7e42aa828b60541ad6d730 : SUCCESS in 1h 05m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/93a7e18905b44087b124dcdd582844d8 : SUCCESS in 1h 41m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ffdfd3c5a33c4800ad6f4d1b4c5758a9 : SUCCESS in 31m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7788ca83973c4a6293c6c535fb9f8c31 : SUCCESS in 7m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a5f558ed5a54cb6b42e1b9330bf0538 : SUCCESS in 9m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32e801d3ed0b4302a3d84d846c950957 : SUCCESS in 14m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ad06c3fa7c94af5a9865433cbb85617 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e18010dfaa349e6a0e2c908263c863a : SUCCESS in 8m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/67033f6075644db585b4bc9784dc4647 : SUCCESS in 50m 45s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a6ecf6874b4a4fe5b641dc460d0ed442 : SUCCESS in 32m 34s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0e6bcbc5acbc429c8b576e8315721ef6 : SUCCESS in 28m 14s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/2208a5675abb4d62afc8c63cbbc77060 : SUCCESS in 38m 18s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fde50b1a3c1e40eca36cd2db1664d051 : FAILURE in 1h 57m 13s\n- nova-next https://zuul.opendev.org/t/openstack/build/33d2ba19785a45bd864afd67885939d0 : SUCCESS in 1h 05m 06s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/dbfe1ba94e4e44b5bc8eaa54f0d5babf : SUCCESS in 3m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/55ec5df9a2ec47d1aa0e50eb1a8d557c : SUCCESS in 1h 35m 16s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c51b265fc2674fadbcd9823f1456a8d4 : SUCCESS in 48m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f7ed78073f843599413a13ddf1529a9 : SUCCESS in 55m 54s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/04d5016223b24d99b69543850d37176b : SUCCESS in 15m 05s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/e14503ad2a68450e972a437de64b4896 : SUCCESS in 48m 59s","accounts_in_message":[],"_revision_number":15},{"id":"6d7eb114fd407f83c5d1927ac091ffb94ad4cd5b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-28 04:39:54.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"6763db3f725ea6a3fe37174953d9bfd36b5e234e","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-28 04:46:04.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/a56d8cfb723f459e89e79b7dffcc7e3a\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/1511b289f35746e68972223c00ce5334 : FAILURE in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"39d89a6ebefdb46f777d30467fccdc909cea8961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 06:12:58.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb25a0d155324efb96ec00cd0004313b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a6c7b05492e54747ab3d0ff295a43bf3 : SUCCESS in 37m 10s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0719fe9dba5d440792f6e647afdf8ba8 : SUCCESS in 54m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a41ae99fcad64751b74aa155ba899ae3 : SUCCESS in 33m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57060ba2d78e45a2b2d382bd07904dbf : SUCCESS in 1h 01m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc8346743d5946a3a459ae0045fee176 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2dd7354db4b84594a4d057166706c83b : SUCCESS in 10m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c399e07db07475c841b53e229b535c9 : SUCCESS in 8m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9fc643841f744cb394130a51d519ecff : SUCCESS in 13m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f81b96beb44f5a87929d296134c0bf : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6244ec8cf22c4fecb731537ad990d80e : SUCCESS in 6m 52s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/311274740f67487ba68cb561272df919 : SUCCESS in 27m 04s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/066d27825d804717a0426a02008084fb : SUCCESS in 1h 25m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f954e28299914962be744836adec03aa : SUCCESS in 46m 48s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/5e1efd2928884a9dbf0112e92ac7cf66 : SUCCESS in 37m 18s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/05336a4e880e4a7084348a2504e04c47 : SUCCESS in 1h 03m 11s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e823e6b7e4fe456482aef391826f7ea7 : SUCCESS in 52m 31s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7a2dba9afa3045b1ae71da102d01e850 : SUCCESS in 1h 00m 43s\n- nova-next https://zuul.opendev.org/t/openstack/build/6da9c682d8fd4543a13d394aec3727f5 : SUCCESS in 1h 03m 37s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/c8f12949646c4cd7b5cae3461e0bcb38 : SUCCESS in 28m 16s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/78607d8370654702aa8471b6841adb03 : SUCCESS in 4m 46s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b74d65e222db4bd591090c895106eafd : SUCCESS in 28m 30s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f160388eb9ee43d98c9a0d5fe6737a62 : SUCCESS in 28m 46s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/6844042b098843349c34b8d569f57240 : FAILURE in 33m 56s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/39b70e7b586041e986e59a8309963a58 : FAILURE in 53m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cf46bb16bf44fc9bf2beca8d404f769 : SUCCESS in 45m 33s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0867ddb7827b42beab76702bb354884e : SUCCESS in 25m 39s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7c019e3a43dd4a13aa0e19dca1389438 : SUCCESS in 29m 09s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/69bc5b13de164912b36e2aa98e88414b : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":15},{"id":"e19382649c28b6d2d41e3e2d02e73883625729e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 06:19:31.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/610caba094254b7e8945fd431ba641dc\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1b350d5873384d588dbae0a684d74ea1 : SUCCESS in 28m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b8e218acd256433a98789f938d7c64db : NODE_FAILURE Node(set) request 0bde4ab4f68a486486922efca5a0636e failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c485bb6c3d80408b89db53947facf312 : SUCCESS in 42m 31s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"6f1dd0f2684c03c16222adb9c75b890dd88a81fd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-28 14:48:31.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"5b7b8c0c1c43c2b2ab638d1c50ad183adf0ee0e0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-28 14:54:33.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/27c9324b343c447cb3f778f0d1ad6e82\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/b336407e834e45cea6f60d7cd629a8b5 : FAILURE in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"923619010f673ae652bd7ae8dc4c4d9e1c79c654","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 15:25:39.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f11a4ad79a3c48499c69170d498bfd77\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/47b2a7249a334cf1af5bb30e984fa33a : SUCCESS in 14m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6800ec157ec449d5af4f5e6ad6723be3 : SUCCESS in 17m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9574508b8a3a4185845942553069d7f0 : SUCCESS in 19m 45s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"fe9c80b50c19b6b4078dbaa8d7e64249451cf480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 16:33:17.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac9c6beb97d44c6fb6f43aec6964a499\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea084a43eb67444da9237977deefbbfd : SUCCESS in 1h 08m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e05c0c1cc3a74c41985172c268169973 : SUCCESS in 47m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4359ac523ca47d08ea13f39dc3cc3f8 : SUCCESS in 44m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8375bb2b056b47b98ba6238b6a57f604 : SUCCESS in 57m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c0e501470074b2f9a4330bd4ddcbbfe : SUCCESS in 9m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/288c3abf2ac9459dac7ad22e0703b013 : SUCCESS in 9m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c88d055dc9e4b299dc7882be6258097 : SUCCESS in 11m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6b59e763777a4e2ea6ff9c043766a58c : SUCCESS in 20m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d86f9d4835bf4143ace97c2f61147d8a : SUCCESS in 13m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f28ba6e687c4cd7aa98a71fe49d18c1 : SUCCESS in 4m 01s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/b2c0905f7666453ea78de0f67d2d036c : SUCCESS in 45m 39s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ab261a2240794b34956c18539c96a64b : SUCCESS in 46m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7ba55977b41b463b9bf77d2917f33e91 : SUCCESS in 48m 08s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/8f35ffc8b56b4b63b648f2ed51b4e929 : SUCCESS in 36m 59s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/4ca84ef2c3f343e887d5bae7834f385b : SUCCESS in 1h 01m 47s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0b5b9f302aa64173b1d8eb4c3ff038ed : SUCCESS in 38m 15s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b72dab48965245d1b5190b790e00b79b : SUCCESS in 1h 39m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/975cd0eca62a47799f26c2924ddfac36 : SUCCESS in 1h 02m 42s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/4712af2f788f4b3eb240f69b0d1861f4 : SUCCESS in 48m 06s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f834fef749fa4056955b3d42281c0f25 : SUCCESS in 3m 30s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3e94b5da13b6453d9e3ffcaac81626a1 : SUCCESS in 28m 10s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6f463e0004fb439eab263209528a939d : SUCCESS in 28m 08s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/5086e9556725400aba6f179cabeec51c : SUCCESS in 34m 16s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d1eaaf74c9349caaf8df37553e29723 : SUCCESS in 1h 12m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34b1c126a329448ab7ec9f0958efd91c : SUCCESS in 26m 57s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0eed071f963748a58fe41d8164543a64 : SUCCESS in 28m 10s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ecd297f2abb14fa1a1bea5880fac0e56 : SUCCESS in 18m 09s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/0a5488584a1f4f0296d936203d8cbed1 : SUCCESS in 11m 05s","accounts_in_message":[],"_revision_number":15},{"id":"05cd54b7f838e0335a894e4651f2826fb10584a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 16:34:01.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"c5359bee1e7249bfe884d4958455a1f48cc66aee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-28 16:43:58.000000000","message":"Patch Set 15:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":15},{"id":"63dc35e81e37cd36b379454b30c3edaad969fead","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 18:15:29.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b531d709908c47a799a5dd4a498fe8dc\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/432f836187344e298a64eac113553cc3 : SUCCESS in 49m 51s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e1768ce5554c4f3dab392a522187416b : SUCCESS in 1h 22m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c33820084b18469792b0363e7462ab77 : SUCCESS in 29m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74263cbef900419f9a26208943ad3a76 : SUCCESS in 9m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee43cee448d64b3f9c9654b40f74e197 : SUCCESS in 10m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/74b9ab17424547d5ae862de0e8505418 : SUCCESS in 14m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eaa13533cb84faab79beacd38209863 : SUCCESS in 13m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19733345f2df4df6b2afdb3fc790b8a5 : SUCCESS in 9m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/60616a68cda64bcb8eaa5246d4c11c16 : SUCCESS in 34m 38s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3d217300d658432c9a66826771ae090c : SUCCESS in 33m 40s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e161d133d32543de992c021ccff5ddc3 : SUCCESS in 29m 40s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/342146a4653e43eb9884f28e39c6c07b : SUCCESS in 32m 33s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5f3777ee9a904a4abe941364aa6e7142 : SUCCESS in 1h 11m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/1ec6384c02f34135bca10b40490a9fe1 : SUCCESS in 1h 00m 30s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/89c3b440ec5b47d7a96c1323bada4a1e : SUCCESS in 3m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f1e53879f1e54b1dbd3b68719b43509b : SUCCESS in 1h 36m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a8027402fb84b83aa494a540d01cc66 : SUCCESS in 1h 19m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65a30d47c1f14766ae8da820927d6a6f : SUCCESS in 56m 03s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/fb34f2ed001f4b359de0b9a382963b67 : SUCCESS in 14m 01s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/0beb9fdfc5b643ff90c7c6cd09397784 : SUCCESS in 26m 58s","accounts_in_message":[],"_revision_number":15},{"id":"e5fbe5178b3b469120f01757728bab19a8cfa2d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 18:15:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"af7fc47bd185b875cf12fae09668a58c17b7cb27","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 18:16:41.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0332a5bafc9148d2bbb1bba8b9b3f064\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d91e920418dc4fb8aa13a224d3ca98b7 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"f712a55eeb21ddeb6447c243e8faf964db634e77","revisions":{"ba0f53ebb134b3dbb455864c7dead3863a4dc056":{"kind":"REWORK","_number":1,"created":"2026-08-08 20:02:22.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/1000211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/1000211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/1000211/1"}}},"commit":{"parents":[{"commit":"2002a7198bebea694d18d0db255990090a26227c","subject":"Unify the thread poool executors into a central factory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2002a7198bebea694d18d0db255990090a26227c"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"subject":"Shutdown thread pool executors during service shutdown","message":"Shutdown thread pool executors during service shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done at the end of\neverything else and with wait\u003dfalse so that it will let in-progress\ntasks to complete and will not block shutdown if any task still\nrunning on any executor.\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: I59a3091d53e3e6ab990e80ed3974b2153569b2bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba0f53ebb134b3dbb455864c7dead3863a4dc056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba0f53ebb134b3dbb455864c7dead3863a4dc056"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2002a7198bebea694d18d0db255990090a26227c","is_merged_in_target_branch":false,"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","change_number":998571,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"93ec1286e8c2b663063656fda019c65e6fd12b1b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-11 04:14:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/1000211/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/1000211/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/1000211/2"}}},"commit":{"parents":[{"commit":"f6a4aff20132ed07ab30cdfc5211da76135dc055","subject":"Unify the thread poool executors into a central factory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f6a4aff20132ed07ab30cdfc5211da76135dc055"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-11 04:14:56.000000000","tz":0},"subject":"Shutdown thread pool executors during service shutdown","message":"Shutdown thread pool executors during service shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done at the end of\neverything else and with wait\u003dfalse so that it will let in-progress\ntasks to complete and will not block shutdown if any task still\nrunning on any executor.\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: I59a3091d53e3e6ab990e80ed3974b2153569b2bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/93ec1286e8c2b663063656fda019c65e6fd12b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/93ec1286e8c2b663063656fda019c65e6fd12b1b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f6a4aff20132ed07ab30cdfc5211da76135dc055","is_merged_in_target_branch":false,"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","change_number":998571,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"2002a7198bebea694d18d0db255990090a26227c","ours":"ba0f53ebb134b3dbb455864c7dead3863a4dc056","theirs":"f6a4aff20132ed07ab30cdfc5211da76135dc055","merge_strategy":"recursive","contains_conflicts":false}},"23203294405ad8d8522e5404d2107ebc064ecd33":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-08-11 04:27:24.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/1000211/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/1000211/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/1000211/3"}}},"commit":{"parents":[{"commit":"b1b8b6062a695339e37802a1b6ef2aa5771d2ec4","subject":"Unify the thread poool executors into a central factory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1b8b6062a695339e37802a1b6ef2aa5771d2ec4"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-11 04:27:24.000000000","tz":0},"subject":"Shutdown thread pool executors during service shutdown","message":"Shutdown thread pool executors during service shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done at the end of\neverything else and with wait\u003dfalse so that it will let in-progress\ntasks to complete and will not block shutdown if any task still\nrunning on any executor.\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: I59a3091d53e3e6ab990e80ed3974b2153569b2bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/23203294405ad8d8522e5404d2107ebc064ecd33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/23203294405ad8d8522e5404d2107ebc064ecd33"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b1b8b6062a695339e37802a1b6ef2aa5771d2ec4","is_merged_in_target_branch":false,"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","change_number":998571,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"f6a4aff20132ed07ab30cdfc5211da76135dc055","ours":"93ec1286e8c2b663063656fda019c65e6fd12b1b","theirs":"b1b8b6062a695339e37802a1b6ef2aa5771d2ec4","merge_strategy":"recursive","contains_conflicts":false}},"e60a4bbb43166912d9cfac2e30bbf3a847659386":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-08-13 04:22:05.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/1000211/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/1000211/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/1000211/4"}}},"commit":{"parents":[{"commit":"af7bbc157227ad984c28328bec1853025bef9a20","subject":"Unify the thread pool executors into a central factory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/af7bbc157227ad984c28328bec1853025bef9a20"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-13 04:22:05.000000000","tz":0},"subject":"Shutdown thread pool executors during service shutdown","message":"Shutdown thread pool executors during service shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done at the end of\neverything else and with wait\u003dfalse so that it will let in-progress\ntasks to complete and will not block shutdown if any task still\nrunning on any executor.\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: I59a3091d53e3e6ab990e80ed3974b2153569b2bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e60a4bbb43166912d9cfac2e30bbf3a847659386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e60a4bbb43166912d9cfac2e30bbf3a847659386"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"af7bbc157227ad984c28328bec1853025bef9a20","is_merged_in_target_branch":false,"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","change_number":998571,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"b1b8b6062a695339e37802a1b6ef2aa5771d2ec4","ours":"23203294405ad8d8522e5404d2107ebc064ecd33","theirs":"af7bbc157227ad984c28328bec1853025bef9a20","merge_strategy":"recursive","contains_conflicts":false}},"f4334fe94f4a0ddb0e3aecf5175b5438d737529d":{"kind":"REWORK","_number":5,"created":"2026-08-13 05:07:17.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/1000211/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/1000211/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/1000211/5"}}},"commit":{"parents":[{"commit":"af7bbc157227ad984c28328bec1853025bef9a20","subject":"Unify the thread pool executors into a central factory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/af7bbc157227ad984c28328bec1853025bef9a20"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-13 05:05:26.000000000","tz":-420},"subject":"Shutdown thread pool executors during graceful shutdown","message":"Shutdown thread pool executors during graceful shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done in manager\u0027s\ncleanup_host() and after self.driver.cleanup_host() as that\nshutdown(wait\u003dTrue) the delayed_executors which will need the main\nexecutor to be up to finish the in-progress things. That is the last\nthings manager needs the executors for and safe to shutdown the\ncompute.\n\nFor other services, it will be done at the end of manager\u0027s\ngraceful_shutdown()\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: I59a3091d53e3e6ab990e80ed3974b2153569b2bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f4334fe94f4a0ddb0e3aecf5175b5438d737529d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f4334fe94f4a0ddb0e3aecf5175b5438d737529d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"af7bbc157227ad984c28328bec1853025bef9a20","is_merged_in_target_branch":false,"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","change_number":998571,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6f7a16e60f9bbcc4d05152678bea61b3f438039c":{"kind":"REWORK","_number":6,"created":"2026-08-13 19:40:09.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/1000211/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/1000211/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/1000211/6"}}},"commit":{"parents":[{"commit":"af7bbc157227ad984c28328bec1853025bef9a20","subject":"Unify the thread pool executors into a central factory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/af7bbc157227ad984c28328bec1853025bef9a20"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-13 19:39:56.000000000","tz":-420},"subject":"Shutdown thread pool executors during graceful shutdown","message":"Shutdown thread pool executors during graceful shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done in manager\u0027s\ncleanup_host() and after self.driver.cleanup_host() as that\nshutdown(wait\u003dTrue) the delayed_executors which will need the main\nexecutor to be up to finish the in-progress things. That is the last\nthings manager needs the executors for and safe to shutdown the\ncompute.\n\nFor other services, it will be done at the end of manager\u0027s\ngraceful_shutdown()\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: I59a3091d53e3e6ab990e80ed3974b2153569b2bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6f7a16e60f9bbcc4d05152678bea61b3f438039c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6f7a16e60f9bbcc4d05152678bea61b3f438039c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"af7bbc157227ad984c28328bec1853025bef9a20","is_merged_in_target_branch":false,"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","change_number":998571,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8813ee935d2b95c282ac9d8ad5e4d7a5e429bf75":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-08-14 14:01:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/1000211/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/1000211/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/1000211/7"}}},"commit":{"parents":[{"commit":"7c7c68136de2f63de76579fbc6c37bb3ea68333a","subject":"Unify the thread pool executors into a central factory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7c7c68136de2f63de76579fbc6c37bb3ea68333a"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-14 14:01:49.000000000","tz":0},"subject":"Shutdown thread pool executors during graceful shutdown","message":"Shutdown thread pool executors during graceful shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done in manager\u0027s\ncleanup_host() and after self.driver.cleanup_host() as that\nshutdown(wait\u003dTrue) the delayed_executors which will need the main\nexecutor to be up to finish the in-progress things. That is the last\nthings manager needs the executors for and safe to shutdown the\ncompute.\n\nFor other services, it will be done at the end of manager\u0027s\ngraceful_shutdown()\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: I59a3091d53e3e6ab990e80ed3974b2153569b2bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8813ee935d2b95c282ac9d8ad5e4d7a5e429bf75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8813ee935d2b95c282ac9d8ad5e4d7a5e429bf75"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7c7c68136de2f63de76579fbc6c37bb3ea68333a","is_merged_in_target_branch":false,"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","change_number":998571,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"af7bbc157227ad984c28328bec1853025bef9a20","ours":"6f7a16e60f9bbcc4d05152678bea61b3f438039c","theirs":"7c7c68136de2f63de76579fbc6c37bb3ea68333a","merge_strategy":"recursive","contains_conflicts":false}},"4bba3d4073ee8323d960004b68ace20bdb636a15":{"kind":"REWORK","_number":8,"created":"2026-08-14 17:58:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/1000211/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/1000211/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/1000211/8"}}},"commit":{"parents":[{"commit":"7c7c68136de2f63de76579fbc6c37bb3ea68333a","subject":"Unify the thread pool executors into a central factory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7c7c68136de2f63de76579fbc6c37bb3ea68333a"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-14 17:57:54.000000000","tz":-420},"subject":"Shutdown thread pool executors during graceful shutdown","message":"Shutdown thread pool executors during graceful shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done in manager\u0027s\ncleanup_host() and after self.driver.cleanup_host() as that\nshutdown(wait\u003dTrue) the delayed_executors which will need the main\nexecutor to be up to finish the in-progress things. That is the last\nthings manager needs the executors for and safe to shutdown the\ncompute.\n\nFor other services, it will be done at the end of manager\u0027s\ngraceful_shutdown()\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: I59a3091d53e3e6ab990e80ed3974b2153569b2bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4bba3d4073ee8323d960004b68ace20bdb636a15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4bba3d4073ee8323d960004b68ace20bdb636a15"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7c7c68136de2f63de76579fbc6c37bb3ea68333a","is_merged_in_target_branch":false,"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","change_number":998571,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d060fb5a8405791b5e3f003b1663b3d9129fa3c0":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-08-14 18:30:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/1000211/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/1000211/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/1000211/9"}}},"commit":{"parents":[{"commit":"8af9f3bbf9440fdeab2c9b6a31f6e490aeb6087c","subject":"Unify the thread pool executors into a central factory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8af9f3bbf9440fdeab2c9b6a31f6e490aeb6087c"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-14 18:30:15.000000000","tz":0},"subject":"Shutdown thread pool executors during graceful shutdown","message":"Shutdown thread pool executors during graceful shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done in manager\u0027s\ncleanup_host() and after self.driver.cleanup_host() as that\nshutdown(wait\u003dTrue) the delayed_executors which will need the main\nexecutor to be up to finish the in-progress things. That is the last\nthings manager needs the executors for and safe to shutdown the\ncompute.\n\nFor other services, it will be done at the end of manager\u0027s\ngraceful_shutdown()\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: I59a3091d53e3e6ab990e80ed3974b2153569b2bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d060fb5a8405791b5e3f003b1663b3d9129fa3c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d060fb5a8405791b5e3f003b1663b3d9129fa3c0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8af9f3bbf9440fdeab2c9b6a31f6e490aeb6087c","is_merged_in_target_branch":false,"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","change_number":998571,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"7c7c68136de2f63de76579fbc6c37bb3ea68333a","ours":"4bba3d4073ee8323d960004b68ace20bdb636a15","theirs":"8af9f3bbf9440fdeab2c9b6a31f6e490aeb6087c","merge_strategy":"recursive","contains_conflicts":false}},"fe632a8dd269051f6f728446bffefb9ba741fb02":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-08-19 02:50:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/1000211/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/1000211/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/1000211/10"}}},"commit":{"parents":[{"commit":"0842edaa998d12e77d622dae5aedfb5e0024c16a","subject":"Unify the thread pool executors into a central factory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0842edaa998d12e77d622dae5aedfb5e0024c16a"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-19 02:50:21.000000000","tz":0},"subject":"Shutdown thread pool executors during graceful shutdown","message":"Shutdown thread pool executors during graceful shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done in manager\u0027s\ncleanup_host() and after self.driver.cleanup_host() as that\nshutdown(wait\u003dTrue) the delayed_executors which will need the main\nexecutor to be up to finish the in-progress things. That is the last\nthings manager needs the executors for and safe to shutdown the\ncompute.\n\nFor other services, it will be done at the end of manager\u0027s\ngraceful_shutdown()\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: I59a3091d53e3e6ab990e80ed3974b2153569b2bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fe632a8dd269051f6f728446bffefb9ba741fb02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fe632a8dd269051f6f728446bffefb9ba741fb02"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0842edaa998d12e77d622dae5aedfb5e0024c16a","is_merged_in_target_branch":true,"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","change_number":998571,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"8af9f3bbf9440fdeab2c9b6a31f6e490aeb6087c","ours":"d060fb5a8405791b5e3f003b1663b3d9129fa3c0","theirs":"0842edaa998d12e77d622dae5aedfb5e0024c16a","merge_strategy":"recursive","contains_conflicts":false}},"85b3dbd9f27abde2578dc290c948b7163f24ee70":{"kind":"REWORK","_number":11,"created":"2026-08-19 16:50:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/1000211/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/1000211/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/1000211/11"}}},"commit":{"parents":[{"commit":"0842edaa998d12e77d622dae5aedfb5e0024c16a","subject":"Unify the thread pool executors into a central factory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0842edaa998d12e77d622dae5aedfb5e0024c16a"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-19 16:50:56.000000000","tz":0},"subject":"Shutdown thread pool executors during graceful shutdown","message":"Shutdown thread pool executors during graceful shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done in manager\u0027s\ncleanup_host() and after self.driver.cleanup_host() as that\nshutdown(wait\u003dTrue) the delayed_executors which will need the main\nexecutor to be up to finish the in-progress things. That is the last\nthings manager needs the executors for and safe to shutdown the\ncompute.\n\nFor other services, it will be done at the end of manager\u0027s\ngraceful_shutdown()\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: I59a3091d53e3e6ab990e80ed3974b2153569b2bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/85b3dbd9f27abde2578dc290c948b7163f24ee70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/85b3dbd9f27abde2578dc290c948b7163f24ee70"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0842edaa998d12e77d622dae5aedfb5e0024c16a","is_merged_in_target_branch":true,"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","change_number":998571,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a113345e15bc9738f128b2891ec7aa41f1843c1c":{"kind":"REWORK","_number":12,"created":"2026-08-19 17:11:09.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/1000211/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/1000211/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/1000211/12"}}},"commit":{"parents":[{"commit":"0842edaa998d12e77d622dae5aedfb5e0024c16a","subject":"Unify the thread pool executors into a central factory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0842edaa998d12e77d622dae5aedfb5e0024c16a"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-19 17:10:45.000000000","tz":-420},"subject":"Shutdown thread pool executors during graceful shutdown","message":"Shutdown thread pool executors during graceful shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done in manager\u0027s\ncleanup_host() and after self.driver.cleanup_host() as that\nshutdown(wait\u003dTrue) the delayed_executors which will need the main\nexecutor to be up to finish the in-progress things. That is the last\nthings manager needs the executors for and safe to shutdown the\ncompute.\n\nFor other services, it will be done at the end of manager\u0027s\ngraceful_shutdown()\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: I59a3091d53e3e6ab990e80ed3974b2153569b2bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a113345e15bc9738f128b2891ec7aa41f1843c1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a113345e15bc9738f128b2891ec7aa41f1843c1c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0842edaa998d12e77d622dae5aedfb5e0024c16a","is_merged_in_target_branch":true,"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","change_number":998571,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6862570c979b8999b11c0a9814df9e9f30370e31":{"kind":"REWORK","_number":13,"created":"2026-08-19 21:20:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/1000211/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/1000211/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/1000211/13"}}},"commit":{"parents":[{"commit":"0842edaa998d12e77d622dae5aedfb5e0024c16a","subject":"Unify the thread pool executors into a central factory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0842edaa998d12e77d622dae5aedfb5e0024c16a"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-19 21:20:06.000000000","tz":-420},"subject":"Shutdown thread pool executors during graceful shutdown","message":"Shutdown thread pool executors during graceful shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done in manager\u0027s\ncleanup_host() and after self.driver.cleanup_host() as that\nshutdown(wait\u003dTrue) the delayed_executors which will need the main\nexecutor to be up to finish the in-progress things. That is the last\nthings manager needs the executors for and safe to shutdown the\ncompute.\n\nFor other services, it will be done at the end of manager\u0027s\ngraceful_shutdown()\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: I59a3091d53e3e6ab990e80ed3974b2153569b2bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6862570c979b8999b11c0a9814df9e9f30370e31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6862570c979b8999b11c0a9814df9e9f30370e31"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0842edaa998d12e77d622dae5aedfb5e0024c16a","is_merged_in_target_branch":true,"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","change_number":998571,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"28dfa7616ac157ede50dd9e85c146989de16b10d":{"kind":"REWORK","_number":14,"created":"2026-08-20 03:03:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/1000211/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/1000211/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/1000211/14"}}},"commit":{"parents":[{"commit":"0842edaa998d12e77d622dae5aedfb5e0024c16a","subject":"Unify the thread pool executors into a central factory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0842edaa998d12e77d622dae5aedfb5e0024c16a"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-20 03:02:45.000000000","tz":-420},"subject":"Shutdown thread pool executors during graceful shutdown","message":"Shutdown thread pool executors during graceful shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done in manager\u0027s\ncleanup_host() and after self.driver.cleanup_host() as that\nshutdown(wait\u003dTrue) the delayed_executors which will need the main\nexecutor to be up to finish the in-progress things. That is the last\nthings manager needs the executors for and safe to shutdown the\ncompute.\n\nFor other services, it will be done at the end of manager\u0027s\ngraceful_shutdown()\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: I59a3091d53e3e6ab990e80ed3974b2153569b2bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/28dfa7616ac157ede50dd9e85c146989de16b10d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/28dfa7616ac157ede50dd9e85c146989de16b10d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0842edaa998d12e77d622dae5aedfb5e0024c16a","is_merged_in_target_branch":true,"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","change_number":998571,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f712a55eeb21ddeb6447c243e8faf964db634e77":{"kind":"REWORK","_number":15,"created":"2026-08-26 00:20:11.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/1000211/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/1000211/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/1000211/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/1000211/15"}}},"commit":{"parents":[{"commit":"0842edaa998d12e77d622dae5aedfb5e0024c16a","subject":"Unify the thread pool executors into a central factory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0842edaa998d12e77d622dae5aedfb5e0024c16a"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:58:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-26 00:20:02.000000000","tz":-420},"subject":"Shutdown thread pool executors during graceful shutdown","message":"Shutdown thread pool executors during graceful shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done in manager\u0027s\ncleanup_host() and after self.driver.cleanup_host() as that\nshutdown(wait\u003dTrue) the delayed_executors which will need the main\nexecutor to be up to finish the in-progress things. That is the last\nthings manager needs the executors for and safe to shutdown the\ncompute.\n\nFor other services, it will be done at the end of manager\u0027s\ngraceful_shutdown()\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: I59a3091d53e3e6ab990e80ed3974b2153569b2bf\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f712a55eeb21ddeb6447c243e8faf964db634e77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f712a55eeb21ddeb6447c243e8faf964db634e77"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0842edaa998d12e77d622dae5aedfb5e0024c16a","is_merged_in_target_branch":true,"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","change_number":998571,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
