)]}'
{"id":"openstack%2Fnova~1001212","triplet_id":"openstack%2Fnova~master~Ice578bc491dac41ae4de5ef6f2f0397515f0bd26","project":"openstack/nova","branch":"master","topic":"bug/2163448","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-08-26 12:51:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ice578bc491dac41ae4de5ef6f2f0397515f0bd26","subject":"Improve the graceful shutdown job ansible tasks name","status":"MERGED","created":"2026-08-17 18:38:31.000000000","updated":"2026-08-26 12:52:27.000000000","submitted":"2026-08-26 12:51:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1001212-bug/2163448","meta_rev_id":"91a9d73abf8901f5dbe0162f9a716a823c232536","_number":1001212,"virtual_id_number":1001212,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 12:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2026-08-18 21:01:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2026-08-25 08:49:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2026-08-25 08:49:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-17 19:01:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-17 19:33:51.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-17 20:27:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-18 03:35:53.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-08-18 21:01:35.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2026-08-25 08:49:18.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-08-25 12:04:56.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"a0e900a9d3374615e6db292fb20b72e750c15744","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-17 18:38:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e4a63b855695cc924854e0ff8fe3e0cc9ebad9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 19:01:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa7f44cabca84ed783393f622f4f39f5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fef9882bfb4a4ea5b337b1c8c6fbd7d0 : SUCCESS in 18m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/da71cdc0fc0a4f1c9464402d2fb58e87 : SUCCESS in 17m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ea3b08ca6c0249bd97a1ca78636c8ef1 : NODE_FAILURE Node(set) request 6b316ec7354d4ff89e8a8fb68265e1d2 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73c4edab7a7dd2cba5d64e41452a4929dd4d2906","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-17 19:33:51.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ice578bc491dac41ae4de5ef6f2f0397515f0bd26 .","accounts_in_message":[],"_revision_number":1},{"id":"bd19267af8775d244bcf8b6f1c902d975c034978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 20:27:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a245ef8108c0440f90bf7285d64a73d5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dabc63d539d74c688d92a3b8b10e69cf : SUCCESS in 42m 16s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/31c5b55d116144a2a23498f1c7c0e2f2 : SUCCESS in 1h 42m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fbf09f2c1a34b9eb31b9d6e6a09ace7 : SUCCESS in 36m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cea503a105bb43d9a934588effc1dbf9 : SUCCESS in 1h 28m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f92462c211c9431ea6730d9794c0745f : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f674277bae6c4f0a940dec131b392527 : SUCCESS in 16m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b67cc6e27c34f9e94dbc0f9934d984a : SUCCESS in 14m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9749520d5b814ae3b9abf344a34e4f1a : SUCCESS in 17m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ec08458ee1341b5bb3a1b3f2c78c3bc : SUCCESS in 13m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3758e66f47834da7a04fb9c24ac060ac : SUCCESS in 8m 59s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/f6faa2a2791c4305ae3c35d396897925 : SUCCESS in 35m 05s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/721bf71317b048ffa20dc02c7aed6781 : SUCCESS in 1h 24m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8bbe85ca3428414db2a54a2d19be16d5 : SUCCESS in 33m 26s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/5c8a4b3ff29d41fb98eeddbb0d903537 : SUCCESS in 32m 47s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/9207dd4091cc4c7498facf7c9a365d63 : SUCCESS in 39m 03s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5bb721172aef489e9beb89c2b40782a8 : SUCCESS in 26m 47s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1a220b547e3544eaa31a7d86e15ad8a8 : FAILURE in 1h 17m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/45eab3a29b904bd78dd0c6c4178b0481 : SUCCESS in 1h 34m 43s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6ef611e316e34b64867a8143e4184702 : SUCCESS in 50m 28s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3187ebd018644d8b876ee9eec5f4cffc : SUCCESS in 3m 03s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6153d9e66b6d4532b28ef6129ca46aa8 : SUCCESS in 28m 11s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/330cf358e7454ac6a592b9a1a1260f9f : SUCCESS in 26m 37s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/7b46eeb3bb614b7190c5af5848738f84 : SUCCESS in 32m 30s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/89aea87a8a6e46f49625c877ad6bfb01 : SUCCESS in 1h 29m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4977495b5a534404ae8c691026f8bcb4 : SUCCESS in 1h 03m 54s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/eab534872b3b4e7eacb6ad0690b565af : SUCCESS in 52m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f9d36ee6078247cd910530b58b188dee : SUCCESS in 21m 16s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7e6c7a23c8ca48da9e2ddd95a7e9ca6c : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":1},{"id":"77fd28ffddc47669ac3ec9e58a0d402d5cf62c19","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-18 01:06:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9bb5da297dd96b331ecd7a42f0315f2952baa92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 02:03:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9739c36137324b09985bcd68851af905\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/60e7ac74c138432c803d2c8162405cc2 : FAILURE in 4m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9adff2a59240449dba8dc0ca65a5da74 : SUCCESS in 24m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5b4af4efea2845258b3e5cb533bd6969 : SUCCESS in 37m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79a35fa00a28761c16586428009b08679de07e9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 03:35:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae5cc9a6b6b74843a08cd34994b6abc1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9df1801dcfe744bbbeb23e7110206468 : SUCCESS in 30m 56s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f7c02e28df1d4d6e85145a27aeb9592e : SUCCESS in 1h 36m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9d5f5c373c348d39862bfcc1f0ff781 : SUCCESS in 1h 00m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adf805c644b54bbf8bffbb94d2481e26 : SUCCESS in 1h 18m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee8b17ad1ca34e6191593a38bebbc29a : SUCCESS in 7m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd6063b705a54615885f261152252225 : SUCCESS in 13m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/69b0d61381ff486bb0860a4d6abc3fcb : SUCCESS in 9m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eaee5d881c9c464aac2e1e089049a5f8 : SUCCESS in 18m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3e423c05f6048fba22da2f919a73409 : SUCCESS in 13m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2721e5e62b6c474b80fbcdc8f6215661 : SUCCESS in 5m 03s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/630328b92cc246029eead690ca3b93da : SUCCESS in 31m 39s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/82cae1e6174e488398a6a1e3a9939c86 : SUCCESS in 1h 11m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1a6e077eff6a43d38ceba6ceb56394cb : SUCCESS in 36m 53s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/6d17e384abbe4b419a9ea1811ea5c037 : SUCCESS in 34m 50s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/cdc29852cc8142b48691abb1b27d970d : SUCCESS in 1h 02m 48s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e74a7471a8f947f7b64efd204a7d352f : SUCCESS in 50m 56s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f529b07011e9431d9d46cd3196082f38 : SUCCESS in 1h 46m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/ecc30d92f52647f8a6a74c8c007b0d5a : SUCCESS in 2h 12m 08s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/47b605b9f72e4c84bd9875bcad62c995 : SUCCESS in 48m 44s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bc3daf61b9344407ab60e20f731ea72f : SUCCESS in 6m 01s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4ea55633254a442385b8e7fa4960794e : SUCCESS in 28m 43s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/05b28269c0b14823a04f2742131fc05e : SUCCESS in 23m 09s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c8f9ea230f9746a284a6a14c82f2912f : SUCCESS in 32m 32s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0190f7442f574fa5bdaceaf2e0333e6f : SUCCESS in 1h 13m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba62f08117d5436fb995ea3cb2ff1c93 : SUCCESS in 51m 39s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/b02a6695cbad45ca8a99eb861405f31a : SUCCESS in 56m 03s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/99a371e58fc94230ae173b6944d3047d : SUCCESS in 34m 05s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/076b59af4df24528b191e723f8fd3489 : SUCCESS in 13m 31s","accounts_in_message":[],"_revision_number":1},{"id":"dde1cd63444be06c79272a332ce4810cd24dd222","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-18 03:46:33.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"702e7d2cbf153c65743e773122ca4c652f40020f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-18 19:22:25.000000000","message":"Topic gs removed","accounts_in_message":[],"_revision_number":1},{"id":"cc47f02ee936344cd4bf90a699821e3d915ea427","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-18 19:22:31.000000000","message":"Topic set to bug/2163448","accounts_in_message":[],"_revision_number":1},{"id":"e8f4b84cab2e99137e026a11440e8cd8668c1edb","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-08-18 21:01:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f7d0f9ee792ab3bf678237bdb8325bc59fc8cbf","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-25 08:49:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bac1ce76e87f0fc24369189aeffc983e415c8143","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 08:50:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2984e0623ed556da7deb6272bf6d8abdb683ea7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 10:42:34.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/1001017 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"38b6df778dacec9621bd0a9868ac3960afd8f8d0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-25 12:04:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"984c01368e9f23e36e8a9bf0e0a61d8f313c95bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 14:15:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10968af179ff4eea8d4702133f655c0f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6136dfbbaedb4ce5aa3a58977e9392e3 : SUCCESS in 29m 02s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/03f23c6a04864338843dd11edc70a0e6 : SUCCESS in 51m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3bcda93c7f3a49f6ade610952b0a9b40 : SUCCESS in 43m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb74773aefc442dabfc17ead03e907e1 : SUCCESS in 58m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37fdd9347e6c42209fc2e392a0a5ff39 : SUCCESS in 8m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee407d4a00734abe82d978c8f8a31e78 : SUCCESS in 13m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d6f38fb82fd84a288eb2c1a9736b2bb7 : SUCCESS in 12m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d80ebe12506a4bbf971d392baa899e27 : SUCCESS in 25m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a377f9a6d424f8b9cb0fcaaa7ee69a2 : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa38f04f65c94cbd96187b2027c0186b : SUCCESS in 8m 32s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/f9ad2778a026446b8b82618dc652323a : SUCCESS in 31m 46s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/813d7e3219ab41edb2c0b2d06fd66527 : SUCCESS in 1h 25m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/048235770a5b4a34ac5214a74df766b2 : SUCCESS in 26m 50s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/079437509ca74d0ea4436a3aa29a6811 : SUCCESS in 35m 09s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/ffa4697bb8d0413db8e7300e70ea9a88 : SUCCESS in 1h 00m 18s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/735684c6da7f41de99006d5068875c1a : SUCCESS in 53m 30s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1480d40a59b44b0db199725b98d0c186 : SUCCESS in 1h 57m 58s\n- nova-next https://zuul.opendev.org/t/openstack/build/2b72bdac3353496f97fcad31db0b51a1 : SUCCESS in 1h 55m 30s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/469135620f864e72bb76f374cef357ad : SUCCESS in 41m 48s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c723c9ce2e5f4b68b30a6b417cad8bd2 : SUCCESS in 5m 13s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ccf26e1b61e541c59bcfeb1dc643c314 : SUCCESS in 28m 11s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/62aef50c07a04ba5948e71fb05687849 : SUCCESS in 22m 09s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/1fd8debb1b574bbc8766766e153cefc0 : SUCCESS in 42m 57s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a1420860f524cfab16d3842cc32b70a : SUCCESS in 1h 37m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f48e4d35e85246148694d5320a0ed77c : SUCCESS in 1h 02m 13s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/ca9a9800046e49d0bef98253e9161ef1 : SUCCESS in 1h 00m 11s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6796262ea1394ed0935a22c438200424 : SUCCESS in 43m 06s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/cda6adaa459341dc91856148df32d92f : SUCCESS in 14m 31s","accounts_in_message":[],"_revision_number":1},{"id":"23779d2ecc881f6326c1440bb6e2073172150123","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 14:16:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5d4f8844d32aa2c5cff3900f05a140cf1286719","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-25 14:25:31.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"4717620b84d22bbf6939569368c50acac6717c7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 15:52:22.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/1001017 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"7fd0def64ec6e66f514495b56c173fc956ee17e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 16:30:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f162d62a0a294dc4bd1e17f0d513f9c0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5370ec1c40174ed5b8c91401bddf44ea : SUCCESS in 29m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c9f6f00e006f402da46c71d6140420c5 : SUCCESS in 17m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f94f45575671490d81ee07b612732a78 : SUCCESS in 18m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7925dc2f1c63e48eefecab4642b2a4c375fc663f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-26 08:27:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7fbed888640f0ccca9208d37567c52d6273ccae2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 10:03:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56990219d767450194f7e34fc774a32a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2b1bd2ae4bc54918b06585773597734b : SUCCESS in 11m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/84fdfaa826084ec28bfaae432fec3b73 : SUCCESS in 21m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e72d2e647ab647fd949921088b108d00 : SUCCESS in 40m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04511b246fce46eba6d36761c3d36f5236165925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 10:44:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ee704d8b4c649ddb3997f7f1c232969\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f9b20c0846e548249f5378af53788d66 : SUCCESS in 34m 12s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/aa2846b7dc0945209d57abae0302ede6 : SUCCESS in 1h 28m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33ac8b5adb0d49b3b002cec4ffcb8340 : SUCCESS in 45m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27171c1b434949e7b80adbcd043dd202 : SUCCESS in 1h 29m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f61f12e03c1c40e5b164769136ddc31b : SUCCESS in 7m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc83cec1328f478c8708c2d50e238e29 : SUCCESS in 12m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4592cc066634717a91ef59ab6c648f0 : SUCCESS in 8m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8b026a4179e44c62b18e2e3aa3fd252d : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdf62d69c6d3425d9e26e1273853d564 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b2cf95f4f404a1f8cce042be093f082 : SUCCESS in 8m 52s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/9a5dacb832e44b6b92efbe0cda204003 : SUCCESS in 31m 18s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/72da81c8fcd940bcbf23db4a9c284c38 : SUCCESS in 1h 21m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f6bc3a69ad4b4c119b2358a904d431b3 : SUCCESS in 25m 50s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/460c300231344070888231ce001bc0c3 : SUCCESS in 36m 14s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/37ac18e8974f41a599b33ecfc1b6393a : SUCCESS in 1h 00m 33s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2b4eb298eac54c1481ca234c8722d3fe : SUCCESS in 49m 45s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2cafb080f547431cb7d473cffc2ac12c : SUCCESS in 2h 08m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/a9ecd73ea3414a3abafad643edd4243e : SUCCESS in 1h 00m 58s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/150cffd141fb44339bbbca5863b148c4 : SUCCESS in 47m 26s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fdc0be2a1edb46ffb4e986dd476b947f : SUCCESS in 5m 10s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d5d34aae31f942188c4f8098c099e33d : SUCCESS in 26m 07s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5c3233e185324637a932f2cfe48290e1 : SUCCESS in 28m 49s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3fadc8bc55d245c494a123abf6513ca1 : SUCCESS in 34m 16s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2cf75894c06447dcba397eb918b58318 : SUCCESS in 1h 19m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99f781a21dde4dfba9ca6d010f5c0473 : SUCCESS in 54m 45s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/475e88ddfc2d454285d2a93330015e80 : SUCCESS in 45m 46s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6a510cc13cb04c47b9992738c05928b6 : SUCCESS in 22m 52s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9f3bf12eb9de4fc5a60cb60d0a728ea8 : SUCCESS in 12m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7933b56baf56e83066cf124f51ade3a109418914","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 10:45:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8eeee9fbefd174ec3c3d434d9f858654b44d7bbd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-26 10:55:32.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"74197d457136387965acb0e9bc2016e6fc0b65d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 12:51:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d4608c51c5746aeb7f4a92944bfdf85\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/267005cc5d144b19bb3882df650cbaf3 : SUCCESS in 37m 37s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/cddca47aee4044019a7c31a322df9e0d : SUCCESS in 1h 41m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/efa0fcab309446998ac47a1597a65f62 : SUCCESS in 1h 09m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/582f2bca63184c46a069843de571755c : SUCCESS in 8m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/682afdd3a8154015a4c83043d9c5bb46 : SUCCESS in 14m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e6c5eff6aa54e86bdd8ab2618c54151 : SUCCESS in 13m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a680c5dcb5ea4bbbac7d3f327e83f884 : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b50546edf9c4000afaea23cd7315016 : SUCCESS in 8m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/27baea9b5fa34cd4a8eaa8319f60bd7f : SUCCESS in 51m 02s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ef85b56edd6d4b0da85940ad69072c85 : SUCCESS in 28m 20s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b2075149ba78415896efb98e862fb183 : SUCCESS in 27m 42s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3e54aeceb4544d579d6dee3c94ec82c5 : SUCCESS in 33m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/42372833b55c4ef8bb4d8888c130e7e0 : SUCCESS in 47m 26s\n- nova-next https://zuul.opendev.org/t/openstack/build/992690f812c44ed3940521f1361cbaf7 : SUCCESS in 1h 59m 56s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/945358d34d514b3b8f382c11adb87218 : SUCCESS in 6m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ab14d01faadd4591842b22653ba6bc95 : SUCCESS in 1h 41m 18s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/94c6e336c2a746bab603f3b23d53a31f : SUCCESS in 1h 16m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b87a8de850e471d9b6c18719b0b806f : SUCCESS in 52m 04s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/aa69fbbe8a7e491ab5b1a7190afbc5a2 : SUCCESS in 9m 56s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/2061ab70d90f42e0b6e4dbbd8eb94764 : SUCCESS in 49m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c684db73409b2831553244da30acef18997038c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 12:51:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"91a9d73abf8901f5dbe0162f9a716a823c232536","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 12:52:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c69f1d3b954496da43db0f5f552f404\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/556b75e2db8e49a99145ddfd1fbcd1ce : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"017acd4ac4e08dee49bb2e6ba1f0fca593fad5d7","revisions":{"017acd4ac4e08dee49bb2e6ba1f0fca593fad5d7":{"kind":"REWORK","_number":1,"created":"2026-08-17 18:38:31.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/12/1001212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/12/1001212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/1001212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/1001212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/1001212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/12/1001212/1"}}},"commit":{"parents":[{"commit":"aaa1d9d59ecce606d187e2e967c256ceaaa54b46","subject":"Use service shutdown related config options in nova-graceful-shutdown","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aaa1d9d59ecce606d187e2e967c256ceaaa54b46"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-17 18:35:48.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-17 18:35:48.000000000","tz":-420},"subject":"Improve the graceful shutdown job ansible tasks name","message":"Improve the graceful shutdown job ansible tasks name\n\nnova-graceful-shutdown job run the test scripts in post hook and\nperform the similar testing tasks for multiple operations like\nlive migration source shutdown, cold migration dest shutdown etc.\n\nBut in zuul console output, the naming of tasks are not so clear\nand it does not reflect which operation it is testing. Improving\nthe same for easy debugging or checking the results.\n\nChange-Id: Ice578bc491dac41ae4de5ef6f2f0397515f0bd26\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/017acd4ac4e08dee49bb2e6ba1f0fca593fad5d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/017acd4ac4e08dee49bb2e6ba1f0fca593fad5d7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aaa1d9d59ecce606d187e2e967c256ceaaa54b46","is_merged_in_target_branch":true,"change_id":"Icfa6009be5b70b4cb46618814f0f01add5298493","change_number":1001017,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
